)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~721824","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ib88cbadb622163d41428b153739679cbba7c336d","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"cleanup-py27-support","hashtags":[],"change_id":"Ib88cbadb622163d41428b153739679cbba7c336d","subject":"Cleanup py27 support","status":"MERGED","created":"2020-04-22 11:46:29.000000000","updated":"2020-04-28 06:52:49.000000000","submitted":"2020-04-28 06:49:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721824-1588056590621-bde3a662","meta_rev_id":"d7fb61d9ba895d4c90952c0b46f28ec2689d2649","_number":721824,"virtual_id_number":721824,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-28 06:49:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-04-27 17:27:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-28 06:19:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-28 06:19:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 17:27:45.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-04-28 06:19:39.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-04-28 06:49:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"402d4b38481f7010fc4a5e81ee06f793d9a479c2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-22 11:46:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6a19ee3fd92ca5419309928c404b9f43aaf30a2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-22 11:49:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3cd4c6fa823c0d6a68f006a9d5a44c1e774d6aac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 12:20:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/59f3ae11804e48acbd132cdf9c569c46 : SUCCESS in 12m 22s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/6afa9e74f9194e7fb2bb37254a24f537 : SUCCESS in 20m 34s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/9dc345e1d1dd40b985b232081f46324d : SUCCESS in 23m 59s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6e10f222ef04d81b99b03190f329865 : SUCCESS in 19m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8108213eadb04bf090f2e3e422487988 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/985f1b504f9a4ca5ae95e1913a80a307 : SUCCESS in 4m 16s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/2534b975ab1a45d09b842c5bb6d5516d : SUCCESS in 21m 16s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/80b512c03d794c0ea8381f62fcfc1f99 : FAILURE in 7m 16s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/be9f0f4fca434516af58432494270b7b : SUCCESS in 26m 05s","accounts_in_message":[],"_revision_number":2},{"id":"1dc604fd31391b019500867bec9b1be75463faa9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-27 15:05:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8c33304f30f47d31b57632c848c73cc4130161b9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-27 15:50:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6893f85a85650f5cf38c326181948ca59d40fb6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 16:10:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0ed62f536462430d8109dfdc58e04cdf : FAILURE in 6m 26s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/782d77d946ba4ba385c5332636082cbf : FAILURE in 7m 56s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/928e36076a7242f8aaeedd8b72256ecd : FAILURE in 8m 23s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81cd7898340545f185cba8fea174ea4a : FAILURE in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd44af33358f4aadbb307ce9fd9c3ca0 : POST_FAILURE in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36c68919f3214d90bc39229add02dbfd : SUCCESS in 6m 27s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/1e11ced73478402184c276daba1c9342 : FAILURE in 7m 06s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/e9edfa6fda4b40cfb94654eb7008ad58 : FAILURE in 9m 15s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/89d47072a0c3493e8dc926c0164c87fa : FAILURE in 5m 09s","accounts_in_message":[],"_revision_number":3},{"id":"6b172ae5dea3ebadf16c0169cc8d8ba21cbbdb2e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-27 16:13:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a8aca7e8c08aa654a6ba7bc335551da1bd776bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:01:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9b81733bb36d45c48e10eabde5033555 : SUCCESS in 9m 28s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/face861adbd44e0db21b4edb8354ab51 : SUCCESS in 19m 41s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/33d0948f4a874f20b4c0e16d706e2b79 : SUCCESS in 29m 36s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6d31a25a8304c0a96ef8123cee94bb4 : SUCCESS in 18m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da8a7c39f99b4e49aa37a91d3bba343d : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99f1c9840c414c4fab46c42d0bb58989 : SUCCESS in 4m 08s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/687ec1866ea643ccbec378b4b82ef6ac : SUCCESS in 19m 18s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/ae3120855e764ea99bc98ee1d2a65967 : FAILURE in 7m 06s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/40396be182d743a99e242ec73eb4bdea : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":4},{"id":"79abc756647d5ae9342113d605035e37253070ee","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-27 17:27:45.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"906db492fa9c0156f84b17c4e8e8d0fc5edf0041","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-28 06:19:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7471bb7f1072baf80ecbde94c750b58e3ad8de5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 06:19:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1a9def16012d23751f6787375fe01000c8402516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 06:49:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c8ef888739a44bc4b615a95308aac248 : SUCCESS in 9m 57s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2cf1311f7d444137872113c3bcab74a7 : SUCCESS in 21m 05s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/df21c010959c45df83cc92288a1f392f : SUCCESS in 25m 38s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c3b872c3b044108a66f08a9d8153213 : SUCCESS in 25m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1faac116377f446fb3344ff52e7cad9a : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6700316e8b534118b79f944d89db29bd : SUCCESS in 6m 13s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/94da160e4d9946dcbec0fda1db708f61 : SUCCESS in 23m 45s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/e4f6cdc7a3e84ae49a162bae0e3ddc6e : SUCCESS in 22m 23s","accounts_in_message":[],"_revision_number":4},{"id":"29d68b64d00bf0f6951f2e84d37b52630b88c1dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 06:49:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ccc6d4f1e80a97af46efc1bcaf0318650023ebfe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 06:52:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a9b3dfa30dc40e3b1f7d419469cf774 : SUCCESS in 2m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ac042a1b82247c489e4c40a0a2492ef : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c33db01ecf4ddfe6d902ba6a0c89900b0d4d0c25","revisions":{"8be43b8e27f10791c88a031884a841658da7f052":{"kind":"REWORK","_number":1,"created":"2020-04-22 11:46:29.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/24/721824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/24/721824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/1"}}},"commit":{"parents":[{"commit":"a6f0eab5cf8963c6b9930b87873f0a18b2834a82","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a6f0eab5cf8963c6b9930b87873f0a18b2834a82"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-22 11:42:26.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-22 11:45:56.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Remove obsolete sections from setup.cfg\n- Cleanup */source/conf.py to remove now obsolete content.\n\nChange-Id: Ib88cbadb622163d41428b153739679cbba7c336d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8be43b8e27f10791c88a031884a841658da7f052"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8be43b8e27f10791c88a031884a841658da7f052"}]},"branch":"refs/heads/master"},"94f0bad7eeb0cb17a199c32c143da85b60e56c68":{"kind":"REWORK","_number":2,"created":"2020-04-22 11:49:49.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/24/721824/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/24/721824/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/2"}}},"commit":{"parents":[{"commit":"a6f0eab5cf8963c6b9930b87873f0a18b2834a82","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a6f0eab5cf8963c6b9930b87873f0a18b2834a82"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-22 11:42:26.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-22 11:49:45.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Remove obsolete sections from setup.cfg\n- Cleanup */source/conf.py to remove now obsolete content.\n\nChange-Id: Ib88cbadb622163d41428b153739679cbba7c336d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/94f0bad7eeb0cb17a199c32c143da85b60e56c68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/94f0bad7eeb0cb17a199c32c143da85b60e56c68"}]},"branch":"refs/heads/master"},"3c2803dd39917f6a5a3d3a8aee379194c2742b27":{"kind":"REWORK","_number":3,"created":"2020-04-27 15:50:54.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/24/721824/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/24/721824/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/3"}}},"commit":{"parents":[{"commit":"a6f0eab5cf8963c6b9930b87873f0a18b2834a82","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a6f0eab5cf8963c6b9930b87873f0a18b2834a82"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-22 11:42:26.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 15:50:25.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Remove obsolete sections from setup.cfg\n- Cleanup */source/conf.py to remove now obsolete content.\n- remove install_cmd, move constraints into deps,\n  small cleanup of tox.ini\n\nChange-Id: Ib88cbadb622163d41428b153739679cbba7c336d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3c2803dd39917f6a5a3d3a8aee379194c2742b27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3c2803dd39917f6a5a3d3a8aee379194c2742b27"}]},"branch":"refs/heads/master"},"c33db01ecf4ddfe6d902ba6a0c89900b0d4d0c25":{"kind":"REWORK","_number":4,"created":"2020-04-27 16:13:55.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/24/721824/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/24/721824/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/721824/4"}}},"commit":{"parents":[{"commit":"a6f0eab5cf8963c6b9930b87873f0a18b2834a82","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a6f0eab5cf8963c6b9930b87873f0a18b2834a82"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-22 11:42:26.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 16:13:49.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Remove obsolete sections from setup.cfg\n- Cleanup */source/conf.py to remove now obsolete content.\n- remove install_cmd, move constraints into deps,\n  small cleanup of tox.ini\n\nChange-Id: Ib88cbadb622163d41428b153739679cbba7c336d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c33db01ecf4ddfe6d902ba6a0c89900b0d4d0c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c33db01ecf4ddfe6d902ba6a0c89900b0d4d0c25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
