)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~723581","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Iab961aba8d1957aa40e7b71796e6078c8c20e5db","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"cleanup-py27-support","hashtags":[],"change_id":"Iab961aba8d1957aa40e7b71796e6078c8c20e5db","subject":"Remove setup.* files","status":"MERGED","created":"2020-04-27 15:11:55.000000000","updated":"2020-04-28 13:09:59.000000000","submitted":"2020-04-28 13:08:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723581-1588079294087-2ccb4115","meta_rev_id":"183949144a39b8ab6e00f019e43161baa180602e","_number":723581,"virtual_id_number":723581,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-28 12:17:18.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-28 13:08:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-04-27 17:27:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-28 06:20:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-28 06:20:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 17:27:27.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-04-28 06:20:15.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-04-28 13:08:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9910bc54123ad12cd3e67bab90caca9ae9b9989e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-27 15:11:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab7e9678d803103d0b4bf6de3340735fd5e2b65e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-27 15:12:16.000000000","message":"Patch Set 1: Workflow-1\n\nLet\u0027s test whether our assumptions are correct","accounts_in_message":[],"_revision_number":1},{"id":"aac5c421b30250e9f875287c9157d36b91528ed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:34:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/020662c0d1f148e995038a0383026a21 : FAILURE in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9136f8ae7d4145a29b741deb15b5ad36 : POST_FAILURE in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7991a4b9e350abc6bacea6c31a83bb45fe7ad800","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-27 15:50:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"54a08bcd67c23f299428686507169a54f2a1b04d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-27 16:13:55.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"a4c07999311dbdbd6965d1743cc5002c8270b97f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 16:51:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e9ab7428566f45ad91e2712a6d768e27 : SUCCESS in 6m 53s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c9695ae5a4a64da7a09f2895da10db0d : SUCCESS in 18m 00s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/5df7c83134204f5594663fdef73af9ae : SUCCESS in 18m 37s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63fe6f3f884e451a981dd0411f678b31 : SUCCESS in 18m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b00cd7ae75244488ed41cfb8b3c6bcf : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d550818801754ef696870a4b915f2e1e : SUCCESS in 5m 29s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/53c9ca578fa747eb832d7c66bf38ce41 : SUCCESS in 20m 35s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/125904e0198e4d1f95f322acb403d9de : FAILURE in 5m 52s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/8b456b8fbe884d64b304bb3c16b72810 : SUCCESS in 18m 48s","accounts_in_message":[],"_revision_number":3},{"id":"c717f52d4e657459e611440bba48ff5ffe4442c7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-27 17:27:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a36deed1246e7e9d3b5409aeebea0003b207133d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-28 06:20:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"47f4fb1c27ec3399a789c897d306489a533c5f56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 06:20:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b3d5cd155e38a01958699498a47402f1b1980792","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 07:10:19.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/71076bcbc83045ce9ba0c9add524cc78 : SUCCESS in 10m 38s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/92fab64f4ec44df9b6362a461fb28e8d : SUCCESS in 20m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/a29edc02618d4210897d0ba423ecf339 : SUCCESS in 23m 41s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a349553068494f64ae632bd6f186f8cf : SUCCESS in 20m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7433ebcc49fc43aaa493cc90fe585838 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a6c5e0b93d04783a69e4a0c2c12b61a : SUCCESS in 6m 09s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/fb8834ee214940aa8b2592a2a1df4ab7 : SUCCESS in 21m 23s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/01f176731eb74643b9e2cf9d1c89ac7a : FAILURE in 43m 35s","accounts_in_message":[],"_revision_number":3},{"id":"af280341a859f75c28a82fc6d309871fc8378135","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-28 07:23:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e9cb422eb6ccbf4e984fe21eb6c94823e85e6d3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 07:52:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ac063d74abf24e1482977279423ad045 : SUCCESS in 7m 10s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/15938e67e9034c229f2b4f2ac9e42cc5 : SUCCESS in 22m 57s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/110e2c2fb1594865a24a0302f985c13e : SUCCESS in 22m 45s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6e9dd2ab3a644a0af8dffbfaef44ae0 : SUCCESS in 19m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d807cc0d5dd460dbfcb4fbd1864b40c : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/853570cb620c41658ab06a2967ff1826 : SUCCESS in 7m 34s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/b3a389d1ce844582aa804554758d81c1 : SUCCESS in 21m 50s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/a99142d1688f4bd3bd476a064faf455e : FAILURE in 14m 47s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/18c4058373e24b309aff0b3d70a4f105 : SUCCESS in 22m 48s","accounts_in_message":[],"_revision_number":3},{"id":"328165aad18a8ad12db884db00d1c86b679bda6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 07:52:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"74081f768dadf2690aa00bc3b9d99907270a26f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 08:21:51.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5e36febd0caf4685bbc1fe1106ec8a20 : SUCCESS in 7m 44s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/bc5ebfc7e3cc4b158516dfe31c108112 : SUCCESS in 24m 17s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e4a4bebf03934fd9ad79a8b1342b7558 : FAILURE in 15m 13s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a1fdcc4711547468be4c5295f7b3c18 : SUCCESS in 20m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcf6ab0775964f0fb35893220468d545 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcd9dee1080048288b3a0b10885394b1 : SUCCESS in 5m 25s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/461d1bd22d314a398dfec04ebb585513 : SUCCESS in 23m 37s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/1415257392304176a70ad853b489588d : SUCCESS in 19m 27s","accounts_in_message":[],"_revision_number":3},{"id":"e3ca7a50b633781fd677fbffa596366c030a9080","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-28 08:24:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"69b063c3e1437cc0ff3597edaa36b3de5fd21621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 08:51:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/21ec57d572894b1ab7867122b7ef4d1b : SUCCESS in 7m 32s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/84e8db163ab04b1ebd36e61efbf2c79a : SUCCESS in 19m 40s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c33bf8c94db447fb9d6792b82cfa8d97 : SUCCESS in 20m 30s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab4af8df1f344461b392eb1942d3af79 : SUCCESS in 23m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b5ac749819042c5b9f4eaf74a4769fe : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22d1656305b6406c8362910fa47a39de : SUCCESS in 3m 41s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/ea8e5eb78e694e7687f865589c310da0 : SUCCESS in 21m 05s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/aeb8bdefde894bdebfe17e897e310cba : FAILURE in 10m 36s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/8733dc50af034955a92219015a7b4ec7 : SUCCESS in 20m 51s","accounts_in_message":[],"_revision_number":3},{"id":"3be255e4f7bda6321b6392261bad48263352d5e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 08:51:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"93ee9f2ffce7536bb950214dec6136dc1f715621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 08:56:19.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/77f693609f5a48219561679540aa5edc : EXCEPTION\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/3bbf60d669af416ba54b73cf1920bb58 : EXCEPTION\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/7f7814bd7ae04983a81a3f20c2c7e38c : EXCEPTION\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6ecbf46ceb646dfb673f9dedbe6b50e : EXCEPTION\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5ba09461da84f6e8a00e0a24276d800 : EXCEPTION\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29d61c8fefe74210af457b56a301ca05 : EXCEPTION\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/ad4161be9d0b48178e63f1daa154abe7 : EXCEPTION\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/239f5e564a474b3f8756df6859eb46f1 : EXCEPTION","accounts_in_message":[],"_revision_number":3},{"id":"476d31a88901b574b237ecb60512e0fc9679a49d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-28 12:17:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d2f56dad54a26a610dbde367437fa351fbe51e74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 12:41:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2f021fe2dc374eacb62b2548ca0ea8e9 : SUCCESS in 9m 23s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f462726ae5af443a97c373227acf3748 : SUCCESS in 21m 09s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/7ebbde19e8a445dc9840b04a85dc46d5 : SUCCESS in 19m 14s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38f2d4fb20634c5392aa43184b2288b7 : SUCCESS in 21m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b330d6e1285647e38bc40365245093f8 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/955be61cfaa3487aa45ec1b8be5ea524 : SUCCESS in 3m 53s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/e0d7c1cbb89d4db79d11cf13459d6ef3 : SUCCESS in 18m 40s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/2f2bc4659627431c92f439b8292da8ed : FAILURE in 9m 56s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/a41603916ca2495ba19703a6628dee19 : SUCCESS in 18m 20s","accounts_in_message":[],"_revision_number":3},{"id":"710f2da3a0dadf297c95841d4d96c46a1c7dbe66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 12:41:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e4d64bc96bcd7a4289c808f5a2a844e6f3d5b086","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 13:08:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9af14cf1b6fc4f73bb4aa1fc14f9600c : SUCCESS in 7m 08s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f83fb7592fb4412285cab73ec99d700b : SUCCESS in 18m 57s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b3ac44edf955419bb3583ed574e19d46 : SUCCESS in 20m 03s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d77b426e820244a1ab4ed5df730c7b98 : SUCCESS in 16m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd231ad3ea743098dcfa67122bfa3ad : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/916ceaee72e740af8d476c8fa1455cd4 : SUCCESS in 3m 36s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/6aa9921bf6924c7d9fc47ff51286410c : SUCCESS in 19m 41s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/858ba49e0d2b4f029f062814e2b8b611 : SUCCESS in 25m 04s","accounts_in_message":[],"_revision_number":3},{"id":"1e7858a622b673a87b0940779c34789467e68ff4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 13:08:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"712c67c49f746be7b69618d7d63f6a9aab14940a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 13:09:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd17bf606174f53aef2a4f780f97c80 : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96c91c1aa1fe4831b6cde40442891afe : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1ba6cc04f9ddf5e07e11ff340f3df728e4d6620b","revisions":{"27aaa0cfaf0b3881e5a9f9bc23c22988ea6dff94":{"kind":"REWORK","_number":1,"created":"2020-04-27 15:11:55.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/81/723581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/81/723581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/723581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/723581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/723581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/723581/1"}}},"commit":{"parents":[{"commit":"94f0bad7eeb0cb17a199c32c143da85b60e56c68","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/94f0bad7eeb0cb17a199c32c143da85b60e56c68"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 15:11:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 15:11:04.000000000","tz":120},"subject":"Remove setup.* files","message":"Remove setup.* files\n\nThis repo is not using pbr and not publishing to pypi, the setup.* files\nare unused.\n\nRemove them.\n\nChange-Id: Iab961aba8d1957aa40e7b71796e6078c8c20e5db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/27aaa0cfaf0b3881e5a9f9bc23c22988ea6dff94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/27aaa0cfaf0b3881e5a9f9bc23c22988ea6dff94"}]},"branch":"refs/heads/master"},"fa5c4db5b90d102066ab463690ad6ba36af65598":{"kind":"REWORK","_number":2,"created":"2020-04-27 15:50:54.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/81/723581/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/81/723581/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/723581/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/723581/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/723581/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/723581/2"}}},"commit":{"parents":[{"commit":"3c2803dd39917f6a5a3d3a8aee379194c2742b27","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3c2803dd39917f6a5a3d3a8aee379194c2742b27"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 15:11:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 15:50:47.000000000","tz":120},"subject":"Remove setup.* files","message":"Remove setup.* files\n\nThis repo is not using pbr and not publishing to pypi, the setup.* files\nare unused.\n\nsetup.* files are used by UseDevelop in tox.ini but since there\u0027s no\ncode to isntall, we can change the switch to false.\n\nChange-Id: Iab961aba8d1957aa40e7b71796e6078c8c20e5db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fa5c4db5b90d102066ab463690ad6ba36af65598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fa5c4db5b90d102066ab463690ad6ba36af65598"}]},"branch":"refs/heads/master"},"1ba6cc04f9ddf5e07e11ff340f3df728e4d6620b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-27 16:13:55.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/81/723581/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/81/723581/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/723581/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/723581/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/723581/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/723581/3"}}},"commit":{"parents":[{"commit":"c33db01ecf4ddfe6d902ba6a0c89900b0d4d0c25","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c33db01ecf4ddfe6d902ba6a0c89900b0d4d0c25"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 15:11:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 16:13:49.000000000","tz":120},"subject":"Remove setup.* files","message":"Remove setup.* files\n\nThis repo is not using pbr and not publishing to pypi, the setup.* files\nare unused.\n\nsetup.* files are used by UseDevelop in tox.ini but since there\u0027s no\ncode to isntall, we can change the switch to false.\n\nChange-Id: Iab961aba8d1957aa40e7b71796e6078c8c20e5db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1ba6cc04f9ddf5e07e11ff340f3df728e4d6620b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1ba6cc04f9ddf5e07e11ff340f3df728e4d6620b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
