)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~725271","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I41e2b57a78fac585d6ad71b6bcd606d4248111f6","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"I41e2b57a78fac585d6ad71b6bcd606d4248111f6","subject":"Use more reliable way to get mysql version","status":"ABANDONED","created":"2020-05-04 13:34:34.000000000","updated":"2020-05-04 19:19:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"33bfbd74fe0a6b65d79e1582e318bbac09e8ad06","_number":725271,"virtual_id_number":725271,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-04 15:51:16.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-04 18:00:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 18:00:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c44dc3cd66a72f3887dc39b6f57b43922508838","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-04 13:34:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b23183708f59beb906125b91846407ce5f43a8c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 15:36:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/25c69a95f9af4f2d93153771c8cbe544 : SUCCESS in 7m 32s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/6a944a8c6d4744d8bb2fc13f88e3a50d : SUCCESS in 19m 53s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/7aa4f9b9be0b45819f85b195da5bf706 : SUCCESS in 26m 19s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cea7c6406fc45dcb15a164080d7a530 : FAILURE in 13m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a66f30ceae944b9aa0860e659a268881 : SUCCESS in 1h 09m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e58dbf82425b4d718abf4a21043401ee : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/dcb81bebe8224809b2ae189635040e03 : SUCCESS in 1h 24m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aef458b508884d31b360fe2e58cdb1ef : SUCCESS in 1h 32m 28s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1976b4f8b2a340149375b898f7a6e7d9 : FAILURE in 23m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f28b329c7ba49b19c500e6cdffe22f3 : SUCCESS in 1h 57m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/50bb53ad848e43e681a561f46d940ca6 : SUCCESS in 1h 05m 45s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/09a019747eba4e419cd458a39798f10f : FAILURE in 49m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e8642abc29f4d2aaf65a72f7e0dde2f : SUCCESS in 1h 00m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feae7a1aaadf46ae973cd5cfd9493ce8 : SUCCESS in 7m 09s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/644ad58c8a3c43c7949955fa4166c9cf : SUCCESS in 21m 27s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/8df75a6aeae2415f87166667c4235773 : FAILURE in 11m 49s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/fd8cdd4a347c4725ab3052f4c36b42a6 : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":1},{"id":"af77d24242b68086379a02db83b4427832683aa1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-04 15:51:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ea85eb1c9117b0c69a4d00efdf9028538d48b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 18:00:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e4f4d485c9654182861682828d071de7 : SUCCESS in 6m 37s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8d67d8e8d956479098d921142da84bb3 : SUCCESS in 22m 33s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/30ed1c8f17d245cba4d18710d95f0857 : SUCCESS in 28m 25s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26c2ce9700c1434ea923b41c1ae140d7 : SUCCESS in 31m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4751836a18a646578dc70002420e861a : SUCCESS in 1h 19m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/264aee3fd0ef4847b36364e9eada4d52 : SUCCESS in 1h 39m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/26721f6b0215438f8dc9e73c726bfcd7 : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69867b8086744b6f81cf5979dde42df4 : SUCCESS in 1h 21m 57s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/82afb842412c49259f8acd8f0e41662a : FAILURE in 26m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abb70feb3dc94322b37297bbbeebfa0e : SUCCESS in 2h 05m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/00490069f6dd4527a4685c99df1663b2 : SUCCESS in 1h 18m 25s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1a7c9830d5404b46bdd91e5e52c5db4f : FAILURE in 37m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3ed739208f34367ae061f9f5837a18e : SUCCESS in 1h 01m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a548646eab4475194e39a3219d4ff54 : SUCCESS in 4m 44s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/b4511b06fb6948ab81275d221d512ac2 : SUCCESS in 20m 41s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/2db25295cdea44458dc3308e5605d5c9 : FAILURE in 9m 59s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/33a7597ff9fa4c57b92927f6636f5727 : SUCCESS in 21m 57s","accounts_in_message":[],"_revision_number":1},{"id":"33bfbd74fe0a6b65d79e1582e318bbac09e8ad06","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-04 19:19:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8556bc713891f4bbb300b2ce14e737771f1b2898","revisions":{"8556bc713891f4bbb300b2ce14e737771f1b2898":{"kind":"REWORK","_number":1,"created":"2020-05-04 13:34:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/725271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/71/725271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/71/725271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/71/725271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/71/725271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/71/725271/1"}}},"commit":{"parents":[{"commit":"1ba6cc04f9ddf5e07e11ff340f3df728e4d6620b","subject":"Remove setup.* files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1ba6cc04f9ddf5e07e11ff340f3df728e4d6620b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-04 13:34:25.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-04 13:34:25.000000000","tz":180},"subject":"Use more reliable way to get mysql version","message":"Use more reliable way to get mysql version\n\nmysqladmin --version displays version of client instead of server one.\nSo we should make decision to run mysql upgrade or not based\non the server version. Also this task should be ran only when mariadb\nis already deployed.\nThis will make possible to be more independent from client version.\n\nChange-Id: I41e2b57a78fac585d6ad71b6bcd606d4248111f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8556bc713891f4bbb300b2ce14e737771f1b2898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8556bc713891f4bbb300b2ce14e737771f1b2898"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
