)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~727115","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ic5b2d0c8d9f30284587a61385c2d78517a82f6e8","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"Ic5b2d0c8d9f30284587a61385c2d78517a82f6e8","subject":"Local facts are strings so make this explicit","status":"MERGED","created":"2020-05-12 08:51:31.000000000","updated":"2020-05-12 17:41:35.000000000","submitted":"2020-05-12 17:39:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727115-1589305159807-fc221c0e","meta_rev_id":"7870a7b9a1268bd281995c30279a53f63756cfab","_number":727115,"virtual_id_number":727115,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 17:39:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-05-12 15:51:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":2,"date":"2020-05-12 14:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-05-12 15:51:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 14:52:17.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-12 15:51:41.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-05-12 17:39:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e6015b635f25c76f0b4e0e772c6b7479a52c158","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-12 08:51:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70a65dd333942d7a8164f7e318469415d8ca9b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 11:23:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0576a2c7fa2741c99bdcb33046aa293d : SUCCESS in 7m 11s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b0b1392011544575bd769a4c2c7313c8 : SUCCESS in 17m 18s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b5c690102acb4ca3939d825bccf27ade : SUCCESS in 17m 58s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7a56d3372c04a9bb066a2fad9acfcd4 : SUCCESS in 18m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/be11731d5b2942199020ad6e8a1f8bcc : SUCCESS in 1h 02m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aa573ae2f7a24e3aa322d7d445f1ea9d : SUCCESS in 1h 28m 08s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bbd2619c791c4f7eba747bac0619ac97 : SUCCESS in 1h 19m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebe4871af06c4516981b6d7b7c9bb974 : SUCCESS in 1h 12m 09s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9173296fb7254fb6b55a414b2435bac6 : SUCCESS in 2h 25m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fc93833d2b741169a5b77e3adb245a4 : SUCCESS in 1h 44m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/18608166ca574781b78b75b16312f7b0 : SUCCESS in 1h 13m 21s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0d74bdfa80574a30a6ddc0809231e028 : FAILURE in 36m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17a89fc08d4f4fe38db28ade1d41390d : SUCCESS in 56m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc0a30badb91449b9784d61923ba3ae4 : SUCCESS in 4m 42s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/612eba1d6c4a4b5db6a39a15ecb15ab5 : SUCCESS in 22m 58s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/68671887ef2c43089fc81fffeaaa8f44 : RETRY_LIMIT in 3m 27s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/f567324c4e234440b6a9f314ede4158a : SUCCESS in 19m 09s","accounts_in_message":[],"_revision_number":1},{"id":"37ce8afb4a8839ce3707e2b89d8d7368f14b7cdc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-12 14:52:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8eba633f334f7b346b9da4762fd5bf63f1479a34","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-05-12 15:51:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4bead6a8a54da5edc61834402acd779c5786e5ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 15:51:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"716530da6552078ff9abf6fb9bc7824acaab9889","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 17:39:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"140369dc267e95300b5068baa8e1068a46117f9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 17:39:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5c6684be73d14d678c99b5445c64c5b1 : SUCCESS in 9m 48s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/96984a62d2c94b4b92ac97a5ce98da31 : SUCCESS in 20m 10s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/adae307e65454a58a451f283d544fd33 : SUCCESS in 20m 01s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0fdb6b3b94d422aa9c5be519a9b3a70 : SUCCESS in 21m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b8f975bf5d5e41aa88b66816fb657d89 : SUCCESS in 1h 14m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/98dfd7c5ca1e4612bc44d51f9ea8dee6 : SUCCESS in 1h 30m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d92cfaffa35b485a88b4f6af5f5b71dd : SUCCESS in 1h 41m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/096576e88f6947cf8b07cc6c14b1e023 : SUCCESS in 1h 10m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/86c78839f4c748c9b1476e58c817c798 : SUCCESS in 1h 09m 59s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07d19db532d649039e3835e5a277cf92 : SUCCESS in 1h 02m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e8f8b76bb54ba1837ec59edfd55a5e : SUCCESS in 7m 24s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/52602670cea34755873a86473e5f2ba2 : SUCCESS in 23m 25s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/c63690517928445a9e22ccd32a7a1aa2 : SUCCESS in 21m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7870a7b9a1268bd281995c30279a53f63756cfab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 17:41:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d2ca65e29f24153bd3fd0812cded61b : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9600c682c409aee800b7ba56b1373a07683566c1","revisions":{"9600c682c409aee800b7ba56b1373a07683566c1":{"kind":"REWORK","_number":1,"created":"2020-05-12 08:51:31.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/15/727115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/15/727115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/727115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/727115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/727115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/727115/1"}}},"commit":{"parents":[{"commit":"67c0a5437dea194eec637f6e738e2c3b22c94b0a","subject":"Merge \"Use more reliable way to get mysql version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/67c0a5437dea194eec637f6e738e2c3b22c94b0a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-12 08:48:54.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-12 08:51:18.000000000","tz":60},"subject":"Local facts are strings so make this explicit","message":"Local facts are strings so make this explicit\n\nThis patch fixes the following ansible warning:\n\n[WARNING]: The value True (type bool) in a string field was converted to\nu\u0027True\u0027 (type string). If this does not look like what you expect, quote the\nentire value to ensure it does not change.\n\nChange-Id: Ic5b2d0c8d9f30284587a61385c2d78517a82f6e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9600c682c409aee800b7ba56b1373a07683566c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9600c682c409aee800b7ba56b1373a07683566c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
