)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~731290","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Ftrain~I97e8e9b8cb5aa4901207079c0c45813fe9d7f887","project":"openstack/openstack-ansible-galera_server","branch":"stable/train","hashtags":[],"change_id":"I97e8e9b8cb5aa4901207079c0c45813fe9d7f887","subject":"Ensure python3 mysql client libraries are present","status":"MERGED","created":"2020-05-27 18:31:28.000000000","updated":"2020-07-24 11:38:25.000000000","submitted":"2020-07-24 11:37:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731290-1595590641083-78954b85","meta_rev_id":"7f4a60263a3af2ebf76d36803cd34761601c812f","_number":731290,"virtual_id_number":731290,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-05-28 08:20:13.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 11:37:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-07-22 10:49:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":1,"date":"2020-06-29 13:15:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-29 07:28:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-07-22 10:49:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 07:28:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-06-29 13:15:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-22 10:49:29.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-07-24 11:37:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b6132b7ebe4f327fa74f4b5f799b34b60de21cc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-27 18:31:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4cef4d37ac8474eec54865b3c74647903a9d4a29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 21:40:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/774522ff1e384363950784c2dadbab36 : SUCCESS in 6m 30s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/e62a23acde8547e7b063d5d964ceab27 : SUCCESS in 14m 46s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/a96680b192bd478786577dd1182d1f4b : SUCCESS in 18m 28s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/010430796f094a5081f543072ba39024 : SUCCESS in 16m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ec743647445a41329811c4f9a35f89b4 : SUCCESS in 1h 08m 45s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/22527fec16af4d619218d10d5335e93f : POST_FAILURE in 1h 30m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9979c308020a48d5818d9ed7c75d592f : SUCCESS in 1h 36m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3802934702fe4fa8b8ebe8ed0c0bbaae : SUCCESS in 1h 32m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9058468bf8e049ea9f5783ea61e27a46 : SUCCESS in 1h 17m 36s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/52380f6a8e054c56bb4f592f6e6f48eb : TIMED_OUT in 3h 02m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59aaea4bf4a944db96abbe0d52532d3f : SUCCESS in 1h 57m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d0ceb9fa1c534d8b982599fb76e81991 : SUCCESS in 1h 12m 47s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c51dce3eb61a40bfaac82e5f64ae5784 : FAILURE in 33m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f75429d81b734eddba61da710d795c48 : SUCCESS in 55m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5343d0e262894902b2cb4549b9f6cc0a : SUCCESS in 5m 38s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/2dfcfdb1a6af4d9192d387ccaec675a8 : SUCCESS in 19m 10s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/1543ab8ee16c4ef2a4558f0e297a6fc8 : FAILURE in 11m 16s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/b1b61b934fc044e3aada75405ba29c50 : SUCCESS in 17m 05s","accounts_in_message":[],"_revision_number":1},{"id":"636365be914944dbc8ebb834c6324d213f605cc1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-28 08:20:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e592487184a03189839e876fde3815eae2c02c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:52:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b5f1fc2f30734f3298d65ea505c9f589 : SUCCESS in 6m 13s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/62a32446cf67473d81c5e6653d493923 : SUCCESS in 15m 05s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/90e20c1e7e7247138968ad1241cec953 : SUCCESS in 14m 17s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/884860fd6c1f4a868d6ff1e9e3812227 : SUCCESS in 14m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dd534d30ebaa408eae1e9ddbd172254a : SUCCESS in 49m 46s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/56c9009faf0548aa8d88c55a0b6b58cd : SUCCESS in 56m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7f01fd701af84e48a6e7955a1a1a8983 : SUCCESS in 1h 01m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d7a1a6928b534517904f9ed28d0f1d6f : SUCCESS in 1h 00m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20c60068ea6f4b9dbcbfbd58b8c0eef9 : SUCCESS in 1h 18m 47s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/98d1fb608ccd4902aee2982839a7b3f5 : SUCCESS in 1h 30m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bc8b0299dc9408592f018d863ed0fcb : FAILURE in 1h 25m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/789f0de6789746cfa657fb88a39dde45 : SUCCESS in 1h 19m 40s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f6184669e3814ce6885bca2cfb1a31f2 : FAILURE in 35m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d6547a5bc4b4b0088204122ce270956 : SUCCESS in 54m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a1afef62ed24c6a94ff1ac288791bf0 : SUCCESS in 5m 51s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/0d6ebb4fa60a4a49a00597a491e0f962 : SUCCESS in 18m 38s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/e7ae18a4c3154131953489d3a77f1089 : FAILURE in 10m 21s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/13da2d56ee2c4a00b6fa6fe4cef8b9e5 : SUCCESS in 17m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0a33a1ecb844452c3e27b92d135c73a7c10d4500","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-26 17:13:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"426b5b1a56269845f40c8011d16d3a39ec0849a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 19:56:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/17558e83eb7a47f39745aa5dd683c426 : SUCCESS in 8m 51s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/edf62ea60245466caa91001dfded3160 : SUCCESS in 15m 13s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/554a4dccb5e8431db6116502b94cc9d5 : SUCCESS in 20m 49s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b0f45632c424536b08ddf97c62efc87 : SUCCESS in 19m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8f45df01fc064c59b778e80990420c00 : SUCCESS in 1h 02m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2d6e145d0f3241d98d2bd3b75a2bc0bd : SUCCESS in 1h 36m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8de61878fad5450086c07f66c9613b6a : SUCCESS in 1h 27m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/14e9263f7ac34170a963f959920f35c4 : FAILURE in 1h 15m 26s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be69721c6c7e43d99df02bc3bd6feaa1 : SUCCESS in 1h 10m 58s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/190c0371156e45e386bb0f51c6f02ea4 : SUCCESS in 2h 24m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b4338e882bf4a88b3cca05c0133a6c3 : SUCCESS in 1h 42m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aa5d355ca1ad4887b9d1f1efdc383c72 : SUCCESS in 1h 11m 09s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7cb4ad8457fc4bf48f89c5827a734963 : FAILURE in 29m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e614c2e1c7a4fe3a9677561e62f64e9 : SUCCESS in 49m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/583769085649495b985ab186ce687c22 : SUCCESS in 6m 50s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/6893d957f259496aa22e1d90d55ec631 : SUCCESS in 21m 51s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/e1e2948d6650428bb95bf85a2d7848df : FAILURE in 11m 52s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/fc17cb25ffce4dd4b32d7d332e3df40e : SUCCESS in 19m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fbaef63c925e6486fb430aa526a937bf046f9463","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-29 07:28:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0dd6468ad57eedcf3c38399ab4cd74b108d2bdf5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-29 13:15:58.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d83d3ace9e6c29820dbeb1118618446da42349a","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-07-22 10:49:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5086cd32e71d1653abf3058d4c0a794b98390e22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 10:49:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b435ce8b20dd3df10502be8757ec511740f81925","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 12:57:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/31b0fb277e5240cdb37259a95622fa37 : SUCCESS in 7m 00s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/b512ebe62dae4adab47717dd3406938c : SUCCESS in 16m 57s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/7e9dbf7833ce40379082995612180e86 : SUCCESS in 18m 47s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5bbc3c325c2c4b4b8c51eee20ea8abf6 : SUCCESS in 20m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7187f98cc03d4f4293e610336a3ab417 : SUCCESS in 1h 03m 29s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8b55bd1012194479a4e12544843a1258 : SUCCESS in 1h 18m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/19f5e1e03f044f3499a33ddcace59046 : SUCCESS in 1h 34m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9c4a143cbf9f4321888a4cbda0c52573 : RETRY_LIMIT in 18m 29s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/061ddb6376e7440fae4cfe5aa6240c79 : FAILURE in 24m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bbe698e9c2f46b9ba1d9706deba8be5 : SUCCESS in 1h 47m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b02462b9543243bdb37f36f5d642bdf6 : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1de5eb3b033646ab80b6e9de8bce9168 : SUCCESS in 56m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c006157f4a94c9a9b30a983e9586195 : SUCCESS in 6m 42s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/b5d7b103d5b0427fb492729468a906cc : SUCCESS in 20m 43s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/d2c76597141643f8bc54fa6d81ec2d2b : SUCCESS in 19m 01s","accounts_in_message":[],"_revision_number":1},{"id":"58f2c29b773c6456afa7bef0f83467efa6013462","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-24 07:03:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fe5bb76b1363c23a7c711cf17b397562bd3d3b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 09:38:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/854fc963ca43489c9d29735584afaae0 : SUCCESS in 6m 40s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/ab318e64b000435daa9971f30651d680 : SUCCESS in 13m 33s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/7d36526f92264cd6bc57a0812dac4790 : SUCCESS in 18m 24s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/224f0c5ea5d748e1bf6452677a9eb5b6 : SUCCESS in 18m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/258490ce19d84c5f82931c131665ac3e : SUCCESS in 1h 04m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/311b4024e824441c93a1caecddaa6d91 : SUCCESS in 1h 20m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5ce529c2537b4e1bb41dc75de6efe635 : SUCCESS in 1h 29m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1317ce13a2bd41dd9a9a540d46a4e903 : RETRY_LIMIT in 23m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/860d2e9e62204c6eb70d88b6bbb6720f : SUCCESS in 1h 14m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/50aa212bef804b0f9c28eb6f4cd10405 : SUCCESS in 2h 17m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c1f57bd604b489ca47566d82ce96e2b : SUCCESS in 1h 51m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/893a74af45f049738bed19f3f04d84e4 : SUCCESS in 1h 06m 18s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6f818ab4050240f9816a3bc12abdf261 : RETRY_LIMIT in 21m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89baf17404554d7cbf8b86870fafe271 : SUCCESS in 52m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b266a018bf66451eaef03024b439aeb2 : SUCCESS in 7m 30s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/36b7f9ea21434290b97278bf1572c495 : SUCCESS in 20m 06s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/a898dfffb68e4c47a9f1ac6b9a5ec35f : FAILURE in 13m 52s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/08559380905547f49d031bb3c5af8033 : SUCCESS in 18m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2db36eb568860594457296f87e0f77560a1e0bbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 09:38:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f480a48ead67f5dfd9a1f5ac0da8c9641ea41d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 11:37:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/865a4bb6b2fa412397b30c348ee8f093 : SUCCESS in 9m 01s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/bebf855df08840809a72327e6e2ff2d5 : SUCCESS in 13m 30s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/eeb43b2b164444439ee4b28d778a15a8 : SUCCESS in 17m 02s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12850e2135b145c1aab7c1c7e24a49a8 : SUCCESS in 17m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f6324d2bfb66427a9b3a914afd054da5 : SUCCESS in 1h 19m 11s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9f856cb989704b01a95928c453d514e5 : SUCCESS in 1h 23m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/39b20dea1bf4414d80c72726becfa35e : SUCCESS in 1h 32m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e71d5a689b85432cb759f3435916f022 : RETRY_LIMIT in 17m 29s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49b2a01b9fbc4ef79af5c52642090330 : SUCCESS in 1h 11m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1511ec911494c40b7984c19e8ec6b96 : SUCCESS in 1h 49m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/40424edcc6324e3cb566e1cff76f33f8 : SUCCESS in 1h 05m 24s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d73a174f119849f48546d5d69589c2c9 : SUCCESS in 43m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0639fcc6ef744cc89e64f89e98cab07f : SUCCESS in 8m 25s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/710948f5c0994cf3b5106f3cff7b21b6 : SUCCESS in 16m 43s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/95d49a6e4aa7441ca1c9b296f5e973db : SUCCESS in 18m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3b235097c0565b88e3898a44ae9e217b75738336","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 11:37:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7f4a60263a3af2ebf76d36803cd34761601c812f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 11:38:25.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07b039f078594f948a01b794bb023857 : FAILURE in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"021b1621e4a9d4ac5de606884eecd8ffb9efcc7c","revisions":{"021b1621e4a9d4ac5de606884eecd8ffb9efcc7c":{"kind":"REWORK","_number":1,"created":"2020-05-27 18:31:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/731290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/90/731290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/90/731290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/90/731290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/90/731290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/90/731290/1"}}},"commit":{"parents":[{"commit":"18c563c9ccef6eb15d12c28ea5efd52f919e1896","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/18c563c9ccef6eb15d12c28ea5efd52f919e1896"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-03 20:09:04.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-05-27 18:31:28.000000000","tz":0},"subject":"Ensure python3 mysql client libraries are present","message":"Ensure python3 mysql client libraries are present\n\nThese are required when ansible uses python3 on the target host.\n\nChange-Id: I97e8e9b8cb5aa4901207079c0c45813fe9d7f887\n(cherry picked from commit d74863bc52255f2a470b26a160b78f3009de2c2b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/021b1621e4a9d4ac5de606884eecd8ffb9efcc7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/021b1621e4a9d4ac5de606884eecd8ffb9efcc7c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
