)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~733301","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fussuri~Ic4cdc26ddf1cb0591f0a2218355b7cca7af0b0fb","project":"openstack/openstack-ansible-galera_server","branch":"stable/ussuri","topic":"osa/galera-merge-stable/ussuri","hashtags":[],"change_id":"Ic4cdc26ddf1cb0591f0a2218355b7cca7af0b0fb","subject":"Bump MariaDB version to 10.4","status":"MERGED","created":"2020-06-03 18:30:45.000000000","updated":"2020-06-24 16:12:19.000000000","submitted":"2020-06-24 16:10:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733301-1593015041490-f0b752bd","meta_rev_id":"e45f6f400bc84ae6b044b0e09e309d1d9654cf24","_number":733301,"virtual_id_number":733301,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 16:10:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-06-24 14:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-17 18:19:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-06-24 14:01:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 18:19:04.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-06-24 14:01:39.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-06-24 16:10:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62f09d842f0c3c583d49bc1add7b29beec835468","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-03 18:30:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d7f44ab52fc1a795cf5dbf15dc0cd23333e14b19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 21:16:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/27a65ff48fe24bc38bc3ae962f364823 : SUCCESS in 8m 19s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f743a6f4a9ef422396c74118ddfacd71 : SUCCESS in 27m 55s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e98e0f496ada495792249ca37b354fb3 : SUCCESS in 28m 56s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/d75380b76e764e8da1474cdfe325a627 : FAILURE in 10m 02s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65221bfb0ef249f48362ce5ac9872f77 : SUCCESS in 26m 25s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cacc06319109446e82f567bd578580a8 : FAILURE in 8m 21s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bdcb926bfc074dadbac64e017aa9845c : SUCCESS in 1h 16m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/95c2e99c330e4f0b8ad74df613e4e32b : SUCCESS in 1h 55m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47be47c8f5fc4062b05264f5babbe4b7 : SUCCESS in 1h 26m 23s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/83455cfdbe484ba7809c45b752c1f289 : SUCCESS in 2h 42m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/894af4e58a4c4b17bbc67ae4c2a4519c : SUCCESS in 2h 19m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a1af4a95aaeb41a8bc2e52001696a362 : SUCCESS in 1h 42m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b089d6d51e049b6a3a50358b185d89d : SUCCESS in 1h 30m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d94a234ea12c4657b7ee2d49687d4202 : SUCCESS in 10m 32s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/5e966ac827634139a1515302ad6707d1 : FAILURE in 32m 26s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/62942d53bbe5427ba2bbda29d49530d2 : FAILURE in 13m 05s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/b70f5a49e8de4159993fdc40bafd3f93 : FAILURE in 34m 16s","accounts_in_message":[],"_revision_number":1},{"id":"424b895de905dc9a73ef4d8eee7992c4e315f2f7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-10 08:54:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cfa83d8d7b1be594321c0b1a271d5f29749cda48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 11:39:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c0602ef7333b453e9ea1c14cb1f6dc14 : SUCCESS in 9m 37s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/59e2f4b3a188433786e3029da5974616 : SUCCESS in 19m 19s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/65f62d5f475942e5b74eb7f75cfed449 : SUCCESS in 22m 24s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/7275d69903ce41ecbc4d51f05202d293 : FAILURE in 8m 48s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/601b8df1ed8a467c9c7e1026e86858ab : SUCCESS in 20m 18s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4229268ad024f8f97abfb675cf3157a : FAILURE in 8m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/99aef257020b405aab0b42a6fbb9e2f5 : SUCCESS in 1h 08m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f65952809ff1428d96a400673f54acfe : SUCCESS in 1h 29m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c543b88d579d4f49b47f2bdd642bab94 : SUCCESS in 1h 13m 11s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f141f1b3eae84f29b0eb27d9747d7f04 : SUCCESS in 2h 36m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b55d8a0210a468294956499f83f0302 : SUCCESS in 2h 14m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1007bb2908e946909c575fb36804d0bb : SUCCESS in 1h 10m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/534b6b942a794182bb403bde1ef8c5db : SUCCESS in 1h 03m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83b0b0835a3e4b7ca7428bdc39032a61 : SUCCESS in 7m 41s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/5aea83b514244dd08d68f0acae9d67a3 : SUCCESS in 22m 23s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/9a6c60d636ae4c2c9b279046048036ac : FAILURE in 16m 08s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/988a20047f3840d6b219f80b600e12e6 : SUCCESS in 22m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1a3827ab02a24091734553fd115dd4bc3060d979","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-10 18:22:37.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"17e88b05fe375fd2966ad371d934231de6b50400","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-16 10:20:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa3a0e5ea1d19ed0de0d30bab15cda8aa00b0dda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 13:02:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2555efa9ef334ce1a6a7e4257c91b686 : SUCCESS in 11m 23s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/af98e89a2bfd4deebfcdd66a7528d9c5 : SUCCESS in 20m 28s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2e76cb9ef61c487a983f75850d623854 : RETRY_LIMIT in 8m 26s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/32de4c450f7e487294f19ee473053cfb : FAILURE in 9m 45s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad9e4a32176a4e60aae676ef880a3165 : SUCCESS in 21m 05s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c963f2f87fc4e689d4f97a7805628dc : FAILURE in 9m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cfe6bfe2b1094dbfa52e1d20b31af4d1 : RETRY_LIMIT in 11m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/58881f9ee150460498a031acb98fd073 : RETRY_LIMIT in 16m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e45af36782ef4bb58b21f0f5fd8e85a8 : RETRY_LIMIT in 14m 20s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bd655c660c634fb4a57d1f7c45f40415 : RETRY_LIMIT in 15m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a1411b781b145a0b60a1c7ef2ad5927 : SUCCESS in 1h 52m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c01997985f504e2fb100f02005526234 : RETRY_LIMIT in 10m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15fd7ce71e884306aba200995cec0972 : SUCCESS in 1h 19m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a069d5c1dd0e435c9fb2b8a8f4883057 : SUCCESS in 7m 41s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/7e2fb4e9ce304b30a483341bc7ca7d35 : RETRY_LIMIT in 7m 53s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/89d180ea459042ed88ef7166b47e6955 : FAILURE in 17m 07s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/b738ac4e03724aac8096623d3bbd4dbb : SUCCESS in 20m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a14571b980c44d55eea4f838c5c81d8e2151a565","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-17 10:31:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ba3483756b6fe719e2ad31fed0dc4c2cb8cf149b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 12:56:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4813b7beca8042378a9e49944833c21d : SUCCESS in 8m 08s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2417180c9d8b458d81f0aa1cbfdab321 : SUCCESS in 26m 33s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/725ba6a9ec3a40f6a4e15c282123cbfb : SUCCESS in 18m 26s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/5da6ad8b9a81437e93e63de9ec433768 : FAILURE in 13m 49s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25845c507f4f4a7697d8c9862df678ce : FAILURE in 28m 56s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34ebb3ed77a947e18df4512441339b31 : FAILURE in 14m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fd4124ebdb0f4647b7be888460d97c99 : SUCCESS in 1h 06m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e9671a3d6e624e6e8281c647c79994d2 : SUCCESS in 1h 03m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc936043b92b43d894f15d4dfb0cc5e2 : SUCCESS in 1h 34m 22s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3859583a316b4bc1bd00296759aea5ba : SUCCESS in 2h 20m 58s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07d8722aa6ff4be08a615fdf8eecb994 : SUCCESS in 1h 50m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a5d9d543d371404980c8be1e0c491e48 : SUCCESS in 1h 22m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f4563d9a419446680c1ceb3332502ae : SUCCESS in 47m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/365c63c294744146b4ff0aee518ffc3b : SUCCESS in 5m 13s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/d51305b969e14182b35ce3d3a20e4277 : SUCCESS in 18m 50s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/1577994dae11403f8661456f2061c561 : FAILURE in 17m 58s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/3deaa8aa723044f5b2fe477de2dff089 : SUCCESS in 20m 25s","accounts_in_message":[],"_revision_number":2},{"id":"42cf60ee4f8c3cac6e4ad7496d0fc1d61e42c7a5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-17 15:05:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9ea621cc14197fa1880e44fb6d8fafa32da8fc04","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-17 15:07:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"89223cf40c2585bde02daf19fee81670b1e6f149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 17:58:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b11137f8bb7d49dbb06d42d606bcbccc : SUCCESS in 10m 38s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b0f803ec9ff74429ba21b247b60db4d4 : SUCCESS in 22m 41s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/29315ee95ff749778f2f3c0b2bef489a : SUCCESS in 26m 28s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/ffd25a93295a4e3d82c926d4a77f85cf : FAILURE in 12m 50s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca73fd85b3314e5dafdf0514732aefb9 : SUCCESS in 23m 24s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0baa4b03264842f3a52eb86c8e918177 : FAILURE in 12m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d8356320915b436aa2e658c5c857aaa7 : SUCCESS in 1h 07m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b7ca7754614d451bb24a5c350820348c : SUCCESS in 1h 36m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18466573116645808ee07a536904a386 : SUCCESS in 1h 14m 39s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c688069bb0144cdf9cda30613b54a0ba : SUCCESS in 2h 42m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1203d88c98254496be0ee7e1bb702a80 : SUCCESS in 1h 53m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/565bedc406a143d486c6772ce594cf07 : SUCCESS in 1h 10m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51b53390522f43d1809e9f4697a9d794 : SUCCESS in 1h 01m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f8dbdaab3d1461f94225135492f9a9c : SUCCESS in 12m 18s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/f205600b35614354894a61b2d1a71ff1 : SUCCESS in 25m 44s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/376a06d461a7415caee7233e0817c50a : FAILURE in 17m 26s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/cacdde27517c4162b9b60d5dbddf3dae : SUCCESS in 25m 26s","accounts_in_message":[],"_revision_number":3},{"id":"2b442d0eb399b458542a8ab884821dd944b40dc0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 18:19:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a5739771283d1ad5d282eaea2ed980fe9b76066a","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-24 14:01:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3ce79c3459d4eecf27f342f2c163cb7933d0912f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 14:01:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c6aaf790b018d2f3f6808e1ca2d86f5373ebfd69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 16:10:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a2f63fff0d5626c499be1953e32fafd81c6cf9bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 16:10:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d9ccce249bec457b95dbf76da2499572 : SUCCESS in 11m 34s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/86fa6f63c93541ca941e7d2d4cf89e98 : SUCCESS in 27m 28s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8187e3d487d8471f878c00937f694ff6 : SUCCESS in 29m 10s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/843ec20a435b49ef82a5666150357792 : SUCCESS in 22m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/daa4050b82d74904bf2fd616c223695d : SUCCESS in 1h 05m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/29b422cffbb84c8c862b1cf360ce8486 : SUCCESS in 1h 06m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d37ea4b4736945f796811835c9121a48 : SUCCESS in 1h 03m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7b927d3ed5cd4c98a21937842d81f125 : SUCCESS in 1h 14m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff40cdc4905240e3bde603f0ac136fcb : SUCCESS in 1h 08m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a67778067bd408dbc4f282f69df97b9 : SUCCESS in 8m 43s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/fb230e62136a42afadb4e047d72136b6 : SUCCESS in 25m 28s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/06a98f2d54bb4ad380dd857af7f43493 : SUCCESS in 28m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4beefa26253c409f98196fa0b28754b7 : SUCCESS in 1h 59m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"78efb91f783ad4bf8a6493a03ca2dbf71984072c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 16:12:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f8aebe2463482db72f0822ed5fbf63 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"98056ed243f2309e2b1b30a0fdd0baee82116faf","revisions":{"ac2ec138fd7b20b929943ec734b25c814c10259f":{"kind":"REWORK","_number":1,"created":"2020-06-03 18:30:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/01/733301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/01/733301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/733301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/733301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/733301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/733301/1"}}},"commit":{"parents":[{"commit":"7c165c40cb7e18e5bc17fe9b267d59ef74dc9271","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7c165c40cb7e18e5bc17fe9b267d59ef74dc9271"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-06 18:29:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-06-03 18:30:45.000000000","tz":0},"subject":"Bump MariaDB version to 10.4","message":"Bump MariaDB version to 10.4\n\nPatch aims to use next stable release of MariaDB\n\nRemove no_log from functional tests, the only credentials used here\nare for the purposes of the test and debug output is valuable.\n\nChange-Id: Ic4cdc26ddf1cb0591f0a2218355b7cca7af0b0fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ac2ec138fd7b20b929943ec734b25c814c10259f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ac2ec138fd7b20b929943ec734b25c814c10259f"}]},"branch":"refs/heads/stable/ussuri"},"c03832401b60736c9be40076973685d0024b26e1":{"kind":"REWORK","_number":2,"created":"2020-06-16 10:20:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/01/733301/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/01/733301/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/733301/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/733301/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/733301/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/733301/2"}}},"commit":{"parents":[{"commit":"7c165c40cb7e18e5bc17fe9b267d59ef74dc9271","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7c165c40cb7e18e5bc17fe9b267d59ef74dc9271"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-06 18:29:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-16 10:20:21.000000000","tz":180},"subject":"Bump MariaDB version to 10.4","message":"Bump MariaDB version to 10.4\n\nPatch aims to use next stable release of MariaDB\n\nRemove no_log from functional tests, the only credentials used here\nare for the purposes of the test and debug output is valuable.\n\nChange-Id: Ic4cdc26ddf1cb0591f0a2218355b7cca7af0b0fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c03832401b60736c9be40076973685d0024b26e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c03832401b60736c9be40076973685d0024b26e1"}]},"branch":"refs/heads/stable/ussuri"},"98056ed243f2309e2b1b30a0fdd0baee82116faf":{"kind":"REWORK","_number":3,"created":"2020-06-17 15:07:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/01/733301/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/01/733301/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/733301/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/733301/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/733301/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/733301/3"}}},"commit":{"parents":[{"commit":"32192de0cebe7af69fe9c89ccb424242a6fbf6e9","subject":"Use more reliable option to get cluster ready state and remove ansible_python_interpreter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/32192de0cebe7af69fe9c89ccb424242a6fbf6e9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-06 18:29:03.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-17 15:06:51.000000000","tz":180},"subject":"Bump MariaDB version to 10.4","message":"Bump MariaDB version to 10.4\n\nPatch aims to use next stable release of MariaDB\n\nRemove no_log from functional tests, the only credentials used here\nare for the purposes of the test and debug output is valuable.\n\nChange-Id: Ic4cdc26ddf1cb0591f0a2218355b7cca7af0b0fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/98056ed243f2309e2b1b30a0fdd0baee82116faf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/98056ed243f2309e2b1b30a0fdd0baee82116faf"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
