)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~734888","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I3696b3131de424a4b30ff016da714aef22af4e59","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"mariadb_10.4","hashtags":[],"change_id":"I3696b3131de424a4b30ff016da714aef22af4e59","subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","status":"MERGED","created":"2020-06-10 16:53:16.000000000","updated":"2020-06-15 11:10:03.000000000","submitted":"2020-06-15 11:08:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734888-1592219301185-34ccc9b0","meta_rev_id":"eaf6e0817c5d155042daf733ecb655d2e98f80d6","_number":734888,"virtual_id_number":734888,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-15 11:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2020-06-15 09:23:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-13 19:09:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2020-06-15 09:23:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-12 17:11:10.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-06-13 19:09:07.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-06-15 09:23:01.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-06-15 11:08:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"95a963219f649e7ec7ad6c3d3c16a5aa732cfa1b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-10 16:53:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eef61d58815775fc834938601ec5a27cadffdf1e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-10 16:56:43.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5698e839d7c13449ab02c30a727d1d7183489cbf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-10 18:00:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"96e267efdc5b13ae9818d871b45394997c0cc2e0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-10 18:00:48.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"bb13b021e75f0cddda23d3688908ea0665ca3d1d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-10 18:19:50.000000000","message":"Removed Workflow-1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"70023c3e4ffc2f375de07fce6a995306e3b0f84c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 19:37:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d474e2ea2020401eb17cc70eca1cbba1 : SUCCESS in 8m 09s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/132ebf50797f4cf88975c05f98a2363d : FAILURE in 23m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/dcd5654b7e3d431ab3fbf8308fa098ef : FAILURE in 21m 51s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/40f9cc8a681d414b97533b88e0676b35 : FAILURE in 8m 40s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bfec983ba534943b2be28ab7d41b244 : FAILURE in 20m 33s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a78424b9c3a04b52b73f37772f0ffd0d : FAILURE in 10m 25s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1a789a396aab4b00a4824276d83f9d3f : FAILURE in 22m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9ebe6ae9bdf949c69935269a084cf4b0 : FAILURE in 27m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8abd840bbe5c4930bab799356a20db47 : FAILURE in 22m 35s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/910561264f5a4faeb447cf5c57ea5818 : FAILURE in 1h 29m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b55b45936c6c4c4b900bf6708e5a867a : FAILURE in 1h 03m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5b1fb47201634c34bb5c42ad0a9d1237 : FAILURE in 27m 58s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a41f1af5c41c4f5d9085c143c56a8e6b : FAILURE in 23m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30cbd33d40fb42e1b23b7da48282f777 : SUCCESS in 6m 39s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/b40fb8da9ff041f3a9e39984c5ae4f96 : FAILURE in 27m 30s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/3644bf5949d24b51b5098fa0ebafaefc : FAILURE in 15m 10s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/dabcac0aaf2147d5868c1fc8cd4f8005 : FAILURE in 19m 51s","accounts_in_message":[],"_revision_number":2},{"id":"2e1d0bb49f486893395cc559ba1c6ad05cdd3cef","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-11 06:26:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6603c62c2f188a40a5317182fc73dbafb5f2dc3a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-11 06:29:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9fd4f6d313578182c4af9e2fbad22a4633f3251d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-11 08:20:35.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e023f154354e1f2d92c77b55c4038c5b9b71391c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-11 08:22:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9815819a80676641655ada78590229baf2717fd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-11 08:48:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ffbf568eb24ce06ee104915361fa0d7db477cbd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 10:30:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/017cef33387b43ce90ae28c52cbac51e : FAILURE in 8m 32s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/442ad97e94aa410ba7678d76c8acb3e4 : FAILURE in 19m 12s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/5706825904314b55aad76d267cbfa09b : FAILURE in 23m 55s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/03abfaabdb834b66a277d13a0287e20b : FAILURE in 7m 39s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f067ed43ee34ed0b23b2eb557560d96 : FAILURE in 21m 31s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc0f2d5c730c4a0db965f6c01c7b3828 : FAILURE in 8m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/30a1fef359d94919ada991337bf114f2 : FAILURE in 21m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/720b1b0f77eb420a9fb64adad1ba08c8 : FAILURE in 28m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99fb46b9a3ad4cfc8ba175386390f49b : FAILURE in 23m 01s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2f1a50ac4ad24d3985516d4d08f42f8f : FAILURE in 1h 33m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff3bbf4018e74e198a9b5cb767ca69be : FAILURE in 1h 06m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/26048ad68d0d4c579060b4eaf75e3aab : FAILURE in 32m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7a6a0c77a364aca8f12741afe2bd385 : FAILURE in 19m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1265b458b564c43b5462685f2e5e162 : SUCCESS in 5m 37s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/bedc9efb2f6247ff99fb2cc94bbee73f : FAILURE in 23m 31s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/118b794439fb406382ffc39da46c24f9 : FAILURE in 14m 29s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/18855963a99042f6bc4c9110794be4b5 : FAILURE in 18m 57s","accounts_in_message":[],"_revision_number":4},{"id":"e855b4bc48ec884d709d819961fa9f780086f348","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-11 11:05:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0b5b194cd71f28afeab355531e7d5875deb2b875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 12:40:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9f1d323e20c845b0bc635d081f3db9b7 : SUCCESS in 6m 56s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/80813756f7444a3294c934a42c593e5f : FAILURE in 43m 06s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1ecafbfd82814482b8bdb16db90e4625 : SUCCESS in 20m 35s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/c5aef322ae2941418581df973a72c760 : FAILURE in 7m 19s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff87ca1c521d4ad8b1eed716542a13b6 : SUCCESS in 18m 19s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c738b2ba5a443018bac75257d6ef1c7 : FAILURE in 8m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c99b9861d7c44552af9c71f59f10e2e5 : SUCCESS in 1h 14m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aff8c36fc83b43eb87c6be8ebf59454b : SUCCESS in 1h 23m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d2e7b2b94374e9db37e3fc002649dc5 : SUCCESS in 1h 12m 39s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f67b7f3327694d799e3d78893d7f8daf : FAILURE in 1h 33m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bb714e2445946e695435a2d674b11fc : FAILURE in 1h 05m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/02e6b766e6bf4d3ebd2f345f16a88057 : SUCCESS in 1h 04m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82889b4ae3234ce2a5f81419205806aa : SUCCESS in 54m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5bd50f35e1a4b09852695fdecd339e5 : SUCCESS in 6m 30s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/a8aa4f3a08044e57b31b9294b9f8d938 : SUCCESS in 21m 58s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/a66995300cb94b39830cdab387b445cd : FAILURE in 15m 39s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/f17c3ee49f394fc9898d071d66554097 : SUCCESS in 18m 30s","accounts_in_message":[],"_revision_number":5},{"id":"d2f55941f313d66d2d7231a716356018fded1262","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-11 13:08:16.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a7741ba409ed225f172ed2bf02af22745265dee2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-11 13:08:36.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"bb27b05d2a526089e9509fb8cd57df27dd97eea2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 14:58:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/15017b2fe74a4bb2b1bee1a4eb7b023f : SUCCESS in 11m 36s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9a56ee7e0d52474187d762f5f7ca56e7 : SUCCESS in 23m 38s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/99efeebd5f664d2da8c5cd401361f8f7 : SUCCESS in 25m 11s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/7fc0aa18f57248e7867ba36362e7e341 : FAILURE in 11m 29s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83441349dc4a4fc8b8b70fc0f0c36d41 : SUCCESS in 21m 11s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21f8def387ea4e798f70f87f34668a83 : FAILURE in 7m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fc463f8585574e769fc175164ea15730 : SUCCESS in 1h 46m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3110ba312ad7448697da8194d838d492 : SUCCESS in 1h 40m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d22a2cecb7548599fa1f0902f914713 : SUCCESS in 1h 21m 51s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/80913e6e6df44c1792c9dfb3b3e92fc7 : FAILURE in 1h 47m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cba6287544d4fba8cc9c959bacedd2f : FAILURE in 1h 22m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9fadf92e925f4c01adf23f8719ad69c0 : SUCCESS in 1h 16m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3b6bc0cab2c4bb89fce11e8e412e0c2 : SUCCESS in 1h 04m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da3238009eb340c7bf2dd8f72d8fff39 : SUCCESS in 7m 13s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/779efb6f7ed540928a9e06ec10770533 : SUCCESS in 22m 39s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/149bbb8a87af447f8857ff468eb54bb1 : FAILURE in 14m 49s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/64d3b17e6d3944ebaa1bc9bb2c164dce : FAILURE in 28m 10s","accounts_in_message":[],"_revision_number":6},{"id":"6a4d6af9e400a23940d6bd5f06b042c419f967e1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-11 17:03:42.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"6e1ed43bcb3457349417ea8f2f4c3ca31c6db974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 18:52:26.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/afcaf193534047df8a2c5133d94a9d5f : SUCCESS in 7m 24s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1c155c7f9a36464ea11bb704e2e8f3bd : SUCCESS in 19m 50s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/d8c98cdcb8cf482ca3690360e16ef7a4 : FAILURE in 23m 47s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/2e63fffa352b4116bec3012e6820f33e : FAILURE in 8m 02s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6d97ed4e751433aa8a9b7008908fec3 : SUCCESS in 20m 49s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/772a9d02bc5a4e40bc4f81f587e447bc : FAILURE in 6m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/225d1d2a7594490b9d396659707f882f : SUCCESS in 1h 13m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7b6a3b9d42f641fcbe937a89cad2194b : SUCCESS in 1h 35m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eea6f573c4f94eee81d53b5075ab4ece : SUCCESS in 1h 25m 36s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8a953505f5844ac48ac2d2593323adc8 : FAILURE in 1h 40m 58s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1066a2cbd9d412fadb6260c19326e23 : FAILURE in 1h 18m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f56ff249202d466b935ca5ac5e7f38cc : SUCCESS in 50m 10s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86a67cd985434cc0bbf14469c081076d : SUCCESS in 1h 00m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b464f64722e74f66a7d81846c1f3625a : SUCCESS in 6m 32s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/cf566c71c8db4978a647ada525996795 : FAILURE in 23m 49s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/a6853384927b49f88e8661f4418525c1 : FAILURE in 16m 35s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/2a7b6bf4b87f43b687f4f30ce93a95f6 : SUCCESS in 20m 44s","accounts_in_message":[],"_revision_number":7},{"id":"86e5874e746fc6b3aeaf70de552e8d1e0af915b1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-11 19:05:11.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"b2195614f668ae115416f689daedaf3170000eec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 21:03:52.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d24c110ae41b49579fd474547f358af0 : SUCCESS in 6m 19s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/791e838e8ef442b697031b4d27e92c35 : FAILURE in 9m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0a07cd7179a044a8a14880646d21441e : SUCCESS in 19m 24s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/ccaecd198fff46ae81b102b644d7fca6 : FAILURE in 7m 50s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/880b484d93ae4515922522d9ffebd76a : SUCCESS in 19m 57s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94f6c2347cf547949a489369b69be71c : FAILURE in 10m 11s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/60bff54a48af462aa35819462cff22c2 : SUCCESS in 1h 33m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8736d4d635f24b6198c5465024deedb1 : SUCCESS in 1h 39m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9704fd5007141a9bda626c556f9842b : SUCCESS in 1h 19m 32s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/65a880dd4d7d4261a1b388ef64563af2 : FAILURE in 1h 51m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8c4ea2b26804dc080a48c272aafb868 : FAILURE in 1h 18m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ac708bd009ab417cbeba28509cb6c237 : SUCCESS in 1h 13m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7154d3f3fc99408c8f9e2979f7b100d4 : SUCCESS in 1h 00m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aff3699d145468b8fc0930c8b4014f5 : SUCCESS in 8m 41s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/c277416376bb455683e7af450438bc8e : SUCCESS in 23m 01s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/3cebfb9eac044be0b76c356a122934bb : FAILURE in 18m 41s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/f7c3367c365e40869a6e8b3c9807ebbd : SUCCESS in 18m 43s","accounts_in_message":[],"_revision_number":7},{"id":"52c9897729f321a0422e95d4e52cb752fd4f9af3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-12 11:30:51.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"0232e298cd21303cc667180f70caefadea4cf77e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-12 12:28:47.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"375f61475170cfce3e136dbe0a291558139f484a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-12 13:19:26.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"0bf66c76af51fc638744d9791b17c12055772371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 14:58:00.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/04b2bcc112404d699eff7bd76f816a94 : SUCCESS in 9m 08s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9d7bb1af70d8410baa7e4798591bce80 : SUCCESS in 22m 15s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2f47bd28579749a2a9c598e47b940a8d : SUCCESS in 20m 14s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/8b94bce3d1764065ba287f15b41d5c1f : FAILURE in 11m 53s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97ed3965e4174cdea8e0add9c9e6d680 : SUCCESS in 23m 56s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62fb1562c3d34334a67db300dc398120 : FAILURE in 17m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0dd54c03350749b9989d9c309d7f3302 : SUCCESS in 1h 17m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f750c9a875354696950f3424e750e808 : SUCCESS in 1h 32m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30d6da00e8c34034a9cff485e3528422 : SUCCESS in 1h 21m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1a6f5f1dfd4a4de1a7387d2e102faadb : FAILURE in 1h 20m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bd03010ed54415b822841b110783ea6 : FAILURE in 1h 14m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/21f2986ac47448d1aa3d1067fc49ba05 : SUCCESS in 1h 04m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35161794e5554556ab47d19814cba269 : SUCCESS in 1h 02m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bbb5d81f70448a48ee441ebd55b17e6 : SUCCESS in 5m 16s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/480a7a1d81464cd0acb0337ed509cbff : SUCCESS in 20m 59s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/022d89b40d1c4a19bd3e00d69be19dc0 : FAILURE in 18m 55s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/1a6b2746d746408abc5d2316f21b395c : SUCCESS in 19m 28s","accounts_in_message":[],"_revision_number":9},{"id":"317a3b814c466e3f9ff7d79add948a503f4c9ddb","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-12 17:11:10.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"174ee23e92b5a061b0516e717aa2e08643cbde7b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-12 17:25:02.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"99bd1cb27ed805c018d2e85ffdbcce2519f1738d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-12 18:15:30.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"21c60af0c2e8ce60a5076b3ea818e0674a0a1a90","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-12 18:16:21.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5ba206afc47674e89c5add3c74049068d27a9e71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 19:42:21.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dd63a802c064468cb2e7dc373631dc11 : SUCCESS in 7m 24s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/cd3bd0593c4e4084b169eda3cd261532 : SUCCESS in 19m 27s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c44fbf2f3c09481ca6c953ed48b9e4b7 : SUCCESS in 19m 55s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/dcb04338d5b2446bbe1c8b4993bed43c : FAILURE in 10m 34s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dba18c8c2b2a471b9826e31e45b57b9f : SUCCESS in 19m 10s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1736709495e14a30bedc3b3729d7d1df : FAILURE in 15m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e5c0bef54d934b648e69107260c5b6c9 : SUCCESS in 1h 00m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/99d17754716347ff804831533a093b60 : SUCCESS in 1h 22m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d1f1f589d0943afb2445034001303f5 : SUCCESS in 1h 11m 44s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/61ce88f656094c6e93a87db7ca16fa03 : FAILURE in 1h 22m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea80f083443d4754819d5d31f2bda583 : FAILURE in 1h 12m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aaff0e45410a4d52bb53252891edb6e9 : SUCCESS in 1h 05m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75686a3d26644cb892a0fe793594f100 : SUCCESS in 59m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/362bdb3d8112496180431f099319e831 : SUCCESS in 5m 06s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/05fb067dda9b42be90c8e267cbfef79b : SUCCESS in 18m 39s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/c3900a0cf7b54767a125f8eba066d64b : FAILURE in 15m 38s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/c8251ee80c4445d8b8cf9dfe9a519486 : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":12},{"id":"31e88038b8bda2d0c82b9292be22b5d4db073934","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-13 17:09:52.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"64cd63a8ab021b88c0a0b0e8aa71b18c05dd9028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 19:02:32.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7d1ffdcac4ca47e0bc96be165ec2d009 : SUCCESS in 12m 34s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/57110722c33b47eea2cb872435fc6b15 : SUCCESS in 23m 22s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/d97fe8b002f04e759ce5130b1e85f3ef : SUCCESS in 23m 31s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/63aafc4f5c5b4e8f87d723315585b48e : FAILURE in 11m 27s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e509e3b351284269812ad40d28ce8e7d : SUCCESS in 17m 26s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c48611f30550441884ae5d3da3b1293b : FAILURE in 18m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c396dfd6598742528bd23b63a6655bc2 : SUCCESS in 1h 11m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/077fd52ddab34a1195dfaa38eb0da4ca : SUCCESS in 1h 16m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbe7bbff69e743abb8b5d29b27978ab9 : SUCCESS in 1h 18m 54s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8974943050f341d3870670bd78bbad91 : FAILURE in 1h 49m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c4d78d8bb9c4b39aebf635278ee50f9 : FAILURE in 1h 19m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/022a1e12e0dc4ba59c6cf34df26feddc : SUCCESS in 1h 17m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51f8b831d50e4757b5f69b940fdf0077 : SUCCESS in 53m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cccfb8b4ec641ce898b2fc43bd16bd7 : SUCCESS in 5m 49s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/02f342f5bf514ff5927a04f23416948a : SUCCESS in 28m 33s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/2d3a85fa27944651b15780339393d18d : FAILURE in 14m 20s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/936867329fbc488da217df7d23fff052 : SUCCESS in 23m 54s","accounts_in_message":[],"_revision_number":13},{"id":"79bacd8b4e589522b2da30d5f1a296ab17e50502","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-13 19:09:07.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"f8be39428bf43341409699a494ecc1f71d3e9c3c","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-15 09:23:01.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"a05c87addfba0fe4bfa1ea82d05f7c957f7581dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 09:23:13.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"daf30df808c02c543380a8dfb1c7274c9dcaca64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 11:08:20.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f244295e69444cdeb9376d95c1577912 : SUCCESS in 7m 42s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/85efff53983d4a66863c6a89661600f9 : SUCCESS in 17m 47s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c353fae857184c53a208203a66f5e429 : SUCCESS in 21m 47s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/846af83802384e3eb4c6dc9dab5031b1 : SUCCESS in 19m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fa8289cfe428468891351fd15961db8f : SUCCESS in 1h 04m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c8bb5c980a1a46028bc9086424a22cca : SUCCESS in 1h 39m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8919886048584127bca59739e3ec811d : SUCCESS in 54m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4c0841f0c2db42ed81a2e374911e4d61 : SUCCESS in 1h 07m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27f7339b41604452a535269de598b6bf : SUCCESS in 45m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe8906e165354f179602bda2fcdf2049 : SUCCESS in 5m 36s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/07e1481b46e340eda7a594e956f6b946 : SUCCESS in 24m 03s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/a5f199b2a8a14ae69438495fe5fc2211 : SUCCESS in 18m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bb1be2371404e41bcc6a69b923df480 : FAILURE in 50m 44s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"54cffca917611f3187050c86452436e5de76bfb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 11:08:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"95d4f9be6a021108a05051fb38506a3151d71edd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 11:10:03.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da09e7720bab4e81af90c699496798ad : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"9e5497aa0a01539e38cdf17e32b8c631e41dab06","revisions":{"9bad90a0e2553a176422887148474be214bba181":{"kind":"REWORK","_number":1,"created":"2020-06-10 16:53:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/734888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/734888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/1"}}},"commit":{"parents":[{"commit":"91ce47d1fc270f0a750377a8f6aa4e084ec8cb03","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/91ce47d1fc270f0a750377a8f6aa4e084ec8cb03"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:52:30.000000000","tz":180},"subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","message":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path\n\nAfter bumping version to mariadb 10.4 upgrade path has been broken\nas service name and packages have new naming patterns\nfor a while.\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch.\n\nChange-Id: I3696b3131de424a4b30ff016da714aef22af4e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9bad90a0e2553a176422887148474be214bba181"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9bad90a0e2553a176422887148474be214bba181"}]},"branch":"refs/heads/master"},"6ff52c28d2bfcc0ed43e4b63a7bedfdff735f238":{"kind":"REWORK","_number":2,"created":"2020-06-10 18:00:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/734888/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/734888/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/2"}}},"commit":{"parents":[{"commit":"91ce47d1fc270f0a750377a8f6aa4e084ec8cb03","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/91ce47d1fc270f0a750377a8f6aa4e084ec8cb03"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 18:00:29.000000000","tz":180},"subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","message":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path\n\nAfter bumping version to mariadb 10.4 upgrade path has been broken\nas service name and packages have new naming patterns\nfor a while.\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch.\n\nChange-Id: I3696b3131de424a4b30ff016da714aef22af4e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6ff52c28d2bfcc0ed43e4b63a7bedfdff735f238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6ff52c28d2bfcc0ed43e4b63a7bedfdff735f238"}]},"branch":"refs/heads/master"},"63efbf9a55895e404d3f39dced6b0135c334aa02":{"kind":"REWORK","_number":3,"created":"2020-06-11 08:20:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/734888/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/734888/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/3"}}},"commit":{"parents":[{"commit":"91ce47d1fc270f0a750377a8f6aa4e084ec8cb03","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/91ce47d1fc270f0a750377a8f6aa4e084ec8cb03"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-06-11 08:20:31.000000000","tz":0},"subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","message":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path\n\nAfter bumping version to mariadb 10.4 upgrade path has been broken\nas service name and packages have new naming patterns\nfor a while.\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch.\n\nChange-Id: I3696b3131de424a4b30ff016da714aef22af4e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/63efbf9a55895e404d3f39dced6b0135c334aa02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/63efbf9a55895e404d3f39dced6b0135c334aa02"}]},"branch":"refs/heads/master"},"69eec806f398afca42ec96a2b1e2284e4e04cbf7":{"kind":"REWORK","_number":4,"created":"2020-06-11 08:48:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/734888/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/734888/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/4"}}},"commit":{"parents":[{"commit":"91ce47d1fc270f0a750377a8f6aa4e084ec8cb03","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/91ce47d1fc270f0a750377a8f6aa4e084ec8cb03"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-11 08:48:48.000000000","tz":180},"subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","message":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path\n\nAfter bumping version to mariadb 10.4 upgrade path has been broken\nas service name and packages have new naming patterns\nfor a while.\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch.\n\nChange-Id: I3696b3131de424a4b30ff016da714aef22af4e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/69eec806f398afca42ec96a2b1e2284e4e04cbf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/69eec806f398afca42ec96a2b1e2284e4e04cbf7"}]},"branch":"refs/heads/master"},"7218e9fb8e68c54d82712ad33fdcd68a56737792":{"kind":"REWORK","_number":5,"created":"2020-06-11 11:05:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/734888/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/734888/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/5"}}},"commit":{"parents":[{"commit":"f10fef73fc24e426d9c714b305efca45548697d7","subject":"Use more reliable option to get cluster ready state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f10fef73fc24e426d9c714b305efca45548697d7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-11 11:05:05.000000000","tz":180},"subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","message":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path\n\nAfter bumping version to mariadb 10.4 upgrade path has been broken\nas service name and packages have new naming patterns\nfor a while.\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch.\n\nChange-Id: I3696b3131de424a4b30ff016da714aef22af4e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7218e9fb8e68c54d82712ad33fdcd68a56737792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7218e9fb8e68c54d82712ad33fdcd68a56737792"}]},"branch":"refs/heads/master"},"79e455f30673edf69be0914931edd09d75a8be2d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-06-11 13:08:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/734888/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/734888/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/6"}}},"commit":{"parents":[{"commit":"c88cb0dcc8389454c49cbb5738a9939ef7529cdb","subject":"Use more reliable option to get cluster ready state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c88cb0dcc8389454c49cbb5738a9939ef7529cdb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-06-11 13:08:36.000000000","tz":0},"subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","message":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path\n\nAfter bumping version to mariadb 10.4 upgrade path has been broken\nas service name and packages have new naming patterns\nfor a while.\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch.\n\nChange-Id: I3696b3131de424a4b30ff016da714aef22af4e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/79e455f30673edf69be0914931edd09d75a8be2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/79e455f30673edf69be0914931edd09d75a8be2d"}]},"branch":"refs/heads/master"},"92f1dcdd47529fc2cf6c01ba4936524b7b2a05de":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-06-11 17:03:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/734888/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/734888/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/7"}}},"commit":{"parents":[{"commit":"06412c8b4099971bff707ce4e91bce6087def099","subject":"Use more reliable option to get cluster ready state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/06412c8b4099971bff707ce4e91bce6087def099"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-06-11 17:03:42.000000000","tz":0},"subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","message":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path\n\nAfter bumping version to mariadb 10.4 upgrade path has been broken\nas service name and packages have new naming patterns\nfor a while.\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch.\n\nChange-Id: I3696b3131de424a4b30ff016da714aef22af4e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/92f1dcdd47529fc2cf6c01ba4936524b7b2a05de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/92f1dcdd47529fc2cf6c01ba4936524b7b2a05de"}]},"branch":"refs/heads/master"},"11f9c96a5eba65c7d0451148db4535c2f73ec86e":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-06-12 12:28:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/734888/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/734888/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/8"}}},"commit":{"parents":[{"commit":"9e236344c1e749b3ca4292440e9cf19ac8a8b678","subject":"Use more reliable option to get cluster ready state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9e236344c1e749b3ca4292440e9cf19ac8a8b678"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-06-12 12:28:47.000000000","tz":0},"subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","message":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path\n\nAfter bumping version to mariadb 10.4 upgrade path has been broken\nas service name and packages have new naming patterns\nfor a while.\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch.\n\nChange-Id: I3696b3131de424a4b30ff016da714aef22af4e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/11f9c96a5eba65c7d0451148db4535c2f73ec86e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/11f9c96a5eba65c7d0451148db4535c2f73ec86e"}]},"branch":"refs/heads/master"},"6890e08ef190adbf01d93823664edf10ffa8f9de":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-06-12 13:19:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/734888/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/734888/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/9"}}},"commit":{"parents":[{"commit":"7894a146ed93ad91db12c800cb1b6fd0c43af1cb","subject":"Use more reliable option to get cluster ready state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7894a146ed93ad91db12c800cb1b6fd0c43af1cb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-06-12 13:19:26.000000000","tz":0},"subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","message":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path\n\nAfter bumping version to mariadb 10.4 upgrade path has been broken\nas service name and packages have new naming patterns\nfor a while.\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch.\n\nChange-Id: I3696b3131de424a4b30ff016da714aef22af4e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6890e08ef190adbf01d93823664edf10ffa8f9de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6890e08ef190adbf01d93823664edf10ffa8f9de"}]},"branch":"refs/heads/master"},"94f4fb9c2a994617c13acf9ad8b541e4ad63186a":{"kind":"REWORK","_number":10,"created":"2020-06-12 17:25:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/734888/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/734888/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/10"}}},"commit":{"parents":[{"commit":"7894a146ed93ad91db12c800cb1b6fd0c43af1cb","subject":"Use more reliable option to get cluster ready state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7894a146ed93ad91db12c800cb1b6fd0c43af1cb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-12 17:24:50.000000000","tz":180},"subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","message":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path\n\nAfter bumping version to mariadb 10.4 upgrade path has been broken\nas service name and packages have new naming patterns\nfor a while.\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch.\n\nChange-Id: I3696b3131de424a4b30ff016da714aef22af4e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/94f4fb9c2a994617c13acf9ad8b541e4ad63186a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/94f4fb9c2a994617c13acf9ad8b541e4ad63186a"}]},"branch":"refs/heads/master"},"66d2cb746948f22e5559bb554796b2b1ddb16345":{"kind":"REWORK","_number":11,"created":"2020-06-12 18:15:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/734888/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/734888/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/11"}}},"commit":{"parents":[{"commit":"7894a146ed93ad91db12c800cb1b6fd0c43af1cb","subject":"Use more reliable option to get cluster ready state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7894a146ed93ad91db12c800cb1b6fd0c43af1cb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-12 18:15:16.000000000","tz":180},"subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","message":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path\n\nAfter bumping version to mariadb 10.4 upgrade path has been broken\nas service name and packages have new naming patterns\nfor a while.\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch.\n\nChange-Id: I3696b3131de424a4b30ff016da714aef22af4e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/66d2cb746948f22e5559bb554796b2b1ddb16345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/66d2cb746948f22e5559bb554796b2b1ddb16345"}]},"branch":"refs/heads/master"},"ab7cbd31ec80077139295aa6ce35c1bac25b0941":{"kind":"REWORK","_number":12,"created":"2020-06-12 18:16:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/734888/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/734888/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/12"}}},"commit":{"parents":[{"commit":"7894a146ed93ad91db12c800cb1b6fd0c43af1cb","subject":"Use more reliable option to get cluster ready state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7894a146ed93ad91db12c800cb1b6fd0c43af1cb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-12 18:16:12.000000000","tz":180},"subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","message":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path\n\nAfter bumping version to mariadb 10.4 upgrade path has been broken\nas service name and packages have new naming patterns\nfor a while.\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch.\n\nChange-Id: I3696b3131de424a4b30ff016da714aef22af4e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ab7cbd31ec80077139295aa6ce35c1bac25b0941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ab7cbd31ec80077139295aa6ce35c1bac25b0941"}]},"branch":"refs/heads/master"},"9e5497aa0a01539e38cdf17e32b8c631e41dab06":{"kind":"NO_CHANGE","_number":13,"created":"2020-06-13 17:09:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/734888/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/734888/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/734888/13"}}},"commit":{"parents":[{"commit":"36f35bee263f1d87440a4204d018dda45d19d98e","subject":"Use more reliable option to get cluster ready state and remove ansible_python_interpreter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/36f35bee263f1d87440a4204d018dda45d19d98e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-10 16:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-06-13 17:09:52.000000000","tz":0},"subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","message":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path\n\nAfter bumping version to mariadb 10.4 upgrade path has been broken\nas service name and packages have new naming patterns\nfor a while.\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch.\n\nChange-Id: I3696b3131de424a4b30ff016da714aef22af4e59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9e5497aa0a01539e38cdf17e32b8c631e41dab06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9e5497aa0a01539e38cdf17e32b8c631e41dab06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
