)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~735204","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I6f79c2c40f6708db22f925040c0dba8e0fdfd303","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"interpreter","hashtags":[],"change_id":"I6f79c2c40f6708db22f925040c0dba8e0fdfd303","subject":"remove ansible_python_interpreter","status":"ABANDONED","created":"2020-06-11 17:39:45.000000000","updated":"2020-06-13 17:11:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f20067b873a35fb124b2986533c5fcdef7880302","_number":735204,"virtual_id_number":735204,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-13 13:32:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2020-06-12 13:15:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-12 17:28:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-12 17:28:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-12 13:15:55.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-06-12 17:28:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-06-13 13:32:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"226bd9113aedcf0eda5e0c8a082fd739590ae9d0","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-11 17:39:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e14c9e1f4a21da5a24b4641a8aa9f74716f1b0d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 18:15:41.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"129b4a16cab9e639a72be1796d0a9401daa89d83","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-11 18:17:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8b93707f19b72c943a30a23408409d4ad1e49a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 18:43:20.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"31e15e90e4fb27e402510e1ec709f69ff05e8370","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-11 19:37:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8dc6f36aa5c230adf8f98ea76a941919c19c2518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 21:34:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7fb6823e999c49709dd6a0dc718f3095 : SUCCESS in 9m 27s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/855de276e7424ac2bef05dacc9c66acf : SUCCESS in 28m 15s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/6850dbc22a174117a142644f9c9e3266 : SUCCESS in 21m 08s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/15d8821f0cdf407f8d8cca66a14bd386 : FAILURE in 12m 11s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6455cd873b04d3ab848c1655199df42 : SUCCESS in 26m 06s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8e9d3835fae40698b6139b9ff99d659 : FAILURE in 17m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c5082020fe1c4fd082e15ad5dcdfb858 : SUCCESS in 56m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f0a5057e79c343e99bdef75c17736fe1 : SUCCESS in 1h 54m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9cf943e5d91a4b78bc65549840a0aa86 : SUCCESS in 1h 10m 20s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e216313987774f839be4c4633c71d93a : FAILURE in 1h 22m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f133081867bc49af9c8e7fba34dac81b : FAILURE in 1h 21m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8e598580743e41f9b14dc19d57456dd3 : SUCCESS in 1h 12m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d12ebc572ba14cf29fb029338334dd8e : SUCCESS in 1h 17m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe326bcadc14cbc980734e912bbccb2 : SUCCESS in 8m 08s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/30486ad7bdc1415e9b16d1d35bc6b105 : FAILURE in 23m 27s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/563fd5ecf44e48cb846638e02f409751 : FAILURE in 18m 04s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/0d2a4b860ce642dca4f5685b7b17e395 : SUCCESS in 20m 24s","accounts_in_message":[],"_revision_number":1},{"id":"129dab3743bfdbc526c68f9ba365e1a9a50b670a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-12 12:26:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8b306dbb95779d4e0ba1cbd22e534983009bfdcb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-12 13:15:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4dfe789ac2b8e25c140d5020b9cd801f2ffacfe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 13:59:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/461cd9aabf5344f0a020d5011616b140 : SUCCESS in 5m 50s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/5c9ba49d502241bbbae846fa30461504 : SUCCESS in 18m 33s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ccdfd767e72d433eb5c20cbccbe79f78 : SUCCESS in 18m 27s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/cb600f0b1b1e49e99f72551ecc011059 : FAILURE in 8m 54s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/067d19f192254a1ab64823db98c713a0 : SUCCESS in 18m 43s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4657a80768764848ba6f56f262883d80 : FAILURE in 14m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ced0dea1d66b40b5991a9d4540928a68 : SUCCESS in 1h 04m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/49d5b08a6db44845a5e81c9c2061068f : SUCCESS in 1h 27m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10270cde39e549a2bb3dd94188885a53 : SUCCESS in 1h 10m 20s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/594497f302d2461394ab8d948f65affa : FAILURE in 1h 31m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/685cdff990d647d0a583e58504721078 : FAILURE in 1h 10m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/299bc0de57d94c4a9be3676b2332d4e7 : SUCCESS in 1h 07m 20s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5501be27f8074c65a87805cc6121f988 : SUCCESS in 54m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cceac2c210404529956f6bb4b7444b9e : SUCCESS in 6m 03s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/e7765cbed92c4be1992f4b3124a4deab : SUCCESS in 21m 36s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/ef976708e6a64c4791e0f613db4e0666 : FAILURE in 16m 48s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/8e1e73c1d1c9429b8a1408a8793f0bb6 : SUCCESS in 19m 05s","accounts_in_message":[],"_revision_number":2},{"id":"e25cfee85b8afe339a292fd7c520cc4ec4076cc7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-12 17:28:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1cfd14ce5dd66ba73312ee9bc3c94ea119b9da5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 17:29:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6af03eb70d34d7e51ee2fbef8dac4f17982423d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 19:13:05.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8fdc33ba03f347eeb5ed0a7a21d68208 : SUCCESS in 7m 37s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/0beac4d5472e41fa97d8cb3638c002eb : SUCCESS in 20m 15s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/256c1c036dce4f33840fadd946accab6 : SUCCESS in 23m 06s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3a07653221f44838c917600ef9f4431 : SUCCESS in 20m 53s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d9309a092b374dbcbcee6ba95550786a : SUCCESS in 1h 09m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f5e451739bc444e0958d53a9f36f8c33 : SUCCESS in 1h 42m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9cf6fe8972c44388bc3126e7ab70e8e0 : SUCCESS in 1h 22m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/07c6126785474d579049643bba19f9c8 : SUCCESS in 1h 16m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbe46a3e8d474a0aba9ea1bddaf92bef : SUCCESS in 52m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14eb0e74494b49f38125c4eb8d4a4cf8 : SUCCESS in 5m 05s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/ec3cab8506fb41d8abb97f33f31bef1b : FAILURE in 18m 35s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/36bd64dae374441ea5e479465c4274f2 : FAILURE in 22m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7db4bc195e041a8b6a8d5afc3100fb6 : FAILURE in 1h 06m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc61eb11c81c84d137a11d34d6a0559656c74104","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-12 19:51:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"69ef3541fae43b45875b618ce860048c1e721cc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 20:54:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a4816e4bf121497385fb68b9eca435d0 : SUCCESS in 6m 08s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/76ad45ab4b614fe288051ac932740519 : SUCCESS in 16m 56s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/de50fdd5783d4f32b8c070fcd0dd1b39 : SUCCESS in 15m 31s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/efe2b6d351f546da8b70ab9ebbe0711b : FAILURE in 7m 40s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d8e067c694d4f5090c22ae68a4a4e61 : SUCCESS in 17m 51s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/781a6c58f5514930a2069f983c35f122 : FAILURE in 13m 43s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/eac63e9433434e05a3bf064a5768671f : SUCCESS in 53m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b51f5183eca74b84aa74d8a3543dbd01 : SUCCESS in 1h 00m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37b31ec5e00e44938e1b54a5b524e2eb : SUCCESS in 53m 18s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e4285e7794b14deb96c0a0c0e9b3134d : FAILURE in 59m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81aed8b7ffdf4fad8ea39324322155a1 : FAILURE in 50m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fe19cf6860114b649761f16d1dd546c2 : SUCCESS in 46m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e18e34183bcc4114befffced3a45a0c1 : SUCCESS in 41m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7324c10b4f24935a523c1c9c369ed50 : SUCCESS in 4m 26s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/a01c7bab8d4745a892e152c378094a2b : FAILURE in 17m 47s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/33edeca191b8451286dcdcf6b04b5dae : FAILURE in 11m 52s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/db5dc589711b467d86b56da0a1f5a2b6 : SUCCESS in 18m 00s","accounts_in_message":[],"_revision_number":2},{"id":"2af5b788343cba325094f20eb6bda682c5c73697","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-12 21:37:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"380ff879332b4230d5ddbdca725f915b369d774e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 23:21:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b4f4f5fd8631484ba3d0b6bfb6cb7957 : SUCCESS in 6m 21s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/eff47812524a4d939eb0dbc99433959b : SUCCESS in 16m 18s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0749dfc8cd124ed1a372c706ac760795 : FAILURE in 21m 19s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/9d6899da28ce4dde9955ad4e12ef02b6 : FAILURE in 7m 55s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a19028be7e974f33ae564ecf30c76a73 : SUCCESS in 18m 16s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea25603c6b234d19bc4e59a2b424bde8 : FAILURE in 15m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1051a4b6ebbf4ac39c6f1aaf518f5595 : SUCCESS in 55m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/491fe9b9389740a89b1775d7fa23529d : SUCCESS in 1h 15m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fe70c5a510c42ec952b87fd83ecd998 : SUCCESS in 1h 12m 35s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c010b249b1514c1dbb631a0c2eb0770e : FAILURE in 1h 22m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53d67a715a864c8aa6cb320272b406d9 : FAILURE in 1h 02m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/06aab07c38b04240b097d1e938b5f14b : SUCCESS in 1h 05m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33e91e83ddbc413cb87791aa8307aeb4 : SUCCESS in 54m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07648feaaff344118bde6aa16b7fa55c : SUCCESS in 5m 02s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/520c7caba2ca4e15aa9f6ed0a1f58fb6 : TIMED_OUT in 1h 38m 12s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/4e560238e28a4b658ce5d6df11ec4384 : FAILURE in 15m 24s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/6a46d6cb2b6d4ed5942e929e352b2b7e : SUCCESS in 18m 12s","accounts_in_message":[],"_revision_number":2},{"id":"c7b9814fda05cb368185b0213703337c01aef2ea","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-13 06:57:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b3641c0d037e545cf60e4a963f75ced9b77817c5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-13 07:44:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"383174aec922913fc40b572f6796afdb694384e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 08:39:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e8fa44bcbafc44b5a1dbaf37905df44c : SUCCESS in 7m 41s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b590632a969e49a1b14948b3a22ef2db : SUCCESS in 23m 27s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/903205aa4ad542a58f8d0b036081012d : SUCCESS in 21m 34s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/94c57fb6914a4647922f1e8e6caa2d75 : FAILURE in 10m 05s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b78838a55d8f4d3d9dd875ef3ac610d9 : SUCCESS in 20m 53s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b18789b46b714a009bf72dd426fd7a8a : FAILURE in 17m 24s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e017369e501946029b222852d5b59437 : SUCCESS in 1h 17m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/acfc38c518054307860d0158246079fa : SUCCESS in 1h 25m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eda095afb516452184628db3e2c2837f : SUCCESS in 1h 13m 32s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0fc015499ab9461380b396c6ccb58502 : FAILURE in 1h 29m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aacec0c452d84e6cbc5627bd1ef1c8ab : FAILURE in 59m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/adf337308d65468db289dba0fa64b510 : SUCCESS in 1h 39m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/969e30d0f56945d4b53cbfcf83a44bef : SUCCESS in 59m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6aed608613547c4af795a64613dca0b : SUCCESS in 6m 48s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/a38d7804329446118ce8f014c2f559b2 : SUCCESS in 23m 18s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/5350a62ce78041e19449cb8223e0b4d7 : FAILURE in 16m 00s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/2a274feafe754e3989d1ed1304cc18d6 : SUCCESS in 20m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e58650695b3017cea55df08a770d80a94cf342d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 08:39:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5249ad8e6784a94c61a6c9087ffe3cc8fbeec890","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 11:43:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f6bd2950c017401b8bea5a403e2de14f : SUCCESS in 5m 52s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b298eeda20a847429b1e047d1ffb6225 : SUCCESS in 26m 33s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2fde17fe7e5f4661a6d948e23d4717d2 : FAILURE in 22m 50s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a38c838d26f4d8cad276ff57f45971b : SUCCESS in 17m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/54c521fae62e49e89e3a81bce5a7aa71 : SUCCESS in 1h 30m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8f76cb7d8e5748838ca8bab7b8c59bf7 : SUCCESS in 1h 22m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/412c8bda3fef407fa7fcfb838e78a023 : TIMED_OUT in 3h 02m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/21e8fad184db4201b5c3d024439bd6f8 : SUCCESS in 1h 15m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acb29fecfe844a879f041bc1b4cd8483 : SUCCESS in 57m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec5f9e5d5cd0443d9108044204f25381 : SUCCESS in 5m 24s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/5dbd877656a7451293a042bc2a85c4ab : SUCCESS in 28m 42s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/8168fe8f050f48c687eff7811cc50578 : SUCCESS in 18m 54s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bde7d72abcaf42a881916f490ed8e589 : FAILURE in 1h 05m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff4786a8edb81751ca6163327df221f44dbee148","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-13 12:07:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9b9771ea4bbf8f39bc79cc11605b7314b05b713f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 13:32:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/545150b5fc3346b7b49b25abf996ecef : SUCCESS in 9m 32s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/fbf475ba02d549f2920c5689d8d89b8e : SUCCESS in 21m 02s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ac486ed90a4349069efb9ef25c6bd178 : SUCCESS in 22m 46s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/6fdbce5dcd0f490dae95cac39f55b122 : FAILURE in 8m 16s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8690d3482693451e8b93c9ba17c25c38 : SUCCESS in 19m 26s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64d481549fb6434098fe2df1a3bf4f1e : FAILURE in 14m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1a1faaba22ea4ff987c717c1e31343ba : SUCCESS in 56m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/643dc458688944ea93fd0d7cb810e7d2 : SUCCESS in 1h 23m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76e30fc2ac8a4d6ab962a08ae276055d : SUCCESS in 1h 15m 12s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/234c4cd2f50c468abee8b646d8a862f6 : FAILURE in 1h 22m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f5dc70c995f4d8b82ff036395072df8 : FAILURE in 1h 12m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/687362d9f2044332a2d98fdfd1434fe4 : SUCCESS in 1h 03m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94742c427c574faca7208c557abb77bd : SUCCESS in 49m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6566b9561fe0415195995ab6d37b6fdd : SUCCESS in 4m 52s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/7a97cd0eb89049ffbb6e22c7aaf15175 : FAILURE in 23m 35s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/65cd2a46aa74481ea2bccaf84ea4caa8 : FAILURE in 16m 27s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/6cce2a07ffdf4e03a7b93c2377d53c51 : SUCCESS in 21m 25s","accounts_in_message":[],"_revision_number":2},{"id":"9b4958dc6511512bb20c983e7a84bb020fe7fc56","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-13 14:28:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5d91a920ae2ee961b18da124cc605840ccae4db1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 16:07:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d49b097efc5546dc91ca758ad32d98ab : SUCCESS in 7m 37s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/19af2544f340416a817cc4eac2cfb4ed : FAILURE in 23m 21s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/f82232f5b844407abe53c37900e24f78 : FAILURE in 21m 49s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/d50cb5887095415582e7df1850a4f0f4 : FAILURE in 10m 31s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c7b5628f6fa4b75b6cc47bf1af7f0a6 : SUCCESS in 21m 18s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45f76c0b562e4bb7ad3b0d9b8645c956 : FAILURE in 16m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/17c210126b39417a8e2ac82c22c4e026 : SUCCESS in 1h 25m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/feb615951f294541971ba00521858f32 : SUCCESS in 1h 36m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ce5af0895ab43999cf455bb638d0bb1 : SUCCESS in 52m 20s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4351d601f0254e79aada825065fd4333 : FAILURE in 1h 37m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28c7ff52b7454a728454d7a039d4cccd : FAILURE in 1h 10m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f5a4bd5fa193425aa50b004969ed164a : SUCCESS in 1h 12m 28s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3049f9dfc5ab4d97a3ad47c136f7d3c4 : SUCCESS in 1h 07m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ca2e91b410468fb6f78871fd44d77e : SUCCESS in 5m 56s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/a49a1ac1181b48a68224b5f94e194717 : SUCCESS in 26m 57s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/f5bb011938f245f0a592530dc556a21e : FAILURE in 16m 20s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/b26a88326ff74efbaf09f47112bcf53e : SUCCESS in 21m 30s","accounts_in_message":[],"_revision_number":2},{"id":"522a3b862bbfd59b80a4e78a4d6c168d943a5740","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-13 16:17:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"244bea80026eda6168d9083b0aaaa1a92cd36221","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-13 17:11:40.000000000","message":"Abandoned\n\nin favor of https://review.opendev.org/#/c/735424/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f5ed9975e92d468d40fa45428384964d2a7fe48b","revisions":{"04c757886afbe1c4c2a351a900ebf1f702fd4d59":{"kind":"REWORK","_number":1,"created":"2020-06-11 17:39:45.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/04/735204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/04/735204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/735204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/735204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/735204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/735204/1"}}},"commit":{"parents":[{"commit":"91ce47d1fc270f0a750377a8f6aa4e084ec8cb03","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/91ce47d1fc270f0a750377a8f6aa4e084ec8cb03"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 17:37:36.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 17:37:36.000000000","tz":-240},"subject":"remove ansible_python_interpreter","message":"remove ansible_python_interpreter\n\nDepends-On: https://review.opendev.org/734676/\nChange-Id: I6f79c2c40f6708db22f925040c0dba8e0fdfd303\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/04c757886afbe1c4c2a351a900ebf1f702fd4d59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/04c757886afbe1c4c2a351a900ebf1f702fd4d59"}]},"branch":"refs/heads/master"},"f5ed9975e92d468d40fa45428384964d2a7fe48b":{"kind":"REWORK","_number":2,"created":"2020-06-12 12:26:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/735204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/04/735204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/735204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/735204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/735204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/735204/2"}}},"commit":{"parents":[{"commit":"91ce47d1fc270f0a750377a8f6aa4e084ec8cb03","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/91ce47d1fc270f0a750377a8f6aa4e084ec8cb03"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 17:37:36.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-12 12:26:15.000000000","tz":180},"subject":"remove ansible_python_interpreter","message":"remove ansible_python_interpreter\n\nWe also temporary disable ubuntu upgrade jobs to be able\nto merge patch and follow up one, which should resolve\ncheck failure.\n\nDepends-On: https://review.opendev.org/734676/\nChange-Id: I6f79c2c40f6708db22f925040c0dba8e0fdfd303\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f5ed9975e92d468d40fa45428384964d2a7fe48b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f5ed9975e92d468d40fa45428384964d2a7fe48b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
