)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~735228","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I50f2d2dc39dc6e73b532cb6b7e94d0d5195fcf90","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"interpreter","hashtags":[],"change_id":"I50f2d2dc39dc6e73b532cb6b7e94d0d5195fcf90","subject":"remove ansible_python_interpreter","status":"ABANDONED","created":"2020-06-11 19:59:01.000000000","updated":"2020-06-12 12:24:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"521f5b5a241e91eb060e31a72e41d18d0332572d","_number":735228,"virtual_id_number":735228,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-11 21:57:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-11 21:57:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43632c0ac1d036f14e6f0c281824f23d33d4c29d","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-11 19:59:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e979318bcbddd005d5c46703fb59b75f97978b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 21:57:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7d6fccf58ff6477690fde7de8114c3ec : SUCCESS in 10m 16s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/186e2796a06a4c76940552e90ead88b3 : SUCCESS in 22m 45s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/3fd848ae218549e88a43797d90ead80b : SUCCESS in 27m 35s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/4b627deadc83489ebf97939f12ef7124 : FAILURE in 14m 12s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66e4247bff664dac800ec1f55bab7bf7 : SUCCESS in 24m 13s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3de95fe0799d46fdacd8cce634a84dc9 : FAILURE in 18m 47s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/afc47b8e902f42a18f77611faf0011dc : SUCCESS in 1h 15m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7026aab4b5224ebdab6a218b37d89f5e : SUCCESS in 1h 35m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9062664ee4c840f8a2ff2f90304fb744 : SUCCESS in 1h 14m 00s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bb57c5cae43c4bb4a13ded5ac0eb9144 : FAILURE in 1h 38m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56796fa92edd457f8e21924802c1e290 : FAILURE in 1h 32m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ed16d95aa06545e0854bb9a65ecb7d65 : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/967bfec6004040688ff4947ad849978d : SUCCESS in 1h 08m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac8e74ca7704fec966be0ed30f472f2 : SUCCESS in 7m 41s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/5312ce55a7704737a9fd5aad8616fc35 : SUCCESS in 24m 21s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/1b8a082c49334b759e9275a63b83d229 : FAILURE in 18m 26s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/54839d1ff62641ecb4304f0b0ad73a95 : SUCCESS in 21m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0cc40e126b3e8a0d306fce0393b00613f52d1fbf","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-12 12:24:01.000000000","message":"Abandoned\n\nduplicate..","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"363be05f8751bb82c58c95d68631e2dce06647b6","revisions":{"363be05f8751bb82c58c95d68631e2dce06647b6":{"kind":"REWORK","_number":1,"created":"2020-06-11 19:59:01.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/28/735228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/28/735228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/28/735228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/28/735228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/28/735228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/28/735228/1"}}},"commit":{"parents":[{"commit":"91ce47d1fc270f0a750377a8f6aa4e084ec8cb03","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/91ce47d1fc270f0a750377a8f6aa4e084ec8cb03"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 17:51:58.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 19:59:01.000000000","tz":-240},"subject":"remove ansible_python_interpreter","message":"remove ansible_python_interpreter\n\nansible_python_interpreter is now set to auto in the tests repo.\nit doesn\u0027t need to be overwritten in the test anymore.\n\nDepends-On: https://review.opendev.org/734676/\nChange-Id: I50f2d2dc39dc6e73b532cb6b7e94d0d5195fcf90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/363be05f8751bb82c58c95d68631e2dce06647b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/363be05f8751bb82c58c95d68631e2dce06647b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
