)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~737781","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fussuri~I4bb36e779002a6ce7a22fe308195de3b10aa1aba","project":"openstack/openstack-ansible-galera_server","branch":"stable/ussuri","topic":"bind-to-mgmt-stable/ussuri","hashtags":[],"change_id":"I4bb36e779002a6ce7a22fe308195de3b10aa1aba","subject":"Add defaults for galera server and wrep cluster bind addresses","status":"MERGED","created":"2020-06-24 13:32:31.000000000","updated":"2020-06-26 05:25:35.000000000","submitted":"2020-06-26 00:41:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737781-1593132069676-39350062","meta_rev_id":"4819ab02b3024a0cd1a9d6858cc1fe7aba76964f","_number":737781,"virtual_id_number":737781,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 00:41:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-06-25 13:21:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-24 16:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-06-25 13:21:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 13:21:28.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-06-26 00:41:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7bcf193e1957c9851792a173d059e3efc7c29bf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 13:32:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"03a401f0219147646292252c7982e3df523c4953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 16:27:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7d0db4f5c1664328a4e6aa25df47a831 : SUCCESS in 7m 10s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9b4b78c5a0914f8c875742b7ab18adc4 : SUCCESS in 25m 20s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/44599abb051a4b3cb7377cb2d65b7f4b : SUCCESS in 22m 05s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/f306a0f4ff614f959ccc0d73914aea4d : FAILURE in 9m 06s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ada36eb904a04ada9e64f789da802cb5 : SUCCESS in 21m 38s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/140943e1fb674a998ed23a487068ba09 : FAILURE in 9m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/225fd3fdf6e7410681fc09fc4d16c826 : SUCCESS in 1h 12m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8e66c15417cd4f0c979d95f56bda82c5 : SUCCESS in 1h 30m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53edfcc7aa9646a4ae618e73eb196c8d : SUCCESS in 1h 21m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9555e762b2c04ec1a9f18f3f26c01106 : SUCCESS in 2h 30m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a2f0ba02fbc4b6e8812da4a4d11b6e8 : SUCCESS in 1h 54m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1654c890264f4e6cbedc9b11d52ae54c : SUCCESS in 1h 07m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c7d4182ce8849418d8e97064beb6b28 : SUCCESS in 51m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d523a870d61149a1aab87e9815473dec : SUCCESS in 7m 58s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/6dcf19e7c24a4a9bbd7c80f30dc63f03 : SUCCESS in 27m 17s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/f083d92831354351ab5f19b1323d5064 : FAILURE in 24m 34s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/b3ab20bb94174cfa98bcc4b82f33d6bd : SUCCESS in 22m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1168118f07057861c07cb88282112e5e0a04e75e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 16:53:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d6b510f2106dce0704972e41abb45af85151793d","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-25 13:21:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea6028555848f677b8c606893ab051a6a1604ee0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 13:23:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2405c528a0db2ba81c22b0b8164edc038aec10fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 15:00:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0ea32a942f3e473daf540a5a44b570db : SUCCESS in 7m 48s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/5499ea65763a4ed5a63ea211254033a8 : SUCCESS in 19m 44s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b8cfa52c7ccb4d99be22cc7a7f612362 : SUCCESS in 22m 49s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cdeea0432bc343049cfa24ba8471344a : SUCCESS in 21m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dfc01e883ef843ad90d69b84eb00ea64 : SUCCESS in 1h 10m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d3c5f89b65b640df8da752f087f56f34 : SUCCESS in 1h 30m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e81469b8dd245078a4dc8447235e154 : SUCCESS in 1h 14m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ecc4cd8c7d0f42f49f431bfe04ad51aa : SUCCESS in 1h 06m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/856e1c535cb642e9b24510df5491699f : SUCCESS in 1h 02m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7b5bda2cb2a4f6d90eb41b371446e5c : SUCCESS in 6m 18s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/d5248c7885094631ad6368439e7bf1bd : SUCCESS in 18m 09s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/a82d0161d90146348fc98ef3da40fec1 : TIMED_OUT in 1h 35m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1515b7ca669f46359b257d28ab8a6027 : SUCCESS in 1h 24m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7312fecc3e3a75cd70d9c32ace7f2ca6582e897a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-25 15:02:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"70d9d0808acc1892db98c541003b88fb83550f93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 17:43:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7af46d21c2a9412090ee3f86d5ad5a5d : SUCCESS in 6m 48s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7de762fbc8354913a33b7d756577701f : SUCCESS in 21m 37s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/5945b93e91b5420c9e8f828795760a30 : SUCCESS in 21m 07s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/69518426443941f6875f8928bc49d6d9 : FAILURE in 10m 46s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbe0788d0f1d43c4bc243cc236f33f4e : SUCCESS in 20m 39s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4163b1001f7043e1a7a8bfab139236e1 : FAILURE in 16m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f66ae0928084470780f48480f2b2e22d : SUCCESS in 1h 19m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dc952d05dba344dab0a7861c6c6b9a5e : SUCCESS in 1h 33m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad9ed65f2c994f8397624c163e365b86 : SUCCESS in 1h 17m 48s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c944f1b63af54a47923fb759822dc72a : SUCCESS in 2h 26m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40b7c92661f645ce9f3d4adf65d5b1c8 : SUCCESS in 1h 55m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9ad08bcdc43842a99cf9feaef2455a36 : SUCCESS in 1h 08m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87a0fadad333413faa5b7f72fb422b71 : SUCCESS in 55m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c0731544d06414c86c89e13359a5e14 : SUCCESS in 7m 55s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/0887e6e85f084a7fb88dc0088c64e9ab : SUCCESS in 24m 27s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/187c75956ad14fc7a403d53830dc8768 : FAILURE in 16m 42s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/d4754b6731664752a82ead4a1c3b7de5 : SUCCESS in 22m 02s","accounts_in_message":[],"_revision_number":1},{"id":"50ede31784b54353e4723b6069391d1bb05a7e70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 17:43:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19429e9ffbea04122c544d5ca99b9c2055f43ddf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 19:49:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/31b20a85f4854a5283c650cc465d7a3b : SUCCESS in 8m 02s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f595f36622f54e67925fd28ebd55cb12 : SUCCESS in 21m 18s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/4589b8b7351c4b8ba00bb25bf896afb9 : SUCCESS in 23m 29s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcc98fdfc01f4ff3838bf9fe04817e68 : SUCCESS in 18m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fba52660af74432fa4d8e20a81560836 : SUCCESS in 52m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e38dc32335ca48dfa366d975a7776725 : SUCCESS in 1h 28m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55b45c5cd2a04781900b4bbd0633d1f5 : SUCCESS in 1h 19m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dcc8a7b43bbb415b916f6fdf269c9af1 : FAILURE in 1h 09m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b58d191f71b42e2ac6b2255592afaed : SUCCESS in 1h 00m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23613b663f344fdc87ec9ea353918000 : SUCCESS in 10m 36s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/20ea1fea1505404d9170e4266906cd1f : SUCCESS in 23m 49s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/3a8862e5141c4df297990cfdc549e274 : SUCCESS in 20m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a037af0edecb4e6bb1ffb9fea847c5ce : SUCCESS in 1h 58m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74f325a3c6c53336d533ef3a4361e506cac3148a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-25 19:56:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ca2a2a7100208673a02f2c76a8ee89a6528e051d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 22:35:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9df401ff5d124904832f9075f8486b02 : SUCCESS in 12m 25s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/653f9aea454f492c8ca5a1213560ea0f : SUCCESS in 18m 40s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/46f6e2ad7de040819438a068673bc049 : SUCCESS in 27m 51s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/37a97d57c442400e81afb0ae09e45376 : FAILURE in 14m 06s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1e016e92f094dbf921dbfd5fb824be8 : SUCCESS in 23m 31s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88891704289240fa965735073eb564ce : SUCCESS in 20m 46s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/80d583ad6ff348a5964fe91819e4e5c6 : SUCCESS in 1h 05m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e3cfe09657f640a19bc534944b0b222b : SUCCESS in 1h 28m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53324c4760184fa7b177e179f1ab5590 : SUCCESS in 1h 13m 09s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0004ec36664b46a88ce1b8c3b2b0505c : SUCCESS in 2h 32m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0355d463361f4cc4b629cea876d3a26c : SUCCESS in 1h 49m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b2c92ed9e9844dc5aa24bbebbb158bfa : SUCCESS in 1h 02m 00s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fcea41495c34f82af5ac93c49b8ae93 : SUCCESS in 1h 00m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be41cc5d28dd465a89d716d91121ff2d : SUCCESS in 10m 59s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/c218f70c2aab47c7bcbac539554e7ba2 : SUCCESS in 21m 34s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/9ae3ccf2616345ffb9eb915795da105c : FAILURE in 16m 19s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/338770e9642f4bf8b09793bbdef0b29d : SUCCESS in 21m 27s","accounts_in_message":[],"_revision_number":1},{"id":"950103b36d5c7f0c353bb03f4ba08e46ae28b08f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 22:35:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b03bd7227a2210835ff1a7732d2df070c450b99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 00:41:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"36be5a4af98dbe93fac5f3f975fe17409c4fdfea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 00:41:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0734a4d07f5e49b9800a0f025ddad800 : SUCCESS in 6m 42s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/052d3ac7d7c74ed195ec23486ffb33bb : SUCCESS in 20m 32s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/49617dccdc9842eba4fc36384e6d1222 : SUCCESS in 23m 40s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1541fb120185489aa9404c4a41b5eda4 : SUCCESS in 22m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3216e16b46e84e8cb6d63a61c03cb7a9 : SUCCESS in 1h 13m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2c94a0c67ad146caad528510f4ed4310 : SUCCESS in 1h 25m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4373ad59070440b291715b4230c26310 : SUCCESS in 1h 14m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d4e49114fd474107bf2f82a0e081a92b : SUCCESS in 1h 02m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5a59133073c4bcd80edb05c014eebfa : SUCCESS in 54m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b289fe84f414ff89daa1c72a54626c9 : SUCCESS in 6m 15s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/a5edc5ae825a4631a4fc52d954331ae6 : SUCCESS in 16m 43s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/2ab35e393de84c71931dd25d045b346a : SUCCESS in 18m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f49e187a393434aadb7580a7d29af3f : SUCCESS in 1h 57m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd5778299c956f40ce5515356d9b918735f37c78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 00:42:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b7ee1481d245509ba325b42d256864 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6baee05ffb30fcc6a372d485850c6168a1a13680","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-06-26 05:25:35.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis ternary looks weird or?","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4735f98f391a9b1943a1320109ffbb5074e83d9c","revisions":{"4735f98f391a9b1943a1320109ffbb5074e83d9c":{"kind":"REWORK","_number":1,"created":"2020-06-24 13:32:31.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/81/737781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/81/737781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/737781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/737781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/737781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/737781/1"}}},"commit":{"parents":[{"commit":"32192de0cebe7af69fe9c89ccb424242a6fbf6e9","subject":"Use more reliable option to get cluster ready state and remove ansible_python_interpreter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/32192de0cebe7af69fe9c89ccb424242a6fbf6e9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-11 22:13:08.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-24 13:32:31.000000000","tz":0},"subject":"Add defaults for galera server and wrep cluster bind addresses","message":"Add defaults for galera server and wrep cluster bind addresses\n\nThis is defaulted and can be overridden in a deployment\nwhen the correct managment network address is known\n\nDepends-On: https://review.opendev.org/734672\nDepends-On: https://review.opendev.org/735289\nChange-Id: I4bb36e779002a6ce7a22fe308195de3b10aa1aba\n(cherry picked from commit 955ee8da9bbdaaa5f898e00e324f657157ddbd74)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4735f98f391a9b1943a1320109ffbb5074e83d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4735f98f391a9b1943a1320109ffbb5074e83d9c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
