)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~742104","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ifc4ead03ac914275a302620a67a6f9cdb7275116","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"Ifc4ead03ac914275a302620a67a6f9cdb7275116","subject":"Remove Centos-7 support","status":"ABANDONED","created":"2020-07-21 08:52:38.000000000","updated":"2022-01-04 10:57:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4a83f942207998cba379ca53d6057e5a36d1e9a9","_number":742104,"virtual_id_number":742104,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-21 10:29:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-05 19:37:09.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 10:29:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5a9946a7a7ca95fe1b0aab0343a45af2ee340ea","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-21 08:52:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41b6aac6a6821d7ed03e95f40f09fd090bd614ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 11:14:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/901f3b8b60004daa86de09ccddf0ab20 : SUCCESS in 6m 47s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7c58f28c9b494960980f35a810c24228 : SUCCESS in 17m 32s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e2c505bcca8e4909a2eefa642e4bed21 : FAILURE in 19m 28s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/a6bbfd87b3ad4bbba8ff1f6f745546a4 : FAILURE in 10m 26s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b7afd80747440f9a3754ca508ce01f1 : SUCCESS in 21m 32s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/942b64c46893452892716b1d7d169a82 : SUCCESS in 21m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f773eec6bfe74e649da02c012c23529d : SUCCESS in 1h 10m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b7b0c784f7d84165b3b49b2f36765368 : FAILURE in 26m 09s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b3256db1848f4cab9bc0c69aca4ecfd0 : SUCCESS in 57m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05e3561865524e189621b69e8f4c9701 : SUCCESS in 1h 23m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ce797dc5e6f47d9afb7ff0d1ba5faa6 : SUCCESS in 1h 22m 00s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a93b5645fdba40beb7bc80a6e94c18d5 : SUCCESS in 2h 15m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/095a1ed6abb24761ab2ad441100384ce : SUCCESS in 1h 49m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c98c04721ed7485ea2fa13841d21aeac : FAILURE in 21m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4c20195ccfd24493b27028aa38462bfa : SUCCESS in 1h 00m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f3b4abc160846cebd591d4ed435518e : SUCCESS in 1h 03m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13cdaaa6a8f647e7a953841a8d4c0656 : SUCCESS in 8m 08s\n- openstack-ansible-functional-centos-8-ssl https://zuul.opendev.org/t/openstack/build/aa8ef37cbf134c8aad2a587875c8a54a : FAILURE in 9m 42s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/a3e598ce56de4780b4fda11a5a11908d : SUCCESS in 22m 41s\n- openstack-ansible-functional-ubuntu-focal-ssl https://zuul.opendev.org/t/openstack/build/aa8b5ee3d6034bfa9b556595d0989d4f : SUCCESS in 25m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ced5e84d6ad37fc42b43bf88be694a6c32217de","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-21 12:38:51.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a0a92bebbc3901bf324ab7ced7bdcc7b50cb4f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 15:48:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2ccf1a918d5e48e3ad6eafba72316430 : SUCCESS in 10m 06s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/5ee3ee7eb1d84e3da7b190b04a720158 : SUCCESS in 18m 57s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/6416596b1bd5442fbb8ecad210a73987 : SUCCESS in 41m 17s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43bd118e3deb4d668603519dfce6dfe0 : SUCCESS in 19m 10s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5236f5181a3a4203a3bfbbd4dab2b300 : SUCCESS in 20m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0f3c051af66c4753a15e9e719b4b74ee : SUCCESS in 1h 04m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/53430fc9fcd04c00a8cf3fc63b571a5e : FAILURE in 26m 41s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1db7acf916cc4a549426d20447f5317e : SUCCESS in 1h 18m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e19a47216f1f479ead337ad8efd3f4ef : SUCCESS in 1h 06m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcb1962743864c05be38b79120b33d0f : SUCCESS in 1h 20m 56s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/61ff5a0d2f39469cb009d090846cca07 : SUCCESS in 2h 51m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba920aae326a4b6caf45f4c15db8f5f6 : SUCCESS in 1h 45m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c6d97e54fd074501bf43c3c98ef1d70e : FAILURE in 19m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b51679d66dd34b089efe953a48e6a160 : SUCCESS in 1h 08m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f5c6b1a5b384e71ab07003e745ae222 : SUCCESS in 48m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5917ac2b1024f96b618b19d672ba662 : SUCCESS in 5m 34s\n- openstack-ansible-functional-centos-8-ssl https://zuul.opendev.org/t/openstack/build/b6e5882fb16d486fa443170b5ed9f4a0 : FAILURE in 41m 37s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/0202ef15784644e5aabe5833a8495f6d : SUCCESS in 19m 21s\n- openstack-ansible-functional-ubuntu-focal-ssl https://zuul.opendev.org/t/openstack/build/95ff73ede1d4475d90e30f5057b3cc67 : SUCCESS in 22m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2aa955afaaeb4ebbc74d9be1ca3d7807d8147f46","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-21 07:54:14.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"45b2ca61a9eee6c28e31d5c0fc0303b343fd8d79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 10:29:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cb4febda67d0424c991402f5fb9e7756 : SUCCESS in 6m 30s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ec0df84507f94334844266e59ad7c8da : SUCCESS in 21m 01s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/b5e764b4a5e64cba990ef8fcf83e34ce : SUCCESS in 37m 44s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6256aa9b062546058a5be40c09036f8b : TIMED_OUT in 1h 34m 26s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a30169a01f1743568e19ab893aa02204 : SUCCESS in 19m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5b15443068be4d01aa64a6924a73161a : SUCCESS in 1h 06m 45s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/76bcf5a895b94002994a2a266ba6d15f : SUCCESS in 1h 20m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7173c3fb5b94ed4b7241094de85b0db : SUCCESS in 1h 03m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0db734626644c939312d881a4de16f6 : SUCCESS in 58m 27s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/314872f11420415c98f9a480fadc11ad : SUCCESS in 2h 16m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/138285c6e5494784bacfefc811a7eba0 : SUCCESS in 1h 43m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1f69ec77b5ca4a228e2dcb945dc57317 : SUCCESS in 1h 06m 50s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10ea8c621b15437db91b4438ee900244 : SUCCESS in 55m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/238dd36c8e5247caa47ffe3357b7df3b : SUCCESS in 17m 53s\n- openstack-ansible-functional-centos-8-ssl https://zuul.opendev.org/t/openstack/build/34a997d15b404f449c4fb6acd0fee227 : FAILURE in 45m 27s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/9764c30209ed4c28b3297fc1ab85af3e : FAILURE in 52m 56s\n- openstack-ansible-functional-ubuntu-focal-ssl https://zuul.opendev.org/t/openstack/build/e41a4d7968d24438ab1310af1d745697 : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":3},{"id":"d1d97bb97e86e54630d1545a675fd35999cf6cc3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-23 12:44:28.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"35e35b1ada33977d66360a4495e22dd2661649de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 15:09:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/25f324d7ac6540b6a3476121f7e6a705 : SUCCESS in 7m 02s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/388608a4aff34f9eab1ef8c048f8706c : SUCCESS in 18m 46s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/832ed800aeeb4f349bc650f36385f216 : FAILURE in 38m 22s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcb95a95274c4d4689bfd5b7741c5f69 : SUCCESS in 20m 05s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e1c10196571413394db6bc8d5b81fc6 : SUCCESS in 18m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0b7c775a1594429aa87d202a831f35c7 : SUCCESS in 1h 02m 49s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/326062afa3cd4228bb21fe7db548db75 : SUCCESS in 1h 19m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f40e51b5493d4062bba35e3c0c5b5ceb : SUCCESS in 1h 07m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6da4dffc5d24225882016c23804925f : SUCCESS in 1h 01m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/59b5c8e249a64e8ba331ca7b5f509dd0 : SUCCESS in 2h 04m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/693f6b9b8ffb4e08b65cfdf10b3018d5 : SUCCESS in 1h 35m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/267a4cc1a28b49de918f4d887280adab : SUCCESS in 54m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef14085d02664a8e951198524ccab237 : SUCCESS in 57m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db36a64115264e27b586b97ace4c94e0 : SUCCESS in 4m 57s\n- openstack-ansible-functional-centos-8-ssl https://zuul.opendev.org/t/openstack/build/eec713dbf8e7412dbdd107bb8995a43c : FAILURE in 36m 04s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/f2182234f367434dbe5e8c7825a1cd61 : SUCCESS in 19m 17s\n- openstack-ansible-functional-ubuntu-focal-ssl https://zuul.opendev.org/t/openstack/build/965fe5fe4ea94146b92cfe8acbf0fd1e : SUCCESS in 18m 27s","accounts_in_message":[],"_revision_number":3},{"id":"b24aec4f616cc2653c8bd93099572d2da1675658","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-23 21:13:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1328eb1b98ff8aa62d4176fc257a01bd6f964512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 23:53:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/abec3a04d187412788746eb07f490795 : SUCCESS in 9m 21s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a63df13452c5447aa4c30fbd046efab9 : SUCCESS in 20m 33s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/6bc03f6a4191438b874cb5febcd74f83 : FAILURE in 41m 50s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43f2cbdf889e4e63832895a1688670fa : SUCCESS in 23m 02s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9af7e502bbee45eb802899fc482e5daa : FAILURE in 37m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1d5740c6519d4af1a99d4ff6e4e165bf : SUCCESS in 59m 39s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8f8875cfb41b4065bdd161bf59342a21 : FAILURE in 25m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49533d1c836c41be8a5254b534303796 : SUCCESS in 1h 04m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c4087672c124f9a97eb7441683b9475 : SUCCESS in 1h 06m 27s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4832dcd2054d4e7aaf5533148a8b3046 : SUCCESS in 2h 21m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7575971365ba4c21a2d3ed23b99cdcb8 : SUCCESS in 1h 34m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6b1e9f0139e243a69e7852785d4b6db2 : FAILURE in 26m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ef8df701a214eeb836b09c1a10c93d5 : SUCCESS in 57m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/867947914a8c48a19b54eb10b65ea74d : SUCCESS in 8m 16s\n- openstack-ansible-functional-centos-8-ssl https://zuul.opendev.org/t/openstack/build/0d4760932f4d4a10b94b43f99e13fd06 : SUCCESS in 45m 35s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/f29e8d39941d4bd9b834a881aa534c3d : SUCCESS in 18m 18s\n- openstack-ansible-functional-ubuntu-focal-ssl https://zuul.opendev.org/t/openstack/build/f213d44784a243e5b9d01c89ec933489 : SUCCESS in 24m 18s","accounts_in_message":[],"_revision_number":3},{"id":"51697bd9c5a3994183f323f6a633fa0e4e8bd430","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-24 21:30:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3dafff9b700fbbd34758f959ab8fd8f10b0c18d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 00:19:05.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/799f4f201a184337bbf7f2b79d09e3f2 : SUCCESS in 7m 36s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/e69ac02ca5084b89a25ff348b0ac6ef3 : SUCCESS in 20m 38s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/b2b529b4b1af4f26b006df7b2ac8a121 : FAILURE in 40m 23s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da869b65be604c60a205c54f7b5a48ae : SUCCESS in 22m 56s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/992f1de0329b475aa9112593c08b3dc0 : SUCCESS in 22m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8ac572097b754dcc98ff54257be80659 : SUCCESS in 1h 11m 08s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/22945acc30a1488bb130ad191db432ae : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b07630d2d174c88a15343a9d8be0884 : SUCCESS in 1h 11m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f66a05940bac4f60867b6b1951182231 : SUCCESS in 57m 57s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b3bf6ca370334afcb0284b6f990d599f : SUCCESS in 1h 57m 19s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/639e308426e84f1bad421aa1765382f3 : SUCCESS in 1h 39m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/54bd714c0a2f41cea31117627241717c : SUCCESS in 57m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b98bedbe3f1e43f48df63ba11b2998e3 : SUCCESS in 49m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b906e4e9cd7c4612933733173006a621 : SUCCESS in 6m 32s\n- openstack-ansible-functional-centos-8-ssl https://zuul.opendev.org/t/openstack/build/007cdd575ddb4006b7e04f015ac43ecb : FAILURE in 35m 27s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/de808440c2d546df822e77a9542b8a7e : SUCCESS in 17m 07s\n- openstack-ansible-functional-ubuntu-focal-ssl https://zuul.opendev.org/t/openstack/build/d7e25668f88d4027a588efa5838de1b8 : SUCCESS in 20m 27s","accounts_in_message":[],"_revision_number":3},{"id":"38f03262fc365a106cac0734ca2cbf76b588c5f0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-05 19:37:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7ec80b0e2bc7122788a8fdca8f5d829840eee17c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 19:37:27.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"4a83f942207998cba379ca53d6057e5a36d1e9a9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-04 10:57:23.000000000","message":"Abandoned\n\ndone through a different patch - abandoning","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a4d6311bcb3f208d1e6a012f9f7f5e2c382b1158","revisions":{"ec4a69b202fd42fa3b0912b057d122317a8f79af":{"kind":"REWORK","_number":1,"created":"2020-07-21 08:52:38.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/04/742104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/04/742104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/742104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/742104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/742104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/742104/1"}}},"commit":{"parents":[{"commit":"dd1b276453f720f3c22c6c470dc86ab3c9e5edc2","subject":"Update CI tests for currently supported operating systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dd1b276453f720f3c22c6c470dc86ab3c9e5edc2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:50:20.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:51:40.000000000","tz":60},"subject":"Remove Centos-7 support","message":"Remove Centos-7 support\n\nUssuri is the last release that openstack-ansible supports Centos-7.\nRemove support for Centos-7 for Victoria.\n\nChange-Id: Ifc4ead03ac914275a302620a67a6f9cdb7275116\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ec4a69b202fd42fa3b0912b057d122317a8f79af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ec4a69b202fd42fa3b0912b057d122317a8f79af"}]},"branch":"refs/heads/master"},"169c24deea74cdd93b2b29e45e822b51441ba905":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-21 12:38:51.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/04/742104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/04/742104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/742104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/742104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/742104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/742104/2"}}},"commit":{"parents":[{"commit":"dd1b276453f720f3c22c6c470dc86ab3c9e5edc2","subject":"Update CI tests for currently supported operating systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dd1b276453f720f3c22c6c470dc86ab3c9e5edc2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:50:20.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 12:38:48.000000000","tz":0},"subject":"Remove Centos-7 support","message":"Remove Centos-7 support\n\nUssuri is the last release that openstack-ansible supports Centos-7.\nRemove support for Centos-7 for Victoria.\n\nDepends-On: https://review.opendev.org/742166\nDepends-On: https://review.opendev.org/742103\nChange-Id: Ifc4ead03ac914275a302620a67a6f9cdb7275116\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/169c24deea74cdd93b2b29e45e822b51441ba905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/169c24deea74cdd93b2b29e45e822b51441ba905"}]},"branch":"refs/heads/master"},"a4d6311bcb3f208d1e6a012f9f7f5e2c382b1158":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-09-21 07:54:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/04/742104/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/04/742104/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/742104/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/742104/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/742104/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/04/742104/3"}}},"commit":{"parents":[{"commit":"dd1b276453f720f3c22c6c470dc86ab3c9e5edc2","subject":"Update CI tests for currently supported operating systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dd1b276453f720f3c22c6c470dc86ab3c9e5edc2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:50:20.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-09-21 07:54:11.000000000","tz":0},"subject":"Remove Centos-7 support","message":"Remove Centos-7 support\n\nUssuri is the last release that openstack-ansible supports Centos-7.\nRemove support for Centos-7 for Victoria.\n\nDepends-On: https://review.opendev.org/742103\nChange-Id: Ifc4ead03ac914275a302620a67a6f9cdb7275116\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a4d6311bcb3f208d1e6a012f9f7f5e2c382b1158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a4d6311bcb3f208d1e6a012f9f7f5e2c382b1158"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
