)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~742105","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I76438e6519eac09be7f9729de3cefb4130f72dea","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"742105","hashtags":[],"change_id":"I76438e6519eac09be7f9729de3cefb4130f72dea","subject":"Update galera to 10.5.6","status":"MERGED","created":"2020-07-21 08:54:35.000000000","updated":"2020-10-17 16:02:17.000000000","submitted":"2020-10-17 16:01:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742105-1602950468209-76a6aaaf","meta_rev_id":"b2c8026a7a49b7392565b403dbabfb18094b1864","_number":742105,"virtual_id_number":742105,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-17 16:01:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-15 12:59:10.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-16 12:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2020-10-16 12:41:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2020-10-17 13:50:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 12:39:20.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-17 13:50:09.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-17 16:01:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"670cb0a9cf195d0551c8b1c78f29fafcabd7ab08","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-21 08:54:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbe1e666b37a339accd7497c67db236181919cd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 11:25:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f9cb7a4b644240419d943fbf0781f847 : SUCCESS in 7m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/be1b7465b2844be987010fea92e49d29 : SUCCESS in 25m 50s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1f5b9e5e1dea4e48bd3a99d4fc2bbcb8 : SUCCESS in 22m 25s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/ac04de72d36e40e69ff0511140cb21e5 : FAILURE in 12m 36s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44b72ec4bc82426dad89fb4ac4bc06d9 : SUCCESS in 20m 06s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bc774c2769c401b843b0464508359ac : FAILURE in 20m 28s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/38de8615a3104028892c418806a06bc9 : SUCCESS in 1h 07m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/65c76d65e3e642c8be5005ad0ccf6756 : SUCCESS in 1h 20m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/076f33adac54470697d5dfa741fc2c8b : SUCCESS in 1h 28m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02a1e56e27354506bd6a39d15c914adc : SUCCESS in 1h 06m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27a3453c51ec448f811fe92deeede04c : FAILURE in 26m 46s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a912cb952a6440869b66b3a1da565f2f : SUCCESS in 2h 19m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c763cb18792f44c9ac1f9011e121c775 : SUCCESS in 1h 48m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aeeca06de0e34314a53c486ae4246b00 : SUCCESS in 1h 02m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f62c0e61ecfd4c8fb2da95a087505a1e : SUCCESS in 58m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1b967f854424bc698ac415131b35f08 : SUCCESS in 1h 01m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b8221561a4466dae0f808854c6ce94 : SUCCESS in 6m 00s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/c76fcc3241fc4c938c4a17dff0cba25f : SUCCESS in 23m 02s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/b3750c7d0d6c4379b226eedf8f135a8d : SUCCESS in 23m 27s","accounts_in_message":[],"_revision_number":1},{"id":"70a776a890edff3b9466805cde50e7b8ad88897c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-21 11:32:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c3adf1c1c71ce19db927a49c19de842871e4ba7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-21 11:35:17.000000000","message":"Patch Set 2: Code-Review+1\n\nSounds nice!","accounts_in_message":[],"_revision_number":2},{"id":"7111fab702c32e7c249f8c67a0df89f0e80797ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 14:14:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/90751dc1107e4a068ece748481b50089 : SUCCESS in 7m 28s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b4b073d037d14483b3ada11e58432a35 : SUCCESS in 19m 09s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/64259abcc43c454bb35cf654bd1db1b8 : SUCCESS in 22m 49s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/9dab32de82f0497987d35a2b0cd8adf9 : FAILURE in 10m 11s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a6641ee7b3f4dcca7bce152e74f398b : SUCCESS in 20m 09s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc6f24081d5f48c6b87bfa3d9df38048 : FAILURE in 16m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/51034d86fb1b4609a9d9332478497bdc : SUCCESS in 1h 08m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/85e9a5d9669a4d2db796b0fc14d0b602 : SUCCESS in 1h 26m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f4307fb7c37140a9856009b1a00f9bef : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6aaf56cc82f24804bbae56fc5eb14567 : SUCCESS in 49m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5b49107700841359451aeb879e48e55 : FAILURE in 21m 18s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eb5690c194744941876096287d64acb3 : SUCCESS in 2h 28m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cd025891e17425fa6dafc52f643a05e : SUCCESS in 1h 49m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bfe97070873f4f619d3c007962f044aa : SUCCESS in 1h 03m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8d0e2be6d3d54fa2bc3b6ed86309a973 : SUCCESS in 1h 08m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb282cbb60e34014b5c47456df5f5827 : SUCCESS in 49m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f25332e5bf485f90ecb050aa86773d : SUCCESS in 6m 42s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/7d8019e854ac4f51b8f938c154759f93 : SUCCESS in 19m 06s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/3b23df2773c74f6d91f6f0596e88506b : SUCCESS in 20m 18s","accounts_in_message":[],"_revision_number":2},{"id":"431b0caa279c90bba02ef8c4fc2df11d7da8bb71","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-20 15:52:19.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0cd74685c8873e959779b7623f8765a654af67e3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-20 15:52:24.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"b77433ac39255fc11e2197a87c0c1988afeff48b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 18:38:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4e12d314fbf04aff814c76788c268438 : RETRY_LIMIT in 2m 30s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/6b6bf109ef9d43d58a797e6550551667 : SUCCESS in 17m 56s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2b2fc04bb9844ebcb7bb558bccc29c9d : SUCCESS in 18m 11s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/46ec7fbceb6043b8900daa14efd80e1e : FAILURE in 38m 47s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/564fd625764847be8f9315bfdf1644d6 : RETRY_LIMIT in 1m 38s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f99f8ae6fdf44c7a2ebecc4b3c0eb0f : RETRY_LIMIT in 1m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d5ab1c3e39664525a4eafcc7539d9ee5 : SUCCESS in 1h 05m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3a6c4f3254e34f3d950c16dc1f59630e : SUCCESS in 1h 28m 00s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1364741a4df24bf2a545a845271ba247 : SUCCESS in 1h 16m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8e58f9aa3df48fd898e8f87f377c527 : SUCCESS in 1h 11m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/405db7ec4494489f9490ae22d0f465f4 : RETRY_LIMIT in 1m 45s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/16a00501497540f0b7f7fcdd408db1c2 : SUCCESS in 2h 18m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/252b9cfcc1dc4deb9f53ddffdff7964a : RETRY_LIMIT in 2m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1929b9c864194bf3b9c068f379051174 : SUCCESS in 56m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6b7c896c37874dd1af18c464909be5e9 : SUCCESS in 58m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e11769228e994dbc87de0b72198b63bd : RETRY_LIMIT in 1m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e42b19e92e374340ac2bc101cf6dfb47 : FAILURE in 21m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed7fb6db23454dc8ae96a6925c242226 : RETRY_LIMIT in 2m 32s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/92722b71bc2e42a4a7766a14120078a1 : SUCCESS in 21m 38s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/9260ca3f59e64a34be12ed0f86c51e8c : RETRY_LIMIT in 1m 33s","accounts_in_message":[],"_revision_number":4},{"id":"055ebc6c0d9133d2f23cd24dc55bbe5c2eb5934d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-21 07:09:00.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"456822a3cba74e72352c13942d5e76c4a226a7f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 09:43:46.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e708cd13380742a3a042e3b588f0ef32 : SUCCESS in 8m 24s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/5c7f9e5c0cfd403e93e63be5169779b6 : FAILURE in 7m 50s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b758c1368f734205a42f1c48cc8ab3c1 : SUCCESS in 21m 48s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/57e7151fac2041e0b8224e42cc745a3f : FAILURE in 36m 10s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41cdfc298f5f4e76a2dd308cd518902f : SUCCESS in 17m 29s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81d8ded267784a32bfadce903142fd21 : FAILURE in 14m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/28fc326e70204c578b51e52a2d4c9891 : SUCCESS in 1h 10m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3d7141bc2ec445eb866aa17ab865a1a7 : SUCCESS in 1h 30m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/35147369059d4e07ae283473ae5d2e17 : SUCCESS in 1h 17m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/682256e77be84c29b4718ca091a4268d : SUCCESS in 1h 14m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f062e31c781748c8bdfc3e1818e2a100 : FAILURE in 20m 07s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d6807af780fb4ca59ea658733594ca5f : SUCCESS in 2h 19m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98602b46d350408892d1fbb33f5ba4ae : SUCCESS in 1h 47m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a38eea0fed1243bdbb574e6cd8ffea63 : SUCCESS in 57m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/24cf77de409840049ac0bf882ef6a8bc : SUCCESS in 1h 02m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/032cafeb2d2541a699ea8fc644f3813c : SUCCESS in 59m 50s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/859a857dd39b44d6b25e8159499a3019 : FAILURE in 20m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb33eab965642a2883b2d53b1bffd95 : SUCCESS in 7m 31s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/3c58c68ec4ec40fcbb73e5d69925adf0 : SUCCESS in 28m 26s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/f231dc18a01c4416875cc487df1bbdb6 : SUCCESS in 22m 38s","accounts_in_message":[],"_revision_number":4},{"id":"0380b7325b353cb9def12fa1709f750a3c86761c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-21 09:49:21.000000000","message":"Patch Set 4:\n\nwonder why this is not including the vars file for focal\n\n| ubuntu-focal | ok: [aio1] \u003d\u003e (item\u003d/home/zuul/src/opendev.org/openstack/openstack-ansible-galera_server/vars/debian.yml)","accounts_in_message":[],"_revision_number":4},{"id":"aa33d6dcf57103317f195642255247e861fc834f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-26 19:19:28.000000000","message":"Patch Set 4:\n\nBecause focal vars are deleted with this patch?","accounts_in_message":[],"_revision_number":4},{"id":"6220f731e433d83e519e7f80c2759cee6a0e8476","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-26 19:23:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5cca251dd6a681a38fb741a85309710fb7b9dea3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-26 19:25:06.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6d25d3ebb9c1158c557fa197e16090efdae717d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 22:00:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e936ecd1b3ca4bfe81ef56436c7a6703 : SUCCESS in 6m 17s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1cebb46cf7d54896bae461db63e2d0f3 : SUCCESS in 17m 48s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2e3d91e95ed2495d959d4bac0c9d5e71 : SUCCESS in 16m 42s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/ee640935f11a428cb670a6a6f95e1c85 : FAILURE in 55m 02s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f45f5ced73db4efe9cdc90e80578fe65 : SUCCESS in 20m 15s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/738d6a0f6f9d4384bc89cb33f9d9a85a : SUCCESS in 21m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/36e3f5f5c1c442a08a54a2fd3a1506a8 : FAILURE in 24m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/684a262e86704e67bfaea132a1e55689 : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/075ecba9a1df4f7b9542251a7304301a : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a84f191cacfe4a38b6535bd4d4046d4f : SUCCESS in 1h 03m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15b3d32bfa714014b335efd0a3298bc0 : SUCCESS in 1h 07m 02s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ab6b7462cea64f1e87400737820e72cb : SUCCESS in 2h 13m 42s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75e56cca7f8945b5b353e10f3ce552f6 : SUCCESS in 2h 20m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8b8ac1c316ac43029a3f74bac126fd3c : SUCCESS in 1h 33m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f9b210a97a2941f8bdad6456ccceb9e6 : SUCCESS in 1h 29m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea235e1d1cfb4e32b6855aaed943dab1 : FAILURE in 28m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52f5b878f4e24936bf7d39a1c609aee9 : SUCCESS in 52m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e4680906bbe473798d6cf4ab98d89fb : SUCCESS in 6m 45s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/d6f022baa47347ecb02486e2afa17ddb : SUCCESS in 21m 50s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/2bec5954e60c4f50908f5968ca654846 : SUCCESS in 20m 26s","accounts_in_message":[],"_revision_number":5},{"id":"84013af96c6866db92f1ccd85288dac2ca224570","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-27 06:52:44.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e048d43f29b297395002faffba4ea5fe1c6dc891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 09:15:30.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c69e91b502164f92a523d2bc1fb1637e : SUCCESS in 6m 13s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2ee4acb8075d49f39e7220c16f3c5169 : SUCCESS in 17m 51s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8786c17399004d90a8a925c2d39cc8c6 : SUCCESS in 19m 05s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/974958f932294eb081e6a24cab38c579 : FAILURE in 36m 43s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a9d5daf44c14512ae8022b97264415b : SUCCESS in 20m 48s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/084365f68de44d33843841bbfeaa78fd : SUCCESS in 20m 29s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/114fd4d92cef45539c859ca2f477eabf : SUCCESS in 1h 08m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f545aa6cd9794025ab36c27f4f575680 : SUCCESS in 1h 26m 53s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1cccacbb7d2c4aa987a98446815124d7 : SUCCESS in 1h 27m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cda480b7f2ad4f94a1cb1b14c9d43d86 : SUCCESS in 1h 03m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3ac68c12d974630b87cdb29bfdf9ced : SUCCESS in 1h 06m 23s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eeec4ebd4dd149f19f9447b55442bfa9 : SUCCESS in 2h 06m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b93bbfbb56b3419b81bbbc5c60f3cd45 : FAILURE in 1h 08m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b7ea8df345c54cee98bc7ff0c56df087 : SUCCESS in 1h 01m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5da185e642d64ec696a081c65a88970c : SUCCESS in 42m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1b937816a2242d79fa7aff7494d692f : SUCCESS in 1h 39m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/840d7bf6a3304497991aad67f099687c : SUCCESS in 53m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41cc2ad30bf642d9bbd64f6a02028b33 : SUCCESS in 5m 04s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/ae883b51c38346eda31edd7e0019d3c7 : SUCCESS in 21m 47s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/56a471f0fb784cfc87894a59b3880232 : SUCCESS in 26m 08s","accounts_in_message":[],"_revision_number":5},{"id":"3a96caf20f037e2fed808b8f43ad9e2065d9ad56","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-27 13:52:13.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"cc235b45767e3efb397da8f6bb7ab7ac236954d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 16:32:05.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cb804c5cb4d147d1a0acb76ed0b77b45 : SUCCESS in 6m 28s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/128b4b846ab8429785a580cad055238e : SUCCESS in 19m 10s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/146c8d8a3ca142a3871f8ece82b0eddd : SUCCESS in 25m 05s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/e2be46a35e9a41139373805883d6a2c2 : SUCCESS in 48m 17s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/258664260e7a47108ff5c0b5dc50fe72 : SUCCESS in 26m 10s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/154153d13a8446c59a0d9638847693f7 : SUCCESS in 20m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/28edd8eb9da349d894bccaee1fa2cd83 : SUCCESS in 1h 07m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6d1b0971916643e583f1d3b8f0c55a12 : SUCCESS in 1h 24m 16s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e534b0915f7d46ac97d2c03780faea3a : SUCCESS in 1h 31m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/764e91f71c50457db42a319996857209 : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6990de5ae58246a682f53242a03ad5e8 : SUCCESS in 1h 00m 44s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7a87e37d7c8f4bb2ba3e148a26d031d9 : SUCCESS in 2h 20m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31ed4a994342429dbf34e031b2026aac : FAILURE in 48m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8654a8d3663849edb50d951699fd8879 : SUCCESS in 1h 03m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/641d3c6ac2534fbd93f6a96736113ce8 : SUCCESS in 59m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bded996d30b14a4ebe109ccbc39416ca : SUCCESS in 1h 00m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47db38e1a6f847f28e2473bf48d89a63 : SUCCESS in 1h 01m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f29c6f523d274b3fac344f281866e052 : SUCCESS in 5m 41s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/df26c325dea64e90b3e41a4c3b79337a : SUCCESS in 21m 58s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/90a6137a210a4f5e9ccc0396c396c8bb : SUCCESS in 17m 45s","accounts_in_message":[],"_revision_number":5},{"id":"1fa47c6bcd09495b57d6263c955f480068e5e5c0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-28 08:30:22.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1530f58eb1b28d60e144956ea75564266fabf358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:11:19.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/198967f530f340968234725c0837f824 : POST_FAILURE in 7m 26s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ac37b1ce352447678cb957efdacd9996 : SUCCESS in 17m 42s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/db59f4bf594041ee980ca04322adc83d : SUCCESS in 19m 53s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/0b0a11d2e6f845389918baa56e3f56ac : FAILURE in 36m 10s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70febb0a153e4823a84d7aed1ad09333 : SUCCESS in 19m 41s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f5de3f50a38418b8ab669b1ca30f292 : SUCCESS in 19m 15s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/68269b1826fe4e43a89bc540e71f5721 : FAILURE in 2h 15m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8630e670d1c246e7adae5debe279fe5e : SUCCESS in 1h 19m 08s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0d0dd59388694eba965c0e034147cd8d : SUCCESS in 1h 17m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6229afe7ba1e45af84850ce719fa8803 : SUCCESS in 1h 06m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4de593f41c87484588d36efc07fcf72f : SUCCESS in 1h 07m 44s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4dd23502076546f483138d5103131252 : SUCCESS in 2h 21m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f981b4fc5414fcebd002e2133a11e4f : SUCCESS in 1h 57m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/206404d54a1d46e58667ee4fed083aa7 : SUCCESS in 1h 13m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2a0e714657df42eeb1aff71f62fd6b03 : SUCCESS in 1h 03m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e011c62eeef64bc5a14a6596b7cb81bd : SUCCESS in 50m 54s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fd7d2dcbdf548e0acbbf7d92ebe295b : SUCCESS in 50m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea27eb30406540b297a714a7e3b93591 : SUCCESS in 6m 45s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/ff915a443b0346b4828c0a406de7cad3 : SUCCESS in 19m 09s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/c252a7c11f554e2d921567acd614bfe9 : SUCCESS in 24m 17s","accounts_in_message":[],"_revision_number":5},{"id":"d43dea37182f4a054d456a9e1b8699ca69e20ef0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-28 11:19:53.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"782dc1979b9425d34eaf5c52489da77be9bf8508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 13:34:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c606779e69a1434aa879e93a17f8bcc9 : SUCCESS in 8m 35s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8697f1698b41492daa8ed96c954c6eb5 : SUCCESS in 16m 59s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/ada38210fe914a67a75e81d2d02aed69 : SUCCESS in 19m 21s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/8f525dfaa381460a8fa46dbc732264f0 : FAILURE in 53m 34s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a06f3623df294af498aa86d637da8184 : SUCCESS in 18m 50s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a0c9796189e4606baaaccedd383ae36 : SUCCESS in 17m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1c61d866c66b4a79b07972182dc9d171 : SUCCESS in 1h 09m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/73c73cecce39484884bcc7abf8e96b94 : SUCCESS in 1h 34m 26s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5a7968d0e4c1457eb6f8a66ae26f72ae : SUCCESS in 1h 16m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5c36b6d930d4af4abbec8d46807ff6e : SUCCESS in 1h 19m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0aac0ceddf784071b0487735df800109 : SUCCESS in 1h 09m 02s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dc3c7d14e6ba4d32bc35e0817f2c3b26 : SUCCESS in 2h 07m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6da1afac5e4e4d3486c7c4b224f8ccaa : SUCCESS in 1h 47m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/af5fe42e48d14d05a2c2704aad2e999a : SUCCESS in 1h 06m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/098f90c407cc4314b03a7d4d9d332f5d : SUCCESS in 1h 00m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2eb0f1bf9c624492bf39a8b20a4e507f : SUCCESS in 58m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe733b83d76e437aaee13098844e4f9d : SUCCESS in 1h 00m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a50c05cdd0fd41e1aacce04b688ef4af : SUCCESS in 7m 15s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/19035ed46ef34ad79156235406458499 : SUCCESS in 18m 06s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/923c55431b1d4ae5adcd68c800c654e6 : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":5},{"id":"ddd066a655503bcbae2f5007c578d8d0bc4ec6ed","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-01 06:27:54.000000000","message":"Patch Set 5: -Code-Review\n\nok, wait, centos8 is failing in cluster:(","accounts_in_message":[],"_revision_number":5},{"id":"26c853802ee4e1fa9e71b339a80c1b3d236694b4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-16 20:45:00.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"fb504af71381f3821aa7db5bc7efa7448e0eee44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 23:01:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d5f0ac2679f545979dded025219ed5b6 : SUCCESS in 9m 03s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/233dce683fa8423bb3680bd5038d5316 : SUCCESS in 20m 54s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/cb96164ce21c45a3b7bcbdf444c63d01 : SUCCESS in 19m 49s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/04761f5a372a4436a60262fbb627bba9 : FAILURE in 36m 12s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d26f8d66e3543f5a47768f9fc0a212d : SUCCESS in 19m 16s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/407990f944604876b9b8ac53d81ab786 : SUCCESS in 24m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b875cef2a113449e9682e03fddb16093 : SUCCESS in 1h 05m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/682b8cf43d5e43ab8b43b6d660bb743e : SUCCESS in 1h 25m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/029bfa509da9498690048fd83318c845 : SUCCESS in 1h 19m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41aef6534d4647048a4b5471d5f3833d : SUCCESS in 1h 04m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/488dcf9bb34f43d9b33af24db3d0bc36 : SUCCESS in 2h 03m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20b05c591dd44d6fabb53a6f6ed04367 : SUCCESS in 1h 29m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/225d406341fc43358d1c900f6086eae5 : SUCCESS in 1h 03m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e5ca019decd406c977f7a25ed28c37d : SUCCESS in 1h 06m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ce81cb374c42aa80148e5ccef56221 : SUCCESS in 14m 51s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/df28ac05988747eab0c7199b55f06ddd : SUCCESS in 20m 20s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/3f75c1bf9dda4b7ea75c9ff554b2285d : SUCCESS in 19m 47s","accounts_in_message":[],"_revision_number":5},{"id":"4084ac1eafc07637052f9ff06bbbc8779383fc55","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-25 08:01:10.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"7cc938710302b1adbfd9e1a79f1d2770f4125162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 10:14:00.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8cf8b181e02e4e77936920dca0030485 : SUCCESS in 16m 07s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9f17485e2e3643eebd4b183b259b6952 : SUCCESS in 17m 58s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/b66b199b9f7145d2b3025686f98b5ac5 : FAILURE in 35m 12s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92f383cd39b749aaa0c9804f7c9687a7 : POST_FAILURE in 48m 28s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0289d99d17a94be8bc25e049dc959d07 : SUCCESS in 19m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/760dd291df4641eab7949cc11ab11b86 : POST_FAILURE in 1h 28m 43s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a1074535c90847e0813b4296fbae39a4 : POST_FAILURE in 1h 39m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98ee6b1d69bd4d3393190aa15ffad3e3 : SUCCESS in 1h 09m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d57b9062c79a4fd5a7ec46b040f8380e : SUCCESS in 1h 03m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f0cb5db1b5b941b3b29fa42065d2ca0a : SUCCESS in 1h 59m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/710d4388c7e840aa8d870223a9e8f4f2 : SUCCESS in 1h 32m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f291958ab6414ef8a089e76f98a883b0 : SUCCESS in 58m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f02bb5c4bf5c48d2b4c11587ee559f86 : SUCCESS in 50m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5da5abc03f6a4f21936ec5c58b482333 : SUCCESS in 6m 31s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/1c8295fea81e462bafca353ad200c9ea : SUCCESS in 20m 44s","accounts_in_message":[],"_revision_number":5},{"id":"7615cd1c20d6f0cde05725657bcdf2a87ed26653","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-25 14:38:40.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"d93041f846fd934d240b1302122c49f40be09355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 16:48:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fbee7c1410e047639d5517aa04032d28 : SUCCESS in 7m 17s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/25a6b5f6b0d24452962ecabc493ebb1f : SUCCESS in 18m 10s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/98abd91e5746497fa594f57c08b86a05 : FAILURE in 36m 57s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f8c96e4bc874140a0d4703ff37e0a37 : FAILURE in 46m 22s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6492612170b43d3af79543643495ffb : SUCCESS in 18m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fb69ce6a2bb549f28e27f69ef41f6a03 : SUCCESS in 1h 08m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/17df7b9e37014b06a35e56df67f23607 : SUCCESS in 1h 16m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3a3dabf667545e5b08cb29f384b2de8 : SUCCESS in 1h 09m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49629e4f7a634b1c9276fc31d1438922 : SUCCESS in 1h 02m 00s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8b51c7d4465144ffa190e59ec91616b1 : SUCCESS in 1h 55m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f21317dbbf445ebbedf1620fba3bdd3 : SUCCESS in 1h 40m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0ed66b0b06384e5a8ffe72a177203ef5 : SUCCESS in 48m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cce92db06ef447d9b7ba295659003fc2 : SUCCESS in 1h 01m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be11395451ca4bce87460dac3a982e2e : SUCCESS in 7m 14s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/ae3834c03cd444148db0eb5dfa61dce5 : SUCCESS in 18m 00s","accounts_in_message":[],"_revision_number":6},{"id":"a20be7f6b0cccf6c9162e76a379efff40dea1c58","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-25 21:22:10.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"9441bae71897aa8ccd0d58e1ac85a83fbe607a37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 23:23:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a179765b1c1f4b8794c999b301419e1f : SUCCESS in 6m 31s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b1121e13d5dc4334a19eaf64c3292ece : SUCCESS in 16m 54s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/40fce9ae48b0439a8c63ea985dc39590 : FAILURE in 35m 58s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f0dabe0b0f84dadace5bca89862837e : SUCCESS in 19m 07s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfdb952d32b1459bb0c760817493a815 : SUCCESS in 19m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/44a70b16313244a790326069b3a70a5e : SUCCESS in 1h 01m 13s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/47a9ea691f2248f592ba3df6513d0de3 : SUCCESS in 1h 16m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bb892f7c49c4541882390669639f2a2 : SUCCESS in 1h 10m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b26d1a77e72144fca67ac74764e31de6 : SUCCESS in 1h 03m 48s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/81b07b1247834f159ad92bd83d562096 : SUCCESS in 1h 54m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7014e68fa5c14dc3be0668f788363923 : SUCCESS in 1h 27m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0e5a0e3675b24dbba3b6ead607126101 : SUCCESS in 54m 24s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/324a3df27b074ecb900aa8bdaf6b0467 : SUCCESS in 46m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a605ebdce0b64570a609170ec50c8f58 : SUCCESS in 5m 19s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/450f5450fb464f758b22d76617714f76 : SUCCESS in 19m 37s","accounts_in_message":[],"_revision_number":6},{"id":"3d894973e66c6d6b0a083e1237c7c300afbb4bac","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-26 06:37:56.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"3d746eacf5293587daa90a49e96a1d9de835dde1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-26 06:38:43.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"6c6a22258db12bda756345e39c084b217d5c1602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 09:00:31.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b191b40480c145ecae10736967ab089c : SUCCESS in 6m 46s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/564468a2d96f4916ab0cc90763b03a91 : SUCCESS in 20m 34s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/7bc47bcd42e84c83951a95382a2a6d61 : SUCCESS in 37m 25s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/111af51b706a4643adef8432b3c8420b : SUCCESS in 19m 38s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5153f064d858498389b0fcb0b68fc298 : SUCCESS in 20m 17s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e63806d20a9745e4bc534775b14bdeac : SUCCESS in 1h 05m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f9847807835e4253b25d1d60a73ffe7f : SUCCESS in 1h 15m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20d62cdaad014887b2484b82439c403c : SUCCESS in 1h 08m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38bad48ab88b4a948ce0577f7c553d9c : SUCCESS in 1h 04m 02s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/75ac15faf6c54a3a807dda455201ec4e : SUCCESS in 1h 59m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b66bda55d1d414d999dc88ae0596ed1 : SUCCESS in 1h 30m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f8250ae0f5b2493eb4e61a3729c7a999 : FAILURE in 23m 14s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dec326f135e4bf0a6442a905bb694d9 : SUCCESS in 52m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe163a174524b16a7c2719f9b6d5960 : SUCCESS in 5m 21s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/0a369113c6cb4d0b9d916f07676af07f : FAILURE in 45m 32s","accounts_in_message":[],"_revision_number":7},{"id":"69ad084dee7970bcb393676125afcdefa45cf846","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-26 11:42:27.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"31dc558721b70f4d863b84fae3354eb4a5b6ef01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 13:48:47.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0a68f03afa834b13abfb481ceda5302f : SUCCESS in 6m 08s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/0af5aaedeb304124840f3a3a3f9f50e6 : SUCCESS in 17m 07s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/30d9715674004824b50908118338333f : FAILURE in 36m 02s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a394a1da49548b892779b36aebfe42d : SUCCESS in 16m 41s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/921ba07088a34557884076b2d239b9fc : SUCCESS in 17m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/23c65c780cf34031bbdd472b0bbe48e4 : SUCCESS in 1h 02m 54s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ef04a737f23b4f8382e14adcbeee27c0 : SUCCESS in 1h 15m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44b177fd00844f95b63ff6eb3a9f48e2 : SUCCESS in 1h 07m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3456673c490441a681dc863cf6125fc9 : SUCCESS in 57m 39s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/56fe129b39a6434b83eae75a5f33a2ca : SUCCESS in 1h 56m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c530c482aed4446b79ffa05861ab828 : SUCCESS in 1h 28m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/45c369febe2f4e3bb17b7007695c5134 : SUCCESS in 56m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec66ef29a51744ee9deb4dfd14f88f5b : SUCCESS in 47m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f20d4a9a9df4cb8a85d77dc0f637705 : SUCCESS in 4m 34s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/401c32703f254035aa22b0125758852e : SUCCESS in 17m 46s","accounts_in_message":[],"_revision_number":7},{"id":"ae9e887c12647e8a00cae02f6d7147ffc3e2d993","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-27 12:33:40.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"50d91613f5aea16d3daa528cbdc9223c8b1242d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-27 14:36:09.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/072fde4c420b4efc8ff4743c53510c3e : SUCCESS in 4m 59s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/26f15496498d4fd3b12ecd665f054c28 : SUCCESS in 17m 15s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/5dec4ca98f7b4009a6de0e8dde791080 : FAILURE in 34m 11s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c23de2623f641c1a7fc5d1fe015de0f : SUCCESS in 17m 05s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c37ac7b08ace415fb941bdaae89e8c13 : SUCCESS in 16m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f21ae07388c349b2b344e51b681227f8 : SUCCESS in 1h 04m 07s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/500d91671f044088a628a72a83b1a1c7 : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7459e5925ea84f32836e87768f4eba79 : SUCCESS in 1h 18m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbf740ce73bf413abbf299998c614511 : SUCCESS in 58m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b7a78d8fcca64c94a7c71e6f6ebddda4 : SUCCESS in 1h 54m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87fb5f0d15f34b8d896012bf924fc081 : SUCCESS in 1h 31m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6ff12c08d9214e68a1bea53d54a7c1fc : SUCCESS in 56m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8382d0cadba04bd79908a183c614cefb : SUCCESS in 49m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/385e67621eb7403383b7028dcb40dc95 : SUCCESS in 5m 03s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/43609b4c62174e07ba4f7904f5f52744 : SUCCESS in 18m 14s","accounts_in_message":[],"_revision_number":7},{"id":"57f5451c8a89fe7f4faa6c9b22ca207f5602b11a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-06 14:29:54.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"82249c9ed2904ea99541d303e68ec3fbeacf1393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 17:16:30.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0a1db3d941c246e68e61e9a38839aa58 : SUCCESS in 7m 53s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ae2d5fc3c83341da9bf5ba28e6ffd5ad : SUCCESS in 20m 16s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/98f2cb48ce684582b74a45cde2c2f7af : FAILURE in 38m 42s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dc65a8268374460a796e026cec4f1ff : SUCCESS in 20m 34s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/109fb17c3c70479f94025449e6179b1f : SUCCESS in 22m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5b065627d13d45daa17d37e8eac1f46b : SUCCESS in 1h 10m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c366121659eb403d9c2e9103573a9306 : SUCCESS in 1h 15m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fa0d6ac254b4034ae57af70e9883641 : SUCCESS in 1h 15m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc66b84f236b47aeb19f7eed91901b0e : SUCCESS in 1h 02m 31s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/523ac6e0ecdd47f9a9fb1e73f8ccfac4 : SUCCESS in 2h 04m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0468ea7a6bf447429063ba94c107662b : SUCCESS in 2h 06m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/697be10b030f4878abdcb20cc05613e1 : SUCCESS in 1h 09m 24s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dd8ddfa5b5245ecb1d4e637a2dec216 : SUCCESS in 56m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98294a88100e4eca8c1bb2ef62589c3d : SUCCESS in 5m 43s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/f01cb40c4b654174b897434d6b6c94b9 : SUCCESS in 21m 37s","accounts_in_message":[],"_revision_number":8},{"id":"d1112d55f341d866ca167b451946093985cd668f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-14 09:37:17.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5dbd0ec06294a3b596e40478797e054bab10cde7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-14 09:37:29.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"c0b79fd26bd6ea0a915f3a2e894feb74851c2a9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 11:42:39.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7985b2c5a334c84a159fc74b7f7c889 : SUCCESS in 5m 27s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff3dd04559114b4f9825730af09083a1 : SUCCESS in 18m 44s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/cf98a7ac42b7454d90e47d9b211ee33b : SUCCESS in 1h 21m 17s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/2c63c82c44e54c13a3ebf544e3a59e53 : SUCCESS in 1h 07m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93d2090f6eee4bc3bffedaea2f2687cf : SUCCESS in 1h 06m 37s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/66e00d32174344b1be2da6e9f84f1cf5 : SUCCESS in 2h 01m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df9181fc59df451094a3111795651f08 : FAILURE in 1h 02m 09s","accounts_in_message":[],"_revision_number":9},{"id":"8a24ff0186273d3cd44d00d1eb839497b9840bb6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-14 12:13:51.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"4bc3c9626458053549a3724974dc92835e821670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 14:16:15.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfbfd3813fb8471a8ffec92beb5b8033 : SUCCESS in 4m 51s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5113ed5c9a9f460ba3a066705224f346 : SUCCESS in 15m 05s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/3b6f22f4a8db4c58955fbfb8a2b6c5e3 : FAILURE in 1h 18m 50s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/20285d1447e843e1b1aced04df30adce : FAILURE in 36m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3cbf07dbc5f41dfba5c15a9f91841dc : FAILURE in 1h 06m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a69d05ef54584f2b80fa0b11a2be298c : SUCCESS in 1h 30m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e08c30a215b4a5b93f9c9ee24aa5ecc : SUCCESS in 1h 53m 10s","accounts_in_message":[],"_revision_number":9},{"id":"98316439715001815542713732516186abee7aa0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-15 09:29:11.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"67b94e38005ae6dd3520c0ca7ffc2327ee4cc121","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-15 10:24:02.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"1ee8a93b6217f767bfcebd8eeedc33b4ac6768b1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-15 11:21:42.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"f517f5cdec5487e3376994b77b57c0d920f2320d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 12:53:56.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeb8de3d08df4996a430a3fe9be16618 : SUCCESS in 5m 54s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df92b8d2f6604bc7a3401b2f0219d3b9 : RETRY_LIMIT in 9m 34s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/8e9693034c0149428c6a0c59f62e8896 : SUCCESS in 1h 11m 25s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/8cb0403555b54acdbc2adda44b97f4d3 : SUCCESS in 1h 03m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6be4488cadc40a7a03cbd7ed71ec03c : RETRY_LIMIT in 12m 57s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4d63fa52716d4659a70a7390b336900f : FAILURE in 11m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/196ff1bce46b421eb9ab5ec60b3672a7 : FAILURE in 59m 36s","accounts_in_message":[],"_revision_number":11},{"id":"d2c1089394dd642403cc62cf947308eeb94da848","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-15 12:59:10.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"14b9661caff27433a179ca5a0d7f4696342bd6b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 14:51:57.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/834e52c3660c441f8b286ab979847104 : SUCCESS in 6m 59s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cf0f197581c4624af8016d8ee2212d0 : SUCCESS in 20m 29s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/c694863e7349412eb63580f018d9b0b3 : FAILURE in 38m 34s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/a650c2786d9c4966a4897c4b5633c220 : RETRY_LIMIT in 12m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29a0bc4e7bc44f7e971ab5e81ab519ab : FAILURE in 32m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/15122ea96236446ebc21f8d089ae6b6c : FAILURE in 1h 39m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8af304dbc08c496da620c4cf754e5f2a : FAILURE in 1h 44m 14s","accounts_in_message":[],"_revision_number":11},{"id":"0c67eddab9bf1081492da889cc1c020af879ad43","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-16 10:28:32.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"13d675555c2e6542d0f506b398958422c9017d13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 12:37:32.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f5b01158a141b19e2f693c09e2a1ea : SUCCESS in 8m 02s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e69d5ae47544fdc898aa6ffb5cdc176 : SUCCESS in 24m 26s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b8fad305cc6c42bfba4d8f16772bc4e6 : SUCCESS in 1h 22m 17s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/33bd03be945a40379d8dc19beb140007 : SUCCESS in 1h 10m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19352ecb29074ddca9768941a0b24bdd : SUCCESS in 59m 09s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f8a158c10385409880f49527908d9c45 : SUCCESS in 2h 01m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d435cd807f7401c908a7bf98265a112 : SUCCESS in 1h 58m 23s","accounts_in_message":[],"_revision_number":11},{"id":"8a41f61a3a83234fd68efb177c1f314543e5cc99","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-16 12:39:20.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"75c9ac0d76101b78e7261508fd9727971de94070","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-16 12:41:27.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"db6fdfe329078a79b1ec9a254fcf61da17794d27","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-17 13:50:07.000000000","message":"Patch Set 11: -Workflow","accounts_in_message":[],"_revision_number":11},{"id":"544bfda6744bd4f9a09bc775934130de18370ea4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-17 13:50:09.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"c2ce415925c72dfb2f5f814c3ce24cd3903b89bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 13:50:20.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"4dfe5a353f856deaebf40f400d5dc8234ff42f8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 16:01:07.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7588b2cc1449849bded86a70873212 : SUCCESS in 8m 01s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48c88548992c47f4992e00f6dae795f3 : SUCCESS in 17m 18s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/4f3770e96cb348e7a76c5f4c88e33277 : SUCCESS in 1h 21m 46s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/59c015a004a346f3b44b4b8bb9eff076 : SUCCESS in 1h 02m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/875ee0668bde487ebc0cf00fc76aa378 : SUCCESS in 1h 05m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/494faffa539a45edbb784a9c1bcbf47a : SUCCESS in 1h 22m 54s","accounts_in_message":[],"_revision_number":11},{"id":"4327a4efbb36ef6613abc60256b5020dc7162f7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 16:01:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"0eafa07ecc0641eb060493c3e249a51826d4e4bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 16:02:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1614fe7436174ad4b27a3e3efbb9d0e1 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"7075e72c654b87870364c85ee2677a164f55afca","revisions":{"6d9bcd9392c9dac37d06a5f7cab5d5744ec44cdc":{"kind":"REWORK","_number":1,"created":"2020-07-21 08:54:35.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/05/742105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/742105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/1"}}},"commit":{"parents":[{"commit":"c567405b567dc6276c0c83b3d166328bd4327e37","subject":"Merge \"Remove Suse tests.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c567405b567dc6276c0c83b3d166328bd4327e37"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:53:59.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:53:59.000000000","tz":60},"subject":"Update galera to 10.5.4","message":"Update galera to 10.5.4\n\nThis is the first stable release from the 10.5 series.\n\nChange-Id: I76438e6519eac09be7f9729de3cefb4130f72dea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6d9bcd9392c9dac37d06a5f7cab5d5744ec44cdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6d9bcd9392c9dac37d06a5f7cab5d5744ec44cdc"}]},"branch":"refs/heads/master"},"61d8103884ca6b5c10b6fed723481e468de9e7c3":{"kind":"REWORK","_number":2,"created":"2020-07-21 11:32:27.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/05/742105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/742105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/2"}}},"commit":{"parents":[{"commit":"c567405b567dc6276c0c83b3d166328bd4327e37","subject":"Merge \"Remove Suse tests.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c567405b567dc6276c0c83b3d166328bd4327e37"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:53:59.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 11:32:11.000000000","tz":60},"subject":"Update galera to 10.5.4","message":"Update galera to 10.5.4\n\nThis is the first stable release from the 10.5 series.\n\nChange-Id: I76438e6519eac09be7f9729de3cefb4130f72dea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/61d8103884ca6b5c10b6fed723481e468de9e7c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/61d8103884ca6b5c10b6fed723481e468de9e7c3"}]},"branch":"refs/heads/master"},"12ec885e5f41c252035dac41491be66ba20e0c29":{"kind":"REWORK","_number":3,"created":"2020-08-20 15:52:19.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/742105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/742105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/3"}}},"commit":{"parents":[{"commit":"c567405b567dc6276c0c83b3d166328bd4327e37","subject":"Merge \"Remove Suse tests.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c567405b567dc6276c0c83b3d166328bd4327e37"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:53:59.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-08-20 15:52:15.000000000","tz":0},"subject":"Update galera to 10.5.5","message":"Update galera to 10.5.5\n\nThis is the first stable release from the 10.5 series.\n\nChange-Id: I76438e6519eac09be7f9729de3cefb4130f72dea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/12ec885e5f41c252035dac41491be66ba20e0c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/12ec885e5f41c252035dac41491be66ba20e0c29"}]},"branch":"refs/heads/master"},"7782abde71e00b7d374c96e2a2587e927a3da192":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-20 15:52:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/742105/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/742105/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/4"}}},"commit":{"parents":[{"commit":"3b44cf919c56fb6225ef9f87ad1829ecf9303de5","subject":"Fix default galera_wsrep_provider_options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3b44cf919c56fb6225ef9f87ad1829ecf9303de5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:53:59.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-08-20 15:52:24.000000000","tz":0},"subject":"Update galera to 10.5.5","message":"Update galera to 10.5.5\n\nThis is the first stable release from the 10.5 series.\n\nChange-Id: I76438e6519eac09be7f9729de3cefb4130f72dea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7782abde71e00b7d374c96e2a2587e927a3da192"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7782abde71e00b7d374c96e2a2587e927a3da192"}]},"branch":"refs/heads/master"},"63052152ebe29232eb848573bf94c41e9d384640":{"kind":"REWORK","_number":5,"created":"2020-08-26 19:23:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/742105/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/742105/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/5"}}},"commit":{"parents":[{"commit":"3b44cf919c56fb6225ef9f87ad1829ecf9303de5","subject":"Fix default galera_wsrep_provider_options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3b44cf919c56fb6225ef9f87ad1829ecf9303de5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:53:59.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-26 19:23:31.000000000","tz":180},"subject":"Update galera to 10.5.5","message":"Update galera to 10.5.5\n\nThis is the first stable release from the 10.5 series.\n\nChange-Id: I76438e6519eac09be7f9729de3cefb4130f72dea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/63052152ebe29232eb848573bf94c41e9d384640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/63052152ebe29232eb848573bf94c41e9d384640"}]},"branch":"refs/heads/master"},"9003d42f3593c268b7ea9d09092aaf4873036eb7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-09-25 14:38:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/742105/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/742105/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/6"}}},"commit":{"parents":[{"commit":"f9b5a696f03b4000b48336503a8b7493951fdb81","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f9b5a696f03b4000b48336503a8b7493951fdb81"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:53:59.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-25 14:38:20.000000000","tz":180},"subject":"Update galera to 10.5.5","message":"Update galera to 10.5.5\n\nThis is the first stable release from the 10.5 series.\n\nChange-Id: I76438e6519eac09be7f9729de3cefb4130f72dea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9003d42f3593c268b7ea9d09092aaf4873036eb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9003d42f3593c268b7ea9d09092aaf4873036eb7"}]},"branch":"refs/heads/master"},"ef761373625399c31a6204f189f46a528ecff681":{"kind":"REWORK","_number":7,"created":"2020-09-26 06:37:56.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/742105/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/742105/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/7"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:53:59.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-26 06:37:45.000000000","tz":180},"subject":"Update galera to 10.5.5","message":"Update galera to 10.5.5\n\nThis is the first stable release from the 10.5 series.\n\nChange-Id: I76438e6519eac09be7f9729de3cefb4130f72dea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ef761373625399c31a6204f189f46a528ecff681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ef761373625399c31a6204f189f46a528ecff681"}]},"branch":"refs/heads/master"},"dce9445c44f2f39c958f7dfdf273f5e70cf3f0b8":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-10-06 14:29:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/742105/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/742105/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/8"}}},"commit":{"parents":[{"commit":"0748f362de866aa2a037d6ea8cf6e985e8b1a4b6","subject":"Merge \"Explicitly use mysql collection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0748f362de866aa2a037d6ea8cf6e985e8b1a4b6"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:53:59.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-06 14:29:54.000000000","tz":0},"subject":"Update galera to 10.5.5","message":"Update galera to 10.5.5\n\nThis is the first stable release from the 10.5 series.\n\nChange-Id: I76438e6519eac09be7f9729de3cefb4130f72dea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dce9445c44f2f39c958f7dfdf273f5e70cf3f0b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dce9445c44f2f39c958f7dfdf273f5e70cf3f0b8"}]},"branch":"refs/heads/master"},"1448554c242445dc81e017e9c03d4bb6eb7fa724":{"kind":"REWORK","_number":9,"created":"2020-10-14 09:37:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/742105/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/742105/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/9"}}},"commit":{"parents":[{"commit":"1d0b447f80b636f8a745a0db4a68513477c4ccc1","subject":"Use infra jobs instead of functional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1d0b447f80b636f8a745a0db4a68513477c4ccc1"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:53:59.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-14 09:37:02.000000000","tz":180},"subject":"Update galera to 10.5.5","message":"Update galera to 10.5.5\n\nThis is the first stable release from the 10.5 series.\n\nChange-Id: I76438e6519eac09be7f9729de3cefb4130f72dea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1448554c242445dc81e017e9c03d4bb6eb7fa724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1448554c242445dc81e017e9c03d4bb6eb7fa724"}]},"branch":"refs/heads/master"},"331a1ad7bd52c28e1d2f64457ed249b32e8e4f0e":{"kind":"REWORK","_number":10,"created":"2020-10-15 09:29:11.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/05/742105/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/742105/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/10"}}},"commit":{"parents":[{"commit":"1d0b447f80b636f8a745a0db4a68513477c4ccc1","subject":"Use infra jobs instead of functional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1d0b447f80b636f8a745a0db4a68513477c4ccc1"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:53:59.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-15 09:29:08.000000000","tz":0},"subject":"Update galera to 10.5.6","message":"Update galera to 10.5.6\n\nThis is the first stable release from the 10.5 series.\n\nChange-Id: I76438e6519eac09be7f9729de3cefb4130f72dea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/331a1ad7bd52c28e1d2f64457ed249b32e8e4f0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/331a1ad7bd52c28e1d2f64457ed249b32e8e4f0e"}]},"branch":"refs/heads/master"},"7075e72c654b87870364c85ee2677a164f55afca":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2020-10-15 11:21:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/742105/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/742105/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/742105/11"}}},"commit":{"parents":[{"commit":"1d0b447f80b636f8a745a0db4a68513477c4ccc1","subject":"Use infra jobs instead of functional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1d0b447f80b636f8a745a0db4a68513477c4ccc1"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 08:53:59.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-15 11:21:39.000000000","tz":0},"subject":"Update galera to 10.5.6","message":"Update galera to 10.5.6\n\nThis is the first stable release from the 10.5 series.\n\nDepends-On: https://review.opendev.org/758399\nChange-Id: I76438e6519eac09be7f9729de3cefb4130f72dea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7075e72c654b87870364c85ee2677a164f55afca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7075e72c654b87870364c85ee2677a164f55afca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
