)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~754610","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I1c1d5f8b80c55b239db89671972dbfa82800bae6","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"I1c1d5f8b80c55b239db89671972dbfa82800bae6","subject":"DNM Try to understand what\u0027s wrong in CI","status":"ABANDONED","created":"2020-09-27 15:12:17.000000000","updated":"2020-10-01 16:26:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"757485f87c06c16cd1a7913d7053eb4f24d4e1e8","_number":754610,"virtual_id_number":754610,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-29 13:05:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 13:05:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e33057438eecf12a5f41464d211f5fcb7e5d3d9b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-27 15:12:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07aa6238eac722d82844db6a886f6f556b2e6d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-27 17:15:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a8000e06ac9349ad84a357419580a08a : SUCCESS in 5m 45s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/01a7f07faa4e4598a15231e196f56979 : SUCCESS in 16m 16s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/2d4047b782064110a9296f1cc7ccc191 : FAILURE in 21m 54s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b5fab2aa6384b779ed8d390855d9efb : SUCCESS in 16m 31s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04db6228225442ab992735fbf8b2a18e : SUCCESS in 18m 47s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0f116afa15a64ba680feca02ac0452b0 : SUCCESS in 1h 01m 05s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d6cb0c56226e4af792d3fbd04f8b9049 : SUCCESS in 1h 16m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2e10378da494ef9afb8cadaf44d3dcd : SUCCESS in 1h 09m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4067b857933641e1865ed881a4942377 : SUCCESS in 56m 42s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e284c8fab2df49349cf68181348c2908 : SUCCESS in 1h 54m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2c87e71c3704955b9c3bebbeaa7f1b4 : SUCCESS in 1h 27m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5ee72ca340794b38ad0119845766ed83 : SUCCESS in 56m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16fbce84e25a47e9b7cdde64efc185b3 : SUCCESS in 49m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f162c0bf8344615af6e7d8774411577 : SUCCESS in 4m 57s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/bb2d26a28fd94e9c9d1733f8243fdcbf : SUCCESS in 16m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0ea40ee6d04b129994fb49f9f34a8939fe0eab3c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-28 08:18:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"07224221cce1e22710246e11fe9c3ad1708e2ada","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 10:44:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b0e24546243a48a4b9b978e99567edab : SUCCESS in 7m 29s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ddd55eb03b5845e49f9ccc65a4f4e949 : SUCCESS in 18m 14s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/70980e72253c47309524b48cb76c9052 : FAILURE in 27m 16s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85d8167c34e04f7e8a0456ca7e4604c2 : SUCCESS in 20m 51s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4237c55ec1bf47bab34e99aa96ea4468 : SUCCESS in 18m 46s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dfad2294ba224b68a3e6eb06cd60f9a6 : SUCCESS in 1h 02m 07s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/56df9edd5f4a472f9372455ed9ea515a : SUCCESS in 1h 26m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ccb063069d545918e81c8f49763b235 : SUCCESS in 1h 03m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34be785be1b744cdb6d29cdc6f206a38 : SUCCESS in 58m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/78f9b1ef941d4fc69ab457711a096066 : SUCCESS in 2h 11m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f22d067410140cd80a06bc17a129bb2 : SUCCESS in 1h 31m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f32f80cae02545f785a4af72cf107a05 : SUCCESS in 1h 09m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69f4ac497c6948ac9d56085f80297ce4 : SUCCESS in 58m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0264f58b4b2e4b6096dd93adcaf57cd9 : SUCCESS in 5m 03s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/fa83af5ef65e47b1a9d64ee190f1faef : SUCCESS in 19m 05s","accounts_in_message":[],"_revision_number":2},{"id":"ada92f92a5ed75c6f9979f18ff501bdd70bb3e19","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-28 12:16:21.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"21d85afc22ad2a665fb955042e372c510ea621fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 14:23:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/985ae8a2566b4620ac1a7ebef21ed5dd : SUCCESS in 7m 16s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/94f99562fda14c499abb4143ad155cd8 : SUCCESS in 17m 18s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/b0cc37c47ccd43e2995db4bbe551d589 : FAILURE in 24m 58s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0013edf4ca844f878c399ddb29cf7e19 : SUCCESS in 19m 36s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/958da5984b5943dda7bafc64b2a237e8 : SUCCESS in 17m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c3ada1d4c5144785aa7e134c8174e287 : SUCCESS in 1h 08m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c23e84ee163e4d2f9df868941d30b624 : SUCCESS in 1h 15m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d04c6d123284928b51fe79d0a034d87 : SUCCESS in 1h 15m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51a41908c79d474587d3ff4adcac0eff : SUCCESS in 1h 02m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/42f1b3d1d65f4d32a1f253ceaed2f626 : SUCCESS in 1h 56m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89ba49b9cacd4bf982a72f4bc6b95312 : SUCCESS in 1h 33m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0fa7e717a38a42cfae1f8bb4a7d207f8 : SUCCESS in 59m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edfc77026f654ebd99b0cfecbcb57297 : SUCCESS in 53m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/949064ad06f44788a591c3b2d8e70ccc : SUCCESS in 5m 24s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/6ef591ebc39a4e87a6b67e417dda1cc8 : SUCCESS in 21m 10s","accounts_in_message":[],"_revision_number":3},{"id":"22512174d086c7d82d013f0a8778002336feac2f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-28 14:59:38.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5ea374800dd770e8cf40dfa65a5cf011e1aaab28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 17:28:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/155db65a1231400692d033698f353a84 : SUCCESS in 6m 02s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/86e921bedce14a3e8586d1e526089fb1 : SUCCESS in 17m 57s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/28fea04132ba470cab86c928cdacd84f : FAILURE in 37m 04s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c965482479154badb6b95e72efd84496 : SUCCESS in 19m 33s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ad4fad1e9ee42609ce8b405d0ad873e : SUCCESS in 22m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0dd84204149348bb9fffcc0febac5fb3 : SUCCESS in 1h 10m 55s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3905abac9ae543b7aadc706f7f01a356 : SUCCESS in 1h 15m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12a36ac88cf0470a843b286340d7fc04 : SUCCESS in 1h 13m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/699e2007bffa4e9e84a6a32d71cdebe2 : SUCCESS in 1h 01m 19s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d3d01c36f71e49229b22d293943c7f4f : SUCCESS in 2h 12m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45171cea88fb4223ae24b973e9d6068d : SUCCESS in 1h 33m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f356955021fa4b98be3c3f9f1f654247 : SUCCESS in 57m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09e95407cc4a43aea084248449691854 : SUCCESS in 49m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ef2c247d3d44f4c8077d6dee36d8c06 : SUCCESS in 5m 32s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/1e3d51968dc849e0a1eccfb1d760b03e : SUCCESS in 20m 04s","accounts_in_message":[],"_revision_number":4},{"id":"9551ed8b0cf93cb86393458e160ac679f7555cb2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-29 08:57:16.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"5df8bc80d682a8f3863e655172a8795cce7b794e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-29 09:27:23.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"ed7765c8c7602de07746beff7f298c377f5126f1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-29 10:59:55.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"d901f6f50cd97b2978f27bee7d8de5a908e6ad4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 13:05:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/604fddfdb7e44dea993a05106deba052 : SUCCESS in 8m 21s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/293ce1c011134ac9bed7ec49aa0a1287 : SUCCESS in 21m 51s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/c4ac94744d4d4d6296e7265d31c5280b : FAILURE in 43m 43s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74a1722f32194843a990184fc1b4b073 : SUCCESS in 21m 14s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d818abcfff64298a5898aa205e0b64f : SUCCESS in 22m 09s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/eeddc519f7c24b2198e164bf41a162e4 : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6c05383b63ff40339edb1d5c4c27ef80 : SUCCESS in 1h 26m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c275d0973344256ae9333ba6a4c6244 : SUCCESS in 1h 22m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56937dcab2f547859f58d2cac8f7fb1c : SUCCESS in 1h 13m 13s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/322031765bca401ca65c060cbff0fa79 : SUCCESS in 1h 58m 56s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e359065c72ee4243ad99ae938e981d88 : SUCCESS in 1h 37m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1a101a24c4fa41a2adb8f247ccbbc3f2 : SUCCESS in 1h 07m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0b130ff7cdd489bbfcf6317fa529fb9 : SUCCESS in 59m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03464d14457c41c08b25b23d2a2dd9d1 : SUCCESS in 6m 33s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/4e287bf320c348138f3112c2aa216a3b : SUCCESS in 21m 04s","accounts_in_message":[],"_revision_number":7},{"id":"757485f87c06c16cd1a7913d7053eb4f24d4e1e8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-01 16:26:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"52e31b667604a5f8e872e43dc703f863305c0c41","revisions":{"ccd82a6a770a0dfd6255dcb0a73fce32f583023d":{"kind":"REWORK","_number":1,"created":"2020-09-27 15:12:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/754610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/10/754610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/1"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-27 15:12:07.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-27 15:12:07.000000000","tz":180},"subject":"DNM Try to understand what\u0027s wrong in CI","message":"DNM Try to understand what\u0027s wrong in CI\n\nChange-Id: I1c1d5f8b80c55b239db89671972dbfa82800bae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ccd82a6a770a0dfd6255dcb0a73fce32f583023d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ccd82a6a770a0dfd6255dcb0a73fce32f583023d"}]},"branch":"refs/heads/master"},"f670d5e21d74d9342067bf25e471c35d997f1da9":{"kind":"REWORK","_number":2,"created":"2020-09-28 08:18:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/754610/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/10/754610/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/2"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-27 15:12:07.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-28 08:18:25.000000000","tz":180},"subject":"DNM Try to understand what\u0027s wrong in CI","message":"DNM Try to understand what\u0027s wrong in CI\n\nChange-Id: I1c1d5f8b80c55b239db89671972dbfa82800bae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f670d5e21d74d9342067bf25e471c35d997f1da9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f670d5e21d74d9342067bf25e471c35d997f1da9"}]},"branch":"refs/heads/master"},"db9bfc8455ca8bd6266456e4330d642457afaf97":{"kind":"REWORK","_number":3,"created":"2020-09-28 12:16:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/754610/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/10/754610/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/3"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-27 15:12:07.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-28 12:15:55.000000000","tz":0},"subject":"DNM Try to understand what\u0027s wrong in CI","message":"DNM Try to understand what\u0027s wrong in CI\n\nChange-Id: I1c1d5f8b80c55b239db89671972dbfa82800bae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/db9bfc8455ca8bd6266456e4330d642457afaf97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/db9bfc8455ca8bd6266456e4330d642457afaf97"}]},"branch":"refs/heads/master"},"d04197a0e7efa5de8c7542c85530ab15f41114ce":{"kind":"REWORK","_number":4,"created":"2020-09-28 14:59:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/754610/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/10/754610/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/4"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-27 15:12:07.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-28 14:59:36.000000000","tz":0},"subject":"DNM Try to understand what\u0027s wrong in CI","message":"DNM Try to understand what\u0027s wrong in CI\n\nChange-Id: I1c1d5f8b80c55b239db89671972dbfa82800bae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d04197a0e7efa5de8c7542c85530ab15f41114ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d04197a0e7efa5de8c7542c85530ab15f41114ce"}]},"branch":"refs/heads/master"},"6eba9bc97d553c30db934131cb1c6b4e711cb789":{"kind":"REWORK","_number":5,"created":"2020-09-29 08:57:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/754610/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/10/754610/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/5"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-27 15:12:07.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-29 08:57:11.000000000","tz":0},"subject":"DNM Try to understand what\u0027s wrong in CI","message":"DNM Try to understand what\u0027s wrong in CI\n\nChange-Id: I1c1d5f8b80c55b239db89671972dbfa82800bae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6eba9bc97d553c30db934131cb1c6b4e711cb789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6eba9bc97d553c30db934131cb1c6b4e711cb789"}]},"branch":"refs/heads/master"},"65a5d7090d916c9683301bdb0517958a94f93ecc":{"kind":"REWORK","_number":6,"created":"2020-09-29 09:27:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/754610/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/10/754610/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/6"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-27 15:12:07.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-29 09:27:12.000000000","tz":0},"subject":"DNM Try to understand what\u0027s wrong in CI","message":"DNM Try to understand what\u0027s wrong in CI\n\nChange-Id: I1c1d5f8b80c55b239db89671972dbfa82800bae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/65a5d7090d916c9683301bdb0517958a94f93ecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/65a5d7090d916c9683301bdb0517958a94f93ecc"}]},"branch":"refs/heads/master"},"52e31b667604a5f8e872e43dc703f863305c0c41":{"kind":"REWORK","_number":7,"created":"2020-09-29 10:59:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/754610/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/10/754610/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/10/754610/7"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-27 15:12:07.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-29 10:59:50.000000000","tz":0},"subject":"DNM Try to understand what\u0027s wrong in CI","message":"DNM Try to understand what\u0027s wrong in CI\n\nChange-Id: I1c1d5f8b80c55b239db89671972dbfa82800bae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/52e31b667604a5f8e872e43dc703f863305c0c41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/52e31b667604a5f8e872e43dc703f863305c0c41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
