)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~755586","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Id50807afaafa47bca409dcb109c8327476bf60ba","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"755586","hashtags":[],"change_id":"Id50807afaafa47bca409dcb109c8327476bf60ba","subject":"Use infra jobs instead of functional","status":"MERGED","created":"2020-10-01 16:25:30.000000000","updated":"2020-10-14 09:36:41.000000000","submitted":"2020-10-14 09:32:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755586-1602667929656-6db76d6a","meta_rev_id":"29f00d23796ba6cbe656477486e374514919846a","_number":755586,"virtual_id_number":755586,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"date":"2020-10-13 13:05:21.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-14 09:32:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-14 07:14:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-13 16:19:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-10-14 07:14:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-13 16:19:36.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-14 07:14:40.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-14 09:32:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e49338056584802541eccf1ece775c7cee1b833","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-01 16:25:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02754274bdb74ad0db74bf6a2ff1cb6c080edefa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 17:46:59.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"fb8cb5737257763561946f1d992ff497dd23ddfe","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-01 17:48:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a73a05562353f2ba7c1c8503a97d3a36f98afd65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 17:49:09.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-ansible-galera_server on branch master.  The\nerror was:\n\n  The project template \"openstack-ansible-deploy-infra_lxc-jobs\" was not\n  found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-ansible-upgrade-aio_metal-jobs\n        - openstack-ansible-deploy-infra_lxc-jobs\n        - check-requirements\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-ansible-functional-centos-7-ssl\n          - openstack-ansible-functional-ubuntu-bionic-ssl\n      gate:\n        jobs:\n          - openstack-ansible-functional-centos-7-ssl\n          - openstack-ansible-functional-ubuntu-bionic-ssl\n\n  in \"openstack/openstack-ansible-galera_server/zuul.d/project.yaml@master\", line 16, column 3","accounts_in_message":[],"_revision_number":1},{"id":"12b4fb6d407f51631ea948d09fc858170a44d06a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-01 17:51:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"45fade882e97abb0bad524f36818e60a43f990c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 21:11:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1ab9e22c812549418e6318dde6361885 : TIMED_OUT in 3h 02m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b355eb278fa4523bc5715c345355e1d : SUCCESS in 1h 37m 41s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b07aa0064bb64863a70d4abec0fedf77 : FAILURE in 52m 20s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/4a84283bc2bb4b4381a68a5d80345fad : FAILURE in 46m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65aae1f3fc164e2689c5e5220ad1f4cf : SUCCESS in 1h 04m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80d7fcd0b6684e52aec1d139dae021ec : SUCCESS in 6m 29s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/0a1bf2dea38b441ebdb3d32b5e39f7f4 : SUCCESS in 20m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d313f362f58e63eaa1bf1fbcf68f6913694823e2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 08:44:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e9357052cc7989973340787f76c460789ae88e9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 08:45:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0bb9e83fc34ac0366bec1ec92ff2029a73b4f3e0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 08:47:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"89171d8b6bb072cefb06434162a91d0c6a9a9ef2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 09:55:15.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"3540f494822ab0d66dc9edd22f116d15d12ea115","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 10:05:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2071d0379de45809a17f9f7dbe6bc2b873865117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 12:32:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f97994d58bb4dc486a0a2ae031c0cf1 : SUCCESS in 5m 23s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad5e75fe2c6448fe846ac918c72dbbc4 : FAILURE in 17m 24s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/98f114fef0654c7ebed7b01cf7f9e326 : SUCCESS in 1h 16m 58s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/ed2e40d29bfe407bbda9f101cd817d3b : FAILURE in 46m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16964c48a5d84434a25305aa90382c29 : SUCCESS in 59m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/54e601df05cd4279a64d5435b1c18649 : SUCCESS in 2h 00m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/116445ea38ec448d92996ec642d182f1 : SUCCESS in 1h 36m 46s\n- openstack-ansible-functional-centos-8-ssl https://zuul.opendev.org/t/openstack/build/3640c8e80d2a4e7797c27bbebaa321df : FAILURE in 53m 57s\n- openstack-ansible-functional-ubuntu-focal-ssl https://zuul.opendev.org/t/openstack/build/98e0a3cac0c84eaca2deb12d726bd7cd : SUCCESS in 21m 14s","accounts_in_message":[],"_revision_number":4},{"id":"93d7d38d40866cc2b48a44ecd4f51d142bdf1f1a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 13:12:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bdecfcc900413efe22e18b135380788ecfb0229d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 15:34:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1325594d2b66478f8bd49453da06f3f2 : SUCCESS in 5m 42s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d57310474ce2432ab2f6c8387d0d6496 : FAILURE in 20m 45s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/bbb82d52ecd04d11902a61d9c91df4dc : FAILURE in 1h 04m 36s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/9396f366559c4791a95f61c1d8465dbf : FAILURE in 49m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35acf9af739d4dfea0d8dc536682506e : FAILURE in 56m 36s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/eb883f2ca6514751ad99b9340fc6970f : SUCCESS in 2h 10m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb545e3433074a399d4b767a5d7207f5 : SUCCESS in 1h 32m 26s","accounts_in_message":[],"_revision_number":5},{"id":"da3f68981f688aa865238a1ee9985f408a9a9eb4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-05 07:19:24.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"af0c61b931c860369716075c3d6a02822473afa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 09:42:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f64a7410bc142ada5abbca06511235c : SUCCESS in 6m 19s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/746ecf7c793344f69fcd094c17fa02f2 : FAILURE in 23m 09s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/68fa35797efb48dd9f461afed1d5f066 : FAILURE in 1h 03m 55s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/385907cd08f6455d88385c289147ea07 : FAILURE in 49m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4349dd271ad54342b44db0f38b5f45aa : FAILURE in 53m 54s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e3f2d949ab4a429caa09e186ed3d0ef0 : SUCCESS in 2h 01m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6863f9b1383f4f1aafd0080aa6667315 : SUCCESS in 1h 28m 45s","accounts_in_message":[],"_revision_number":6},{"id":"c47deb2ec229e344672fe0ad5468c2c8ef199d1b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-05 13:36:27.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f7ef3811c3d86d647ffe9729e36faf204bad2a3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 15:56:58.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72636c3a8cc34b14a090516449217ee5 : SUCCESS in 5m 12s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d63a7d3e184a41d3b965ffd6a9762492 : FAILURE in 14m 07s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/a873e455938b407d99ce68c4b54da4fa : SUCCESS in 1h 14m 16s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e0eb983ca3b14c32875d7f220a83c947 : FAILURE in 51m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6aad3b1d2e740c48166cca99f0e18d3 : SUCCESS in 1h 01m 51s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d1a4156e57af42e7867d1adb77ac33f0 : SUCCESS in 2h 05m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/625c7ba08d8d420fb705dfabf95680a5 : SUCCESS in 1h 36m 41s","accounts_in_message":[],"_revision_number":6},{"id":"dd0f72c17f56042862b64c4e7f26428a5f78408d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-07 09:19:14.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f9e98ac435b0f39764f93588323284444930c603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 11:02:08.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"6cb8c9a5dd58c0e2b2c7a5c51f387ce96c1a56b3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-07 11:02:29.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"d87ba442e7b5ba7e679b6663c10d985d102ac31c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 13:16:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d417eb9425c4576a1b539c443db6720 : SUCCESS in 6m 33s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a18d3b5a71c347508bd5be9587bc836e : FAILURE in 15m 00s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/cb485471d3bc4b258d8f24b4e40bce5e : SUCCESS in 1h 23m 30s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/6bb003828dc7471cb769f2c3285a4f08 : FAILURE in 52m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bedf03a160f40d2a318aac8ece2bd10 : SUCCESS in 1h 13m 27s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a6f3d91bcc5c4127b60702f4425ca5e2 : SUCCESS in 1h 59m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/904d904a82174c62aa93d0e7c4acbbc2 : SUCCESS in 1h 40m 18s","accounts_in_message":[],"_revision_number":7},{"id":"12798752e811e2d87439228fda0be6aa6a771854","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-07 14:46:20.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"ad3092b78200b8586fbbe25a8950960113823045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 16:33:29.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fee11935634b42018088157b4c7a29f4 : SUCCESS in 6m 05s\n- openstack-ansible-linters-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4aa9d7aaa904762b7c9175401a17e40 : FAILURE in 16m 54s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/f76b96d759e84bad9d3b2a1dde68d2b7 : SUCCESS in 1h 19m 29s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/6c714341e8fd4b448d818bc229aeeba9 : SUCCESS in 1h 00m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f9377daa8ea440fad06c14022aae314 : SUCCESS in 52m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0c0112cdd9dc44908e72c1bb22d3ea24 : FAILURE in 1h 07m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67915c2750bb4a1c8fb71787e2be0913 : SUCCESS in 1h 38m 38s","accounts_in_message":[],"_revision_number":7},{"id":"04c70aee40c6c0e574a4ea7b335305bed6d44bff","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-13 13:05:21.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"c40670b5070692aae016278c9a569a5d54d7cbd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 15:29:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1f215a72f6f4f46ae080f7d7cdc265e : SUCCESS in 6m 27s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c440706d0cb4dc49ec92445e12aad88 : SUCCESS in 14m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/2e26dd76b8584c2eb4be6116f196962b : SUCCESS in 1h 17m 26s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/779e61c0602042ffac68c9ad4ed6e6d9 : SUCCESS in 59m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d28496bb8d004a41ace0eeea4ad88d09 : SUCCESS in 1h 02m 05s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b311bad5ab704c808e52a7786ad7fbec : SUCCESS in 2h 05m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8842e5d9029143ca83d4490e309495d8 : SUCCESS in 1h 36m 46s","accounts_in_message":[],"_revision_number":7},{"id":"901543156eb04c24d8655728caab28c3ec42cf02","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-13 16:19:36.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"45728c9a29e1f4e2620d3f18fbc8293de34c5729","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-14 07:14:40.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"023590d794bb416d35d1ba2b33158b13dd4fda79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 07:14:56.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"94d8aa206fb85cdd9539a4c5920f734579e3b6ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 09:32:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"d689bd6f89160b17c1b0687b9a2dd86bf845e966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 09:32:09.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f7885ff0cad4e54ae0b20bf891838e4 : SUCCESS in 6m 38s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66f40fc8d810404fb1f8abb451393ec2 : SUCCESS in 31m 16s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/808ac32987eb483fadfcd7e46d00b314 : SUCCESS in 1h 14m 28s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/9497ec36ca9d4c2ba5fce81b91785634 : SUCCESS in 1h 06m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da28206c619941f6910a236580cd52e4 : SUCCESS in 1h 10m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9d369890606476785b9145f73eeccaf : SUCCESS in 1h 43m 52s","accounts_in_message":[],"_revision_number":7},{"id":"4c9f8c091a6dc039a1dafb89d4a5f329cfd94184","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 09:36:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/281721c53fb34278b45b43a8133dd6d6 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1d0b447f80b636f8a745a0db4a68513477c4ccc1","revisions":{"de7ad6ecf80f7a8bf3cc2a485413299fd910d971":{"kind":"REWORK","_number":1,"created":"2020-10-01 16:25:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/755586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/86/755586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/1"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-01 16:25:21.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-01 16:25:21.000000000","tz":180},"subject":"Use infra jobs instead of functional","message":"Use infra jobs instead of functional\n\nDepends-On: https://review.opendev.org/755497\nChange-Id: Id50807afaafa47bca409dcb109c8327476bf60ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/de7ad6ecf80f7a8bf3cc2a485413299fd910d971"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/de7ad6ecf80f7a8bf3cc2a485413299fd910d971"}]},"branch":"refs/heads/master"},"7fecd10ee58edfe726c11b2519b9ee6863d93893":{"kind":"REWORK","_number":2,"created":"2020-10-02 08:45:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/755586/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/86/755586/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/2"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-01 16:25:21.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-02 08:44:56.000000000","tz":0},"subject":"Use infra jobs instead of functional","message":"Use infra jobs instead of functional\n\nDepends-On: https://review.opendev.org/755497\nChange-Id: Id50807afaafa47bca409dcb109c8327476bf60ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7fecd10ee58edfe726c11b2519b9ee6863d93893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7fecd10ee58edfe726c11b2519b9ee6863d93893"}]},"branch":"refs/heads/master"},"5bb93132c6b6c5c5f5e9f8444916c88cc058dda0":{"kind":"REWORK","_number":3,"created":"2020-10-02 08:47:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/755586/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/86/755586/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/3"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-01 16:25:21.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-02 08:46:55.000000000","tz":180},"subject":"Use infra jobs instead of functional","message":"Use infra jobs instead of functional\n\nDepends-On: https://review.opendev.org/755497\nChange-Id: Id50807afaafa47bca409dcb109c8327476bf60ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5bb93132c6b6c5c5f5e9f8444916c88cc058dda0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5bb93132c6b6c5c5f5e9f8444916c88cc058dda0"}]},"branch":"refs/heads/master"},"def1d978b12fd5b350655d1f176931aea646b428":{"kind":"REWORK","_number":4,"created":"2020-10-02 10:05:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/755586/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/86/755586/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/4"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-01 16:25:21.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-02 10:04:51.000000000","tz":180},"subject":"Use infra jobs instead of functional","message":"Use infra jobs instead of functional\n\nDepends-On: https://review.opendev.org/755497\nChange-Id: Id50807afaafa47bca409dcb109c8327476bf60ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/def1d978b12fd5b350655d1f176931aea646b428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/def1d978b12fd5b350655d1f176931aea646b428"}]},"branch":"refs/heads/master"},"7b84af870b2c3e39b5175508e5db0739aad7e5ff":{"kind":"REWORK","_number":5,"created":"2020-10-02 13:12:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/755586/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/86/755586/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/5"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-01 16:25:21.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-02 13:12:32.000000000","tz":180},"subject":"Use infra jobs instead of functional","message":"Use infra jobs instead of functional\n\nDepends-On: https://review.opendev.org/755497\nChange-Id: Id50807afaafa47bca409dcb109c8327476bf60ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7b84af870b2c3e39b5175508e5db0739aad7e5ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7b84af870b2c3e39b5175508e5db0739aad7e5ff"}]},"branch":"refs/heads/master"},"8e131f4235a34cfeca8eef53c62468a1c94acda3":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-10-05 07:19:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/755586/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/86/755586/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/6"}}},"commit":{"parents":[{"commit":"a7a396dbb8bff2f867c1a1a7d45f7b2e7d86c7f1","subject":"Fix linter errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a7a396dbb8bff2f867c1a1a7d45f7b2e7d86c7f1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-01 16:25:21.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-05 07:19:24.000000000","tz":0},"subject":"Use infra jobs instead of functional","message":"Use infra jobs instead of functional\n\nDepends-On: https://review.opendev.org/755497\nChange-Id: Id50807afaafa47bca409dcb109c8327476bf60ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8e131f4235a34cfeca8eef53c62468a1c94acda3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8e131f4235a34cfeca8eef53c62468a1c94acda3"}]},"branch":"refs/heads/master"},"1d0b447f80b636f8a745a0db4a68513477c4ccc1":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-10-07 11:02:29.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/86/755586/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/86/755586/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/86/755586/7"}}},"commit":{"parents":[{"commit":"0748f362de866aa2a037d6ea8cf6e985e8b1a4b6","subject":"Merge \"Explicitly use mysql collection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0748f362de866aa2a037d6ea8cf6e985e8b1a4b6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-01 16:25:21.000000000","tz":180},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-07 11:02:29.000000000","tz":0},"subject":"Use infra jobs instead of functional","message":"Use infra jobs instead of functional\n\nDepends-On: https://review.opendev.org/755497\nChange-Id: Id50807afaafa47bca409dcb109c8327476bf60ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1d0b447f80b636f8a745a0db4a68513477c4ccc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1d0b447f80b636f8a745a0db4a68513477c4ccc1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
