)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~755594","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ic0e96e219792638b24a8e62517d98e21c3150f46","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"Ic0e96e219792638b24a8e62517d98e21c3150f46","subject":"Explicitly use mysql collection","status":"MERGED","created":"2020-10-01 16:51:10.000000000","updated":"2020-10-06 09:00:16.000000000","submitted":"2020-10-06 08:57:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755594-1601974661139-7963eb8c","meta_rev_id":"20f75fe4002d2fbcc068f7ec2676c4a536702a9e","_number":755594,"virtual_id_number":755594,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 08:57:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-06 07:25:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-05 09:48:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-10-06 07:25:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 09:48:51.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-06 07:25:59.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-06 08:57:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2ed5f170408f25b7778c52f41e0813d4130c545","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-01 16:51:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"138dc4bc81cedd20bbf72983b21075d9d3b87cec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 19:46:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8ddabb4b86044881a16b050f6daf2161 : FAILURE in 7m 49s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7c2e9e1f6a1343e894c8e715fc0be5c9 : SUCCESS in 20m 30s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/c5d063806e4e45f49c3d8e79b1e91eed : FAILURE in 39m 39s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/477c24cbe9ff45efa4528463ddda1999 : SUCCESS in 22m 42s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/598174f47af4456cb0f6d5ddfea0146d : SUCCESS in 21m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9f47f98fc12f4984843ef06d7bfd4a47 : SUCCESS in 1h 12m 31s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f14e04843f4446d0b703023158b2d1ec : SUCCESS in 1h 22m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc5ec9d747e94ffb896029070ada9c92 : SUCCESS in 1h 15m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba02576bed764500a7cfc7932a8fbc91 : SUCCESS in 1h 05m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d49aeed3b7c64872bf369eeac984b770 : SUCCESS in 1h 56m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdf85801458d40f3bf58eeb3beff04ca : SUCCESS in 1h 39m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/449682f5631f42dcb9d9d2f8256f16a3 : SUCCESS in 1h 04m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57ec6acb9d0545e193b515345a5a8ca9 : SUCCESS in 53m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6eb98779e1348099d3a6dda2800eba1 : SUCCESS in 5m 44s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/ab03e8506dfb42edab91c7c9f90a6937 : SUCCESS in 21m 03s","accounts_in_message":[],"_revision_number":1},{"id":"97553c6fb07b189d4d97ad9f67ac61f56cf79f5d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 09:26:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"01b3109502e083ea26a473dce4f72aa134867a1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 11:39:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4e3d32b25f9a402a90a23e663db5a067 : FAILURE in 8m 19s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8bad9776621b4b528b5ca7830d1f216f : SUCCESS in 21m 30s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/0c90c48cbbf24c0799102fdf9f827552 : FAILURE in 39m 12s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60acd3cc2ebd4eb888e81c7405c8e3a0 : SUCCESS in 21m 04s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b8a8dde85da46bebe714464388f14f3 : SUCCESS in 21m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1fe4913640464221b2cca893cdf577d0 : SUCCESS in 1h 08m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4adfc4e1562b4cec8e52fa97a5ee4a99 : SUCCESS in 1h 19m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/885acde14bf947979dc656c6ce91152b : SUCCESS in 1h 08m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d0f83731b484a458f176014248ad47d : SUCCESS in 1h 03m 56s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d4a2c5749ec34347b9b99522354cae30 : SUCCESS in 2h 03m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/404c77dec86d4077aee95e519924eed3 : SUCCESS in 1h 36m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a31d4d5396054d9ca584a13eb8983865 : SUCCESS in 1h 02m 52s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c0c19b659fa428f9e4c67896729742b : SUCCESS in 58m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6849cb277dce466b87a656513d1bd624 : SUCCESS in 5m 11s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/2e9b713b4d634bf396e90e5ff6ec8adb : SUCCESS in 21m 31s","accounts_in_message":[],"_revision_number":2},{"id":"ab6f129bfe8bacc693a3fbf454d01fe479e5a6b9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-05 07:18:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"15bacba303d9d506d153e20a0b3f202587603f42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 09:38:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c80eb4c6b9f84b09b04a92979b28833c : SUCCESS in 10m 24s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8592e3e5e9a74497b4571bd357f79a24 : SUCCESS in 22m 18s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/adf2da39d3994e8fb705aa6c7a037faf : FAILURE in 55m 18s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ec6765bbe0a40a5a2c4d822b2985fe0 : SUCCESS in 20m 43s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6afe900e1d7d46139d96610af8933e31 : SUCCESS in 22m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/38f3e893c8954158bad2d50265126991 : SUCCESS in 1h 03m 47s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/683dc9d5f4ad47a8b23b99ac459400a4 : SUCCESS in 1h 22m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3203088df90c40b5b36910612c6cba4b : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b79ca11419b4024b7ecb6aefb3b9c05 : SUCCESS in 1h 05m 27s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3acc8bef149f44d8968ed8237b63038b : SUCCESS in 2h 07m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83b67e0d88224e89969fab1c1e485b3f : SUCCESS in 1h 36m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e4d55c52f28149eaa3f36927de22e90b : SUCCESS in 1h 08m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0ba2526fb82448d92d18575d48f5aec : SUCCESS in 58m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ecad6ca2e0a4e5f9b006d431769c1d9 : SUCCESS in 6m 15s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/dcdb99eb8f064fdb8f47eebf77d6c027 : SUCCESS in 21m 47s","accounts_in_message":[],"_revision_number":3},{"id":"450feb7fbeecbb7b9983624e7b92a711a99806ea","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-05 09:48:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"91014fe7c2526b8a0e9affbf190f57389578f18b","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-06 07:25:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"25a19d7dd07d8d1a5308787912e9dcf08d6aeeb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 07:26:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1f02bc9115437c470f26e001792f0879cebce140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 08:57:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/106fc384ace1489f8812ec08261ad30c : SUCCESS in 8m 59s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/181eecd0db664c3a8466eb53c4ffd9bf : SUCCESS in 20m 39s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65f7a0b5af4044409ebb884605fc7f28 : SUCCESS in 21m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8197a09de86f485b939ff64346822f2c : SUCCESS in 53m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/091474adac074770932ed2248224ba89 : SUCCESS in 1h 20m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/987b36bd029248a9951c34e819903b9f : SUCCESS in 1h 13m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a95e6e3b07a84c0e9213fa8c490174a2 : SUCCESS in 1h 02m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d5d15d8d540a485a85d3bc0b38e0ad1e : SUCCESS in 1h 02m 59s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de0308441eb04323baff046485eb4dfb : SUCCESS in 54m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5af41dab4625455ba3230fc521ae296c : SUCCESS in 4m 36s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/f1ea2bc84ad04adcbdfda6acb2769266 : SUCCESS in 24m 23s","accounts_in_message":[],"_revision_number":3},{"id":"16a6193ffcea69b88f97a5f9742dcde96d5d4c9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 08:57:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"20f75fe4002d2fbcc068f7ec2676c4a536702a9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 09:00:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d767674a49b4b01b6d3ef1346faf6be : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5391e63c7810f3379f93912d0dc59e9b88c6a633","revisions":{"9d6d6e916251544791bd8027a9be85731ebff68e":{"kind":"REWORK","_number":1,"created":"2020-10-01 16:51:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/755594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/94/755594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/94/755594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/94/755594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/94/755594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/94/755594/1"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-01 16:50:58.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-01 16:50:58.000000000","tz":180},"subject":"Explicitly use mysql collection","message":"Explicitly use mysql collection\n\nSince we\u0027ve switched to ansible 2.10 we should explicitly use modules\nprovided by mysql collection.\n\nChange-Id: Ic0e96e219792638b24a8e62517d98e21c3150f46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9d6d6e916251544791bd8027a9be85731ebff68e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9d6d6e916251544791bd8027a9be85731ebff68e"}]},"branch":"refs/heads/master"},"35005020587ce67255a77a3879b5b9f4ffce6253":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-02 09:26:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/755594/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/94/755594/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/94/755594/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/94/755594/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/94/755594/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/94/755594/2"}}},"commit":{"parents":[{"commit":"3c62c136d3ca2cff0eb16a00e62d711858a324a9","subject":"Fix linter errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3c62c136d3ca2cff0eb16a00e62d711858a324a9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-01 16:50:58.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-02 09:26:52.000000000","tz":0},"subject":"Explicitly use mysql collection","message":"Explicitly use mysql collection\n\nSince we\u0027ve switched to ansible 2.10 we should explicitly use modules\nprovided by mysql collection.\n\nChange-Id: Ic0e96e219792638b24a8e62517d98e21c3150f46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/35005020587ce67255a77a3879b5b9f4ffce6253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/35005020587ce67255a77a3879b5b9f4ffce6253"}]},"branch":"refs/heads/master"},"5391e63c7810f3379f93912d0dc59e9b88c6a633":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-10-05 07:18:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/755594/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/94/755594/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/94/755594/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/94/755594/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/94/755594/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/94/755594/3"}}},"commit":{"parents":[{"commit":"a7a396dbb8bff2f867c1a1a7d45f7b2e7d86c7f1","subject":"Fix linter errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a7a396dbb8bff2f867c1a1a7d45f7b2e7d86c7f1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-01 16:50:58.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-10-05 07:18:37.000000000","tz":0},"subject":"Explicitly use mysql collection","message":"Explicitly use mysql collection\n\nSince we\u0027ve switched to ansible 2.10 we should explicitly use modules\nprovided by mysql collection.\n\nChange-Id: Ic0e96e219792638b24a8e62517d98e21c3150f46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5391e63c7810f3379f93912d0dc59e9b88c6a633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5391e63c7810f3379f93912d0dc59e9b88c6a633"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
