)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~755698","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Iceda4b23bf5347a7de9817d212ca8db8f2befc10","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"osa-linters-update","hashtags":[],"change_id":"Iceda4b23bf5347a7de9817d212ca8db8f2befc10","subject":"Fix linter errors","status":"MERGED","created":"2020-10-02 06:29:43.000000000","updated":"2020-10-05 09:01:08.000000000","submitted":"2020-10-05 08:59:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755698-1601888387513-008d8d24","meta_rev_id":"bd2d46fcf9b8fdd796b7dc63b5491dacfd431540","_number":755698,"virtual_id_number":755698,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-05 08:59:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-05 07:29:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2020-10-05 07:18:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-10-05 07:29:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 07:18:13.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-05 07:29:32.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-05 08:59:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bed763dbca78c4fbb0123527ce8b4107c808301b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-02 06:29:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27bb49a647fc5b68a9127764c011ff63109f75b5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-02 06:42:35.000000000","message":"Topic set to osa-linters-update","accounts_in_message":[],"_revision_number":1},{"id":"f13be7ad2e1f517716c2b57c98399072d22e69b2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 06:47:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbc69769b21e04290c28db03d7b6a98316a26591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 07:03:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d882d8e1302e4e6085dec45af28c8ab4 : FAILURE in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0479c18e0a4f457a9e31ca1bc67cd729 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"929d93c40592a43c7091d569bae06c8ed7ab56a2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-02 07:06:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ebbb894a5a19dfb6a4edea2c4dff5ce131efd404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 09:22:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e131ce9a0022479f888b9acee217e399 : FAILURE in 9m 59s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ffd19d76b23c429cbafc0d28443be22c : SUCCESS in 26m 13s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/1e676411a1ef4b5bba1011f964ecc71d : FAILURE in 56m 20s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abb4558f949c49d5b5f384eea34c3eb5 : SUCCESS in 20m 28s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0cc9d951a50484698110468754fb394 : SUCCESS in 21m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/528e1ac0910b4ec4b778ffecff8b5f7c : SUCCESS in 1h 10m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c7c966470d904e5ab4b98850eacc3817 : SUCCESS in 1h 20m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b40917bb16342a2bad6fe38bbff4fd2 : SUCCESS in 1h 12m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc8bcbded145468199506f424abcf9cb : SUCCESS in 1h 04m 59s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1ee78b9440e348e2a80521c573da022a : SUCCESS in 2h 01m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b322688d6cee4a2290fdb14d1c8a1742 : SUCCESS in 1h 37m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/26789a680b434d2b9d6400f4a7e43194 : SUCCESS in 1h 02m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a593028fbe9429cb6af68f7ae421ac2 : SUCCESS in 55m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/514b11658f3242bba97b9d2dd17ad23e : SUCCESS in 5m 19s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/30d701642bde43c4898d5d65cf043bfc : SUCCESS in 26m 05s","accounts_in_message":[],"_revision_number":2},{"id":"a59910da2056dcd50e5f1fbff8bcc0875a1099ab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 09:26:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"46e0b73488eabb16d02bc52e71fcc3d30fa439c5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 10:35:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fa0b98896a3b7728d5af403b72b4f2d9eb26458d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 11:43:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/139c7bbf52bb4686aa8862547337d09b : FAILURE in 8m 07s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c386d7fdc11c43dc9e0f6b8551c72e48 : SUCCESS in 20m 37s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/4847765b80d348bb883dfdb750f68875 : FAILURE in 38m 32s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4902226ec2f43f8a6b3c1224415582b : SUCCESS in 21m 20s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b71981e130e144fabec5b7d8fd36685e : SUCCESS in 22m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e48ae0853e964016b6fd582ec79022b0 : SUCCESS in 59m 08s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9c5da9f7da1847c9a91b087e66a2815d : SUCCESS in 1h 40m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9162572ba66f40d0835e5c610e06945e : SUCCESS in 1h 18m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34575a7309eb41159ceb6754c14fad31 : SUCCESS in 1h 12m 59s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/46e0a6a41a644d09accb6c7c3ca83212 : SUCCESS in 2h 04m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cf2e053237740f1b664bfec5f7c9b68 : SUCCESS in 1h 30m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/75361509775e4dde95fbe25f287b3371 : SUCCESS in 1h 08m 10s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d91bbb9c1b8d4a1091f98719c54b845d : SUCCESS in 50m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f5f4fb9d5c64646bdb41e5d41891b1a : SUCCESS in 5m 47s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/17dad8a4430943e6aeaae49565728a68 : SUCCESS in 23m 16s","accounts_in_message":[],"_revision_number":2},{"id":"691df39c22b9245b58d25f9206b97e33bfde6993","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-02 15:00:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a781694f5cdce6d7e4a059cee43f3ddb3e6bc1ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 17:08:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6766e1c03d8a491fa04e772c0f287499 : SUCCESS in 10m 55s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/89602643fff64cb0a7ff634e6d5fd592 : SUCCESS in 23m 12s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/b3cfcaa8f78d4ffba3bff50ded8327a8 : FAILURE in 36m 03s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60c86ddf6cac48a586f6b90be2f18fbb : SUCCESS in 26m 25s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80d58eda507749a1b5bfe7916769cc9e : SUCCESS in 20m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7400ae9c604c4f61b94b127b137f2ab1 : SUCCESS in 1h 07m 05s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/89670392aa0b4f81b3cb1ba59681cb42 : SUCCESS in 1h 31m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efda0964c61144599bacc4a3aee0da6f : SUCCESS in 1h 10m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d9001563a7d4a88ae14b3e461471f21 : SUCCESS in 1h 05m 11s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ea17c6d21534449fa3268ece2beec1d9 : SUCCESS in 1h 59m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77dd02393ff94c71be94bd94e0e95fac : SUCCESS in 1h 43m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7ee08d42a7a84668bb1429cd2f25f962 : SUCCESS in 1h 00m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec2b3ba8594c43ab917ffce337330b4e : SUCCESS in 56m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c7902326c744a17aa1411ff2289e10a : SUCCESS in 4m 51s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/4603dc3ec1d04582aa558e7be8449d16 : SUCCESS in 22m 56s","accounts_in_message":[],"_revision_number":3},{"id":"439409c78ebe6173d5d62db7b10cb827f1282f40","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-05 07:18:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d94484e306e0410aa888700b8c8e004839fd42f9","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-05 07:29:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a8e73c1915dbf96a2c179ca5fb60cf470094a915","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 07:29:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ff26f6d570269ae941ca7247c6b7357c78326d41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 08:59:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"dbf0eb5f55c9b533a2fa3234726388831642893e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 08:59:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/305d2ff1aed64049a26ca045ed8ab6db : SUCCESS in 9m 07s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/f17ac9b1069a4bebb9d884a9db255a4a : SUCCESS in 21m 05s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15cb7fafaa0d43dea3ecd889a64c75f1 : SUCCESS in 20m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8b388ea1334a4248801a34349878b6cd : SUCCESS in 1h 04m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5479fc5ff7bc48a0a359c1c162d9d691 : SUCCESS in 1h 17m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4e94dfca0374dbc9cec24478b265b95 : SUCCESS in 1h 11m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/946afb1dcc594f45a27d0715ec4722d4 : SUCCESS in 1h 10m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8d6cc22b5c4a4838aa5fe831fd008f40 : SUCCESS in 1h 04m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a2c4372d6cb4f3ca80108526c3ba5f2 : SUCCESS in 57m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520012498f2d43faaf7f876ccce951ad : SUCCESS in 5m 23s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/256f583145cb4dacb76eb321247e3d05 : SUCCESS in 24m 10s","accounts_in_message":[],"_revision_number":3},{"id":"bd2d46fcf9b8fdd796b7dc63b5491dacfd431540","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 09:01:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5399214a513a4a369739621248e89156 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a7a396dbb8bff2f867c1a1a7d45f7b2e7d86c7f1","revisions":{"1536e8b4fc905af460a54756ede377b1308baaa6":{"kind":"REWORK","_number":1,"created":"2020-10-02 06:29:43.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/98/755698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/98/755698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/755698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/755698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/755698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/755698/1"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-02 06:29:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-02 06:29:21.000000000","tz":60},"subject":"Fix linter errors","message":"Fix linter errors\n\nChange-Id: Iceda4b23bf5347a7de9817d212ca8db8f2befc10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1536e8b4fc905af460a54756ede377b1308baaa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1536e8b4fc905af460a54756ede377b1308baaa6"}]},"branch":"refs/heads/master"},"3c62c136d3ca2cff0eb16a00e62d711858a324a9":{"kind":"REWORK","_number":2,"created":"2020-10-02 07:06:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/98/755698/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/98/755698/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/755698/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/755698/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/755698/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/755698/2"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-02 06:29:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-02 07:06:01.000000000","tz":60},"subject":"Fix linter errors","message":"Fix linter errors\n\nChange-Id: Iceda4b23bf5347a7de9817d212ca8db8f2befc10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3c62c136d3ca2cff0eb16a00e62d711858a324a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3c62c136d3ca2cff0eb16a00e62d711858a324a9"}]},"branch":"refs/heads/master"},"a7a396dbb8bff2f867c1a1a7d45f7b2e7d86c7f1":{"kind":"REWORK","_number":3,"created":"2020-10-02 15:00:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/98/755698/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/98/755698/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/755698/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/755698/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/755698/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/98/755698/3"}}},"commit":{"parents":[{"commit":"e70e392b33f39f756884c82a11093c73134132ad","subject":"Revert \"Improve compatibility with connection delegation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e70e392b33f39f756884c82a11093c73134132ad"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-02 06:29:21.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-02 15:00:25.000000000","tz":180},"subject":"Fix linter errors","message":"Fix linter errors\n\nChange-Id: Iceda4b23bf5347a7de9817d212ca8db8f2befc10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a7a396dbb8bff2f867c1a1a7d45f7b2e7d86c7f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a7a396dbb8bff2f867c1a1a7d45f7b2e7d86c7f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
