)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~758462","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fstein~I543f1f5847fa65b1523ecc21261a40ecaf8158ce","project":"openstack/openstack-ansible-galera_server","branch":"stable/stein","hashtags":[],"change_id":"I543f1f5847fa65b1523ecc21261a40ecaf8158ce","subject":"Bump galera version","status":"MERGED","created":"2020-10-15 16:30:31.000000000","updated":"2020-10-30 13:43:26.000000000","submitted":"2020-10-30 13:38:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758462-1604065113516-8d692ead","meta_rev_id":"b4b4cb85027eaec15f5ea55724bc5bd3f06edbc4","_number":758462,"virtual_id_number":758462,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-30 13:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-28 15:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2020-10-30 13:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2020-10-30 13:08:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 15:07:48.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-30 13:38:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0e23d064b3db7a1b3b77853859297075e031805","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-15 16:30:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"813b2b86cf8a677f3493fc08746dacd088b1c727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 17:27:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/28a001f931df4f0685400de86d65afbc : SUCCESS in 6m 19s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/99a93be14ac24cbc872f7c6f581bd70b : SUCCESS in 15m 41s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/d488da0c3d974de0b0116babe722103c : FAILURE in 37m 49s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47db6ba52a1e49f2ac3b67dcdc762574 : SUCCESS in 17m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68625883632b4ee49f133bb8bb9c31aa : SUCCESS in 5m 17s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/2ea7a0013c79426ea4c28040a10cbeaa : FAILURE in 34m 15s\n- openstack-ansible-functional-ubuntu-xenial-ssl https://zuul.opendev.org/t/openstack/build/0484d955500747e196cace430308a288 : SUCCESS in 16m 00s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/a5edadd8c6f2430193346a70f2a637d8 : FAILURE in 10m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e19760a28d6757bce494cad7f47afd4f8ab7dc8d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-16 09:49:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d580f2e108989c1dc47abfbeec0d76b9d4abe073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 10:29:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/01edd066199e489ba31b001d6a64f16b : SUCCESS in 7m 05s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/5d866dd81ea14637a25be6ee7ad6bfaa : SUCCESS in 14m 35s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e6d2edc02cef443eb3b9f7e4f4c90aea : FAILURE in 33m 30s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34f569c273ac4dc7b7fc04a9d6af86ab : SUCCESS in 23m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56124f74fe9d4c2bb1bd9f25224cdda1 : SUCCESS in 4m 56s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/38deeb65e0074eb1a23b254f7fcee4ca : FAILURE in 34m 12s\n- openstack-ansible-functional-ubuntu-xenial-ssl https://zuul.opendev.org/t/openstack/build/4aec9094bebb4630beaeb7637849942a : SUCCESS in 16m 36s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/d15c6b1ac525411bb53b0f92f920d7c6 : SUCCESS in 21m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e36848ac0ca1d114b81178db9ae681da1c500f17","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-18 21:35:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f4a9d5f991d4e89316ba0e6f92fafc369a69a8d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 22:18:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d88a6f35dff540b6b39efb47a46ebdfd : SUCCESS in 7m 23s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/cc6f83aa2ed24649a95888a40a4969e0 : SUCCESS in 18m 28s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1dd3ccad00cb4c2095f2e5295071ac0e : FAILURE in 36m 42s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e68ddd5f354b4e2d8dcf580d389f58ac : SUCCESS in 18m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6fc4d250746480c82cf5a3361ce0321 : SUCCESS in 5m 17s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/a9297f2566d44e2689f47f14ff9744df : FAILURE in 35m 52s\n- openstack-ansible-functional-ubuntu-xenial-ssl https://zuul.opendev.org/t/openstack/build/e6d8aa6194c8483fa43ca5f9e74afae4 : SUCCESS in 17m 45s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/052d5621fe904803807720880597b74a : SUCCESS in 19m 46s","accounts_in_message":[],"_revision_number":1},{"id":"de9f0b9036e420ff57ff6b8a88a7deeec9278e2f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-21 07:37:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9281b120ff4e0eb1e82ebdefbe67e3a0ff87152e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 08:30:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0db7ff43ad8642bfa954dab1615a923e : SUCCESS in 10m 00s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/26e33049757e46028d6c9c88db096993 : SUCCESS in 19m 09s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/da821e9fb08d4d34817838b9c575f128 : FAILURE in 37m 06s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/124b14fbf539404ba37d11daa937464a : FAILURE in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d42f6a9daf894dbcb06ebb4dbb1a4658 : SUCCESS in 6m 04s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/f92bc72133114018869e1cc56b8da655 : FAILURE in 34m 06s\n- openstack-ansible-functional-ubuntu-xenial-ssl https://zuul.opendev.org/t/openstack/build/c8384a2adeeb486fadc3745b49383330 : FAILURE in 9m 42s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/d937f6743d6646a49dcaf7a4616deb36 : FAILURE in 8m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e940d9ebc4462ea9a1ff2adde902ea39a1cd451b","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-28 09:02:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5538f23c20a8b3725e0f10d77f7d6a362a4be8aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 09:46:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5a235e0a3eaa40f39c44a363f644d98d : SUCCESS in 6m 48s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/8a9a12a9eec1424fb3b2082b51e4b02f : SUCCESS in 15m 28s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b568a155f4e240909e25a85aef00dd8f : FAILURE in 33m 46s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bacc456cc01433cb3c938495e62de9f : SUCCESS in 17m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/177607878785493f81defddc60a05bf1 : SUCCESS in 3m 42s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/f1b590c164bf48d7a9e91c7bdf20738c : FAILURE in 34m 22s\n- openstack-ansible-functional-ubuntu-xenial-ssl https://zuul.opendev.org/t/openstack/build/1ae9b461be3d4c8d9c5101bc8f4974f0 : SUCCESS in 17m 09s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/3cfada9df2c9477fbbccc17641870784 : SUCCESS in 15m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6a2d91b99e1e5683880507f058697236a774c93e","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-28 11:04:36.000000000","message":"Patch Set 1:\n\nLooks like the cluster bootstrapping script isn\u0027t working properly on CentOS but I\u0027ve yet to figure out why.","accounts_in_message":[],"_revision_number":1},{"id":"646b50f277e7c84d71c3ca8e4a3a8d628d97f662","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-28 13:38:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bdedfe7e15b292c04e2abc9d967d108948b9f63f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 15:07:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c7cfe73f5fef4c868a81c66e202e40d8 : SUCCESS in 7m 46s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/8ee26bea3700434e8fd3a09e02dfa0ed : SUCCESS in 16m 58s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/67583444b5254a959dadb4048a03776d : SUCCESS in 19m 53s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/edfc7acba194420a8504bbf77c1f9494 : SUCCESS in 16m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15647059868942d4a1ae92c94c735d1b : SUCCESS in 3m 50s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/c47d3e2732d74bdb836b69b2e5c6cd73 : SUCCESS in 19m 01s\n- openstack-ansible-functional-ubuntu-xenial-ssl https://zuul.opendev.org/t/openstack/build/251d0695809e43fdbc92d6d78d54612f : SUCCESS in 24m 54s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/4487ada2dd0a4975bc8d67c0e0baa5ec : SUCCESS in 24m 42s","accounts_in_message":[],"_revision_number":2},{"id":"6b0bc5f64a60780a9f37c0c099c931cfd4c3a2c0","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-28 15:07:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aef866871df100dcec73b1a909c525ca41da1075","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-30 13:08:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"331ab89febf462372d72144f031085380a92f2bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 13:09:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1ca2633cffc822e3b8daa3f482ebbf9a61cc7fbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 13:38:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1a2adc2da5fc4ea29c6bc5a2939040ae : SUCCESS in 5m 14s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/b45c7b616446466194e3521dfd23379a : SUCCESS in 14m 29s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/be2ce3a32a104f6080e8e5e468a29194 : SUCCESS in 17m 47s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/deccb9b3d6d142ee84b8356610c8a012 : SUCCESS in 16m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df432bc0e1147fbba8c419ee9a966c9 : SUCCESS in 3m 47s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/8b61f361b39b40ab962358c6aa851c4b : SUCCESS in 25m 09s\n- openstack-ansible-functional-ubuntu-xenial-ssl https://zuul.opendev.org/t/openstack/build/9d3c8bcdc4854021ae20cf63ae8930ac : SUCCESS in 23m 25s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/3b8b5f6669bf40ee80ed02b6fd4a108c : SUCCESS in 19m 22s","accounts_in_message":[],"_revision_number":2},{"id":"ff4cbc9f56927c7e902971c3fdd9dc168755c34e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 13:38:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b4b4cb85027eaec15f5ea55724bc5bd3f06edbc4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 13:43:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08359892e9344c7be9fdbd335a58544 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"342cfc0c221a595ddff930460e725d8c493e6fa4","revisions":{"ac3e832481bd50c951715dd108c1e2f4293886b4":{"kind":"REWORK","_number":1,"created":"2020-10-15 16:30:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/758462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/62/758462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/62/758462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/62/758462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/62/758462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/62/758462/1"}}},"commit":{"parents":[{"commit":"fcba2b150c72dcacd0fba7c2391c3d54cc515601","subject":"Move to PyMySQL instead of MySQLDB for ansible-2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fcba2b150c72dcacd0fba7c2391c3d54cc515601"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-15 16:30:21.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-15 16:30:21.000000000","tz":180},"subject":"Bump galera version","message":"Bump galera version\n\nWe update galera minor version to cover CVE [1]\n\n[1] http://cve.mitre.org/cgi-bin/cvename.cgi?name\u003dCVE-2020-2574\n\nChange-Id: I543f1f5847fa65b1523ecc21261a40ecaf8158ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ac3e832481bd50c951715dd108c1e2f4293886b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ac3e832481bd50c951715dd108c1e2f4293886b4"}]},"branch":"refs/heads/stable/stein"},"342cfc0c221a595ddff930460e725d8c493e6fa4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-10-28 13:38:31.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/62/758462/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/62/758462/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/62/758462/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/62/758462/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/62/758462/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/62/758462/2"}}},"commit":{"parents":[{"commit":"fcba2b150c72dcacd0fba7c2391c3d54cc515601","subject":"Move to PyMySQL instead of MySQLDB for ansible-2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fcba2b150c72dcacd0fba7c2391c3d54cc515601"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-10-15 16:30:21.000000000","tz":180},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2020-10-28 13:37:49.000000000","tz":0},"subject":"Bump galera version","message":"Bump galera version\n\nWe update galera minor version to cover CVE [1]\n\n[1] http://cve.mitre.org/cgi-bin/cvename.cgi?name\u003dCVE-2020-2574\n\nDepends-On: https://review.opendev.org/760135/\n\nChange-Id: I543f1f5847fa65b1523ecc21261a40ecaf8158ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/342cfc0c221a595ddff930460e725d8c493e6fa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/342cfc0c221a595ddff930460e725d8c493e6fa4"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
