)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~772550","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fvictoria~Ie2a802ebb8297bed03d74b3cf54907322b858896","project":"openstack/openstack-ansible-galera_server","branch":"stable/victoria","hashtags":[],"change_id":"Ie2a802ebb8297bed03d74b3cf54907322b858896","subject":"Bring db setup vars in line with other roles","status":"MERGED","created":"2021-01-27 08:51:21.000000000","updated":"2021-02-02 16:50:40.000000000","submitted":"2021-02-02 16:48:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772550","cherry_pick_of_change":771652,"cherry_pick_of_patch_set":1,"meta_rev_id":"d0adc2dc794a82836353a9ef946662aab92d0039","_number":772550,"virtual_id_number":772550,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-02 16:48:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-02-02 08:54:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2021-02-02 08:54:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-27 08:51:21.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2021-01-27 10:51:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-27 12:03:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-01-28 16:24:53.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"ef749bfbfa7e91da64487f7d2b9245b8297b9516","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-27 08:51:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8a2f6c27f5ca4996f4cba64932aa0eb11cb920ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 10:51:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc34b699f214a8d98cec888c5d852d3 : SUCCESS in 5m 24s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/857306a430c446dda3e5a379c315f56b : SUCCESS in 19m 56s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/5dd0c6b1f38c43dabd65f2eb10eff5d3 : SUCCESS in 1h 35m 44s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/cd7b68e7c87f4663bd958642bd21d30e : SUCCESS in 1h 11m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a96f078137d04b85a53e5519c0cc8099 : SUCCESS in 54m 33s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8745bd4046ff409fb96093d08327c054 : SUCCESS in 1h 55m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d17afe3706134118905fa2d7a7918d40 : SUCCESS in 1h 36m 08s","accounts_in_message":[],"_revision_number":1},{"id":"510d14156404a0efe0a74cfe30b4cfc2f62ef264","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-27 12:03:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"308166b797120b93562e2860409a119bc06741aa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-28 16:24:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"432ec2fcbb78e2b330076e0e70b1c673cc6c8f46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 16:25:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"069aa45cdd4e31da168818f8242f29b6e969168a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 18:23:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5494d842196a44b3aecdfc67d2f9182e : SUCCESS in 6m 29s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b95eebcbcfb4557aefc32929fd32bec : SUCCESS in 21m 37s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/bbd586ba5e094ea79cb610414991b892 : SUCCESS in 1h 48m 16s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/8a98563ab0c743e0b8cd1164261c8c49 : SUCCESS in 1h 06m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3eb84fb063145c4b506958019325cf5 : SUCCESS in 56m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60d179e92a564b96b1553fb4e864dade : FAILURE in 1h 16m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3d3f24811bf30a0d129d304bb49f0c2564f1e8c1","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-28 18:45:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d4b5eee182989595bfb357d2c66fe2fe249a94d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-28 18:52:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c425ef6abc26a7d103914a4bfd53fe8a195a6f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 20:55:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3b3ee3824f24afa893e739580eb3dc3 : SUCCESS in 7m 44s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0c8e2078f604a46bc647e9832b9d8a0 : SUCCESS in 19m 12s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/5392db2fa2d04eff8596b1805e67346e : SUCCESS in 1h 16m 59s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/58445916e4a140fe9931cbb04bd931dc : SUCCESS in 1h 02m 38s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6476094ac7d54f10bc956e836c2edcd8 : SUCCESS in 52m 31s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e9333faa8423485f9ff350a8af9d4773 : SUCCESS in 1h 59m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ddfe775c9b845738dc280b10a890aae : SUCCESS in 1h 38m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5ffa4798d1f0c498f718389881918e00e8c3ce06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 20:55:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ac69b91a56c668997b43964a6df928cd8fa9400","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 22:20:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43fd864fc5784b1d976e6283a12c0540 : SUCCESS in 4m 53s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48ed18bb20fe4d6f98c384038279ff9e : SUCCESS in 14m 12s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/614f5a7b988941f4ae9896931d122a4b : SUCCESS in 1h 08m 30s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/ec1e68151c8d472d9254ca50f8963b36 : SUCCESS in 1h 13m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/443bc84145914e63a34850df0114b5a8 : SUCCESS in 1h 05m 13s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecbd37c5736a4f048cfda23b1f06654e : FAILURE in 49m 20s","accounts_in_message":[],"_revision_number":1},{"id":"172b98367b86782c0689aecad5ef42ce4c599a58","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-29 10:33:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b951e88d3e2458b1a3e82b3fcbb27b2ae0bf63e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 11:48:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e739b0aa7c074d39b4358f4402beada4 : SUCCESS in 5m 24s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/597e085c1b34428ea9d70dbea2430621 : RETRY_LIMIT in 9m 58s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/e1495921f1644e95b3a0f81cfe00c783 : RETRY_LIMIT in 11m 35s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/f15bc8665ae944aba3e3401df69998c5 : RETRY_LIMIT in 7m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a49f3855dfb04d458daafac84691bf68 : RETRY_LIMIT in 15m 56s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/df34fc1794d7450aa00db4d8c63064e6 : FAILURE in 1h 07m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5e08c0e646b4e6f9ddd4a3e8db74e30 : FAILURE in 56m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d2fa82824e3c8edfb178adeb86f0774f31786a57","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-29 11:52:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e15c11ea4418e23f0e0a1352f0b9a22c7606a48d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 13:32:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/029788ee8bae49e5b3d21ee4b59a843c : SUCCESS in 6m 20s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22b03420649e4375827ef6202e531f5c : RETRY_LIMIT in 55m 06s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/75ab9f7364a84941949617ca8637847e : RETRY_LIMIT in 10m 57s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/2ab7f4a7a1dc41cdbbd06df9c4a62536 : RETRY_LIMIT in 9m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a134b4b263ea48429db08306e0a8b3db : RETRY_LIMIT in 8m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c59ea4b59ea0432eab571ee7b128848a : FAILURE in 1h 13m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/234aaae779cb4656b18784f1a3f94cf6 : FAILURE in 54m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fa23b0abc072f100c111722e47a01bb303011897","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-29 17:39:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fce855f5f2f98d0bee3833868a92b1eb46e23a29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 19:00:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64d3cfdc09954f219a20e0e71bde08a7 : SUCCESS in 5m 08s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ac4a534d4e849de9fa6eaa3446b5736 : RETRY_LIMIT in 16m 42s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/ccb4d783c51f4b35965d78e24c405819 : RETRY_LIMIT in 11m 33s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e48843ff888242919497578d23bf0f82 : RETRY_LIMIT in 9m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e949766aa224726a1d50adbaa265248 : RETRY_LIMIT in 9m 52s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b3835eb4b3e74e09b89794e4e80deb86 : FAILURE in 1h 08m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26801316e62a4ba8bcaaa1b2cd045fa3 : FAILURE in 1h 06m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d8c22217c122355ed75555edf18e16e439721401","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-01 08:24:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"80d64beb0f7998849591c407baf5a3d44d056e04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 09:41:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ddaf587938647a5b8c552babd377a64 : SUCCESS in 4m 53s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f61c63739d0437b968a8e21846c02f9 : RETRY_LIMIT in 16m 34s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/2ebb4cc555c6410aa8288f1af2cf5bda : RETRY_LIMIT in 11m 38s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/578edf9faa5b47e6a522abcfb3abdbd8 : RETRY_LIMIT in 11m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23aab50221c3473bbe0594294174b68f : RETRY_LIMIT in 10m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8adb43bfa3414774b98feaada2616e2a : FAILURE in 54m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecf1a1dc44974b478c7928b9203db8b9 : FAILURE in 50m 25s","accounts_in_message":[],"_revision_number":1},{"id":"47d528c7cea9ce67e0d12c2d8f697bf05dd9f83c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-01 10:21:36.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e3ca7c2b77556f1d96a02af8551a1b788d4072aa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-01 10:22:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4308366a635eba57ce66eda11d4902ad932f9cfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 12:27:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20fb33b301ca4b11a1888ff45e98c3d9 : SUCCESS in 5m 07s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c35c8f124944ba1b2a9fed491961ab3 : SUCCESS in 13m 13s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/97517cf1c3ae4507a7369912a788b2ee : SUCCESS in 1h 08m 50s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/057a6c55e0f54ad7bc1fb586b7ca16cc : SUCCESS in 59m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/080fc793484f4d90b7bd41501c2a363a : SUCCESS in 1h 01m 53s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6e68419aded948059d00dc81b564ebcf : SUCCESS in 1h 57m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffd29ed13c07465a86f5299c835a513d : SUCCESS in 1h 25m 12s","accounts_in_message":[],"_revision_number":2},{"id":"9aa83d1562ca2d3c83ad0fc903ba05bd82972ac0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-01 13:23:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1bb3111acda720b8e37f62be0e4ddb523c5a17f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-02 08:51:21.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"373a1ce1d643cac8c5f53884a803bfb6968f2760","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-02 08:54:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nre-apply votes","accounts_in_message":[],"_revision_number":3},{"id":"c0456b274ff3643a8a76421fb32a1925be993a58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 10:58:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab4487ec7df34610a6575c9074b81b67 : SUCCESS in 5m 59s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90c4391841e74d4fbb95936df2bc3095 : SUCCESS in 14m 53s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/08aa80b93adb4bcdb0ad5a299a60ff62 : SUCCESS in 1h 17m 26s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/9e2d6daae10941af9cafd1ae9c929476 : SUCCESS in 1h 14m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f3b517af9a742608f270d6ae385a05a : SUCCESS in 1h 11m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6f2b3231aa9c469b9d1ea98e48b047e1 : SUCCESS in 1h 56m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7ed1946ec3a468280684889d041085c : SUCCESS in 1h 32m 42s","accounts_in_message":[],"_revision_number":3},{"id":"2caaa5f73ae6b677bf4df77c0b071c208ee81fbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 10:58:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"116e2117720352ac0d6daeef69fc1c0906fa0cec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 12:37:57.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df3dd8bca5e54e06a5b64ba4020b6356 : SUCCESS in 5m 11s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6dd23aa5b7344d3b16f99c21bf5aff5 : SUCCESS in 16m 00s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/3c73b5f626b8473fb426e2e921a90bb5 : SUCCESS in 1h 05m 35s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/ca267b4799c1433c809449c6b86bc1e3 : FAILURE in 32m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f19b5cde59247668a36def8d7ca1c42 : SUCCESS in 1h 07m 10s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c783c0329e12475f954dce562c49c43d : SUCCESS in 1h 32m 17s","accounts_in_message":[],"_revision_number":3},{"id":"0c11248e4fe44f221cfdb5245c7ed1b1b58824d1","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-02 12:41:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"603d332452c3a79e10020a2f331b96f450a0c3c7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-02 14:36:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6f292d9122c667a11e69218eccdf1c215e7203b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 15:02:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ffec6e5a3043759b3124d899646e26 : SUCCESS in 6m 41s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7181b669cb4248969e7d5880210618fa : SUCCESS in 15m 37s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/274b14b13b584f3aae4756b0f83ebc3f : SUCCESS in 1h 28m 00s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/5213f708af034b12a33ea9cee70aabb4 : SUCCESS in 1h 03m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59ed3624399a4b2f98ad0a01d9ac0c0b : SUCCESS in 54m 35s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2d32a6f058ee4da69c753eb012453839 : SUCCESS in 2h 00m 56s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3112b10fd12644a2be4ed53a293ed258 : SUCCESS in 1h 24m 43s","accounts_in_message":[],"_revision_number":3},{"id":"3f8cd2f360da510ac80a74b87c23095abac423f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 15:02:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7ae9691dfa2a5f434eeaba923d3f70ab7bc00bf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 16:48:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f94922dafb4413d9d29dd437c6ef65b : SUCCESS in 6m 02s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c751f2b65c864a0f86a7f5363a4612d0 : SUCCESS in 15m 16s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/c766286ee7414719aa5f6907177bbe2e : SUCCESS in 1h 15m 18s (non-voting)\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/5fc45b9e028a474a9ca6657b77b9f497 : SUCCESS in 1h 04m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdde7949d8024bb6b4c390e6afc77b26 : SUCCESS in 59m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eceaa5b0f3a4f76b294b9ae11549e88 : SUCCESS in 1h 33m 26s","accounts_in_message":[],"_revision_number":3},{"id":"871dc5862bc3ff9a38cda9f7b67a5dceedd9f26d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 16:48:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d0adc2dc794a82836353a9ef946662aab92d0039","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 16:50:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceea77541a214cb39b5c7206ea5f2842 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"adb7551258c626134b54a8e6117860e97b7ef08c","revisions":{"3aa5143205ec7c82a264e7e16e9c26a2906fdf56":{"kind":"REWORK","_number":1,"created":"2021-01-27 08:51:21.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/50/772550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/50/772550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/772550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/772550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/772550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/772550/1"}}},"commit":{"parents":[{"commit":"892a5bfac0947d0b2871191ce098ef3b00607024","subject":"Bump mariadb version to 10.5.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/892a5bfac0947d0b2871191ce098ef3b00607024"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2021-01-20 14:13:55.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-01-27 08:51:21.000000000","tz":0},"subject":"Bring db setup vars in line with other roles","message":"Bring db setup vars in line with other roles\n\nAllows for galera_db_setup_host to be overridden if necessary for delegation.\nBrings the format in line with the db_setup vars in other roles.\n\nChange-Id: Ie2a802ebb8297bed03d74b3cf54907322b858896\n(cherry picked from commit 8ce7da7a3c52126706c90a2bc27cc41c31b87ece)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3aa5143205ec7c82a264e7e16e9c26a2906fdf56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3aa5143205ec7c82a264e7e16e9c26a2906fdf56"}]},"branch":"refs/heads/stable/victoria"},"d44fc4a249ef95253b6449942acdaca5b7531843":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-01 10:21:36.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/50/772550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/50/772550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/772550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/772550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/772550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/772550/2"}}},"commit":{"parents":[{"commit":"892a5bfac0947d0b2871191ce098ef3b00607024","subject":"Bump mariadb version to 10.5.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/892a5bfac0947d0b2871191ce098ef3b00607024"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2021-01-20 14:13:55.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-02-01 10:21:16.000000000","tz":0},"subject":"Bring db setup vars in line with other roles","message":"Bring db setup vars in line with other roles\n\nAllows for galera_db_setup_host to be overridden if necessary for delegation.\nBrings the format in line with the db_setup vars in other roles.\n\nDepends-On: https://review.opendev.org/773264\nChange-Id: Ie2a802ebb8297bed03d74b3cf54907322b858896\n(cherry picked from commit 8ce7da7a3c52126706c90a2bc27cc41c31b87ece)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d44fc4a249ef95253b6449942acdaca5b7531843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d44fc4a249ef95253b6449942acdaca5b7531843"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"},"adb7551258c626134b54a8e6117860e97b7ef08c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-02-02 08:51:21.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/50/772550/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/50/772550/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/772550/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/772550/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/772550/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/772550/3"}}},"commit":{"parents":[{"commit":"892a5bfac0947d0b2871191ce098ef3b00607024","subject":"Bump mariadb version to 10.5.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/892a5bfac0947d0b2871191ce098ef3b00607024"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2021-01-20 14:13:55.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-02-02 08:51:21.000000000","tz":0},"subject":"Bring db setup vars in line with other roles","message":"Bring db setup vars in line with other roles\n\nAllows for galera_db_setup_host to be overridden if necessary for delegation.\nBrings the format in line with the db_setup vars in other roles.\n\nChange-Id: Ie2a802ebb8297bed03d74b3cf54907322b858896\n(cherry picked from commit 8ce7da7a3c52126706c90a2bc27cc41c31b87ece)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/adb7551258c626134b54a8e6117860e97b7ef08c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/adb7551258c626134b54a8e6117860e97b7ef08c"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
