)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~775011","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I1b42c0930d065dc7b79033ace341ae86e2c5df90","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"I1b42c0930d065dc7b79033ace341ae86e2c5df90","subject":"Replace upgrade jobs with infra_lxc","status":"ABANDONED","created":"2021-02-10 20:24:22.000000000","updated":"2022-01-04 13:10:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c408ea26d4d60549ac3104cd9b684043674a2138","_number":775011,"virtual_id_number":775011,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-10 22:16:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 22:16:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aee54251df40c7201f37af38ce4a2d0d96547a82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-10 20:24:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0419ce6426b2cc11e9eab03a4412d8ae1106b76b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 22:16:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f9108dc61cf4e8688fb712dfb136c81 : SUCCESS in 5m 58s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35516fffaa7240428168a818329aee32 : SUCCESS in 16m 46s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/254a62f326cc4e9798a55f30659477e1 : SUCCESS in 1h 09m 08s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/013a00cea1fd45688c4806506f187a80 : SUCCESS in 1h 03m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4db9b14f375b46b08c7af94eb35770c9 : SUCCESS in 59m 51s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/937b3d2472cb44dca6daa5528ea9fb1d : FAILURE in 1h 31m 33s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/633c46d1293d46028ea7c369ae0270ef : FAILURE in 1h 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c408ea26d4d60549ac3104cd9b684043674a2138","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-04 13:10:53.000000000","message":"Abandoned\n\nalready merged via a different patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b373fc1d564cfbe18434d3f58bd8ce19e47e2e2","revisions":{"0b373fc1d564cfbe18434d3f58bd8ce19e47e2e2":{"kind":"REWORK","_number":1,"created":"2021-02-10 20:24:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/775011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/11/775011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/775011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/775011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/775011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/775011/1"}}},"commit":{"parents":[{"commit":"ec58c1d6cc256808d94a13e1e802bdf97ac2890f","subject":"Merge \"[reno] Stop publishing release notes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ec58c1d6cc256808d94a13e1e802bdf97ac2890f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-10 20:24:08.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-10 20:24:08.000000000","tz":120},"subject":"Replace upgrade jobs with infra_lxc","message":"Replace upgrade jobs with infra_lxc\n\nThis will test upgrade between OSA releases in multi container scenario\nconfiguring infra clusters. This should also decrease CI time.\n\nThe downside is that no tempest tests would be run, but probably they\nare not required as well.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/774685/\nChange-Id: I1b42c0930d065dc7b79033ace341ae86e2c5df90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0b373fc1d564cfbe18434d3f58bd8ce19e47e2e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0b373fc1d564cfbe18434d3f58bd8ce19e47e2e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
