)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~777088","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Id28057c9b9043c9ef609f4ed6f40a8a21a2e6a8e","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"777088","hashtags":[],"change_id":"Id28057c9b9043c9ef609f4ed6f40a8a21a2e6a8e","subject":"Bump MariaDB version to 10.5.9","status":"MERGED","created":"2021-02-23 10:26:30.000000000","updated":"2021-04-21 11:00:02.000000000","submitted":"2021-04-21 10:58:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777088-777088","meta_rev_id":"4adc31365ebc0a9084adb31c0301d5796e69b991","_number":777088,"virtual_id_number":777088,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-21 10:58:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2021-04-21 09:34:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":2,"date":"2021-04-21 06:50:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2021-04-21 09:34:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-23 10:43:33.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-02-23 12:15:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-17 13:52:00.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"CC"},{"updated":"2021-04-21 09:34:38.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"}],"messages":[{"id":"08f104d5ac2a662a2a1c482e7855387bc25e9ecf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-23 10:26:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fc5525c2cf5fe89fb0ec1e4347cc0c58c006db2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 10:43:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"75b1b82128be3c48b7154f5ee894d4a7c58c4b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 12:15:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/639621f392274267ac809ef36c9018e4 : SUCCESS in 5m 39s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91f2c2e2c1e74c41829db8dd8c1e48ea : SUCCESS in 16m 46s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/dfc4a6dd4ad14d5f819891674da6bbf7 : SUCCESS in 1h 11m 38s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/cce2a98a1ea94072a962911e8a7bb44b : SUCCESS in 1h 04m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e3870e71142437bacf4f081d3b87e12 : SUCCESS in 1h 04m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/86208ec1103a4a6e800afca992942f94 : FAILURE in 1h 31m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0af1b2a37646470dbae747523eecffc3 : FAILURE in 1h 09m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ad9e973398225b54359e3532913e4fe0ea3e51ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 16:19:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4be71107ccb331fcae28621a9912b84927e46493","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 16:20:37.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c0f9a005448ec050c1314b4238c97ef030c229e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 20:04:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0afe3516eb59409ea0ca01554ea00a03 : SUCCESS in 7m 28s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f370d9865a3d4234b6fd974fa9f5d086 : SUCCESS in 20m 01s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/f4c652f4e1634fa9a032dbc60a507143 : FAILURE in 1h 14m 39s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/8b6f821da22d40f9911efe25c3936348 : FAILURE in 2h 15m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea1d459c887e40aea689a3349e057641 : TIMED_OUT in 3h 05m 04s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/556f76f10ff542b2b1f8f2edd1760052 : FAILURE in 1h 09m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4183f438a5644e1cb391318ef5780a51 : FAILURE in 1h 11m 33s","accounts_in_message":[],"_revision_number":3},{"id":"3f4ce1d6a25e972ffca97c36438fdf156ff79e54","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-02 09:00:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2994dd76a33ddd2c918b9dabb4d2d00d21c4f284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:31:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df1e6f2d4f3741be855565e526a7e020 : SUCCESS in 5m 54s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a652ebb82ed4e82a3aa9d701909126a : SUCCESS in 19m 01s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/01d50194f8be452b85355ca6be58022d : FAILURE in 1h 24m 28s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/5ec2b486389440eab1972ce1e34df612 : FAILURE in 49m 21s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/658c50325dc942fdbf2143b0cbb95b12 : FAILURE in 44m 17s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/943f924c5aff47ccab3081c5317bf27c : FAILURE in 1h 06m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf6dfeb9bc1d4a09a8573cf1f9468dfe : FAILURE in 55m 35s","accounts_in_message":[],"_revision_number":3},{"id":"9d6cce3430ca95cb2fecc61ab53c9a816a477e3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-09 08:47:45.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"399874992cceb973980982596703f1c1c74ebdb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 10:07:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a2af53b21ab4e9b82fb9be7c0b8182b : SUCCESS in 7m 06s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fc0f98423fd46a3a5d10d6b05f15adc : SUCCESS in 24m 26s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/4988b60c19ff491584398c9e8a505d6c : FAILURE in 1h 14m 02s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/01c5e29825f24546ac5edb1aeb21e0bb : FAILURE in 50m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23a88291bbd44788ae656900c48f78f0 : FAILURE in 48m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/18902238f50948099e144554d9ef54c1 : FAILURE in 1h 12m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc948c526fde4787a90ff4fd8417645d : FAILURE in 1h 00m 59s","accounts_in_message":[],"_revision_number":4},{"id":"7795abb9e68cc7199dfe62ad52f3963b32484a83","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2021-03-17 13:52:00.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1540b83995ac7730fc8011ebfb043ea45656b6c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 15:19:47.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed75788bd9be412391c33f4f3261c5c9 : SUCCESS in 8m 10s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39339ee1192c4e2e999cdbaca137d007 : SUCCESS in 22m 31s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/4f74c874be884afab4c45288efc8a487 : FAILURE in 1h 15m 22s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/0cb30539aff8498dada9930853b097a3 : FAILURE in 54m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ffee87bb2fb40bd971c7f39afefb2b5 : FAILURE in 49m 44s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/25280b92d5504ab88fb43e9971fd4933 : FAILURE in 1h 12m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef6c21b2bd464afbae3b4515b5141ce0 : FAILURE in 57m 29s","accounts_in_message":[],"_revision_number":4},{"id":"56ad720e9fc9a26b1e4b3b37604b535526476081","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2021-03-18 10:50:16.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"023dcc9e105861ab46621e4398077590b5e32cb1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-18 12:35:17.000000000","message":"Patch Set 5:\n\nI believe that upgrade might fail beause of that https://review.opendev.org/c/openstack/openstack-ansible/+/775684/7/scripts/upgrade-utilities/deploy-config-changes.yml","accounts_in_message":[],"_revision_number":5},{"id":"3fefad2c8f19a5c284f3ba885637407d9f7f12b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 12:47:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d014ad904534c8eac56e086d0adefcf : SUCCESS in 6m 43s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77a880e0a5c3495fb22d6c69d2c4dd04 : SUCCESS in 19m 20s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b0b9bbb08a6c44d2b5b20030d2f2f04e : SUCCESS in 1h 18m 24s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/92cfbd18110348f4a7e84889f63ce551 : SUCCESS in 1h 08m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bc29224bd6543429ccb66523891c762 : SUCCESS in 1h 13m 51s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d8274e39872e4f48a723b44312fdabcc : FAILURE in 1h 06m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b6e0c7384fc4c2291764903fb417219 : FAILURE in 59m 49s","accounts_in_message":[],"_revision_number":5},{"id":"35566cd3cb1af93a89fb0583b0e7d8146cfc0ce0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-18 14:11:54.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b44c582a0005e60b2e27b284c5e0e85c448810a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:02:53.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41faf102d954fe99ad73d74267e4a86 : SUCCESS in 6m 03s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40b903a6286649bca55ebfe752819f1e : SUCCESS in 27m 53s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/9ac9b74bd2724a779e8fa853041c0fc6 : SUCCESS in 1h 19m 25s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c2bcdb3b2fe14b229ff5bc69bbebd039 : SUCCESS in 1h 06m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/935334350a4644959418e881e072ea29 : SUCCESS in 1h 06m 02s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/377d2b99258b4288a3852791ef54d501 : FAILURE in 1h 06m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b35b5a74dc3e46dd9785fddf535a7b51 : FAILURE in 1h 04m 43s","accounts_in_message":[],"_revision_number":5},{"id":"c9fe3872691d048727de33f680ee507de6f44704","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-22 15:05:04.000000000","message":"Patch Set 5:\n\nOk, so upgrades fail with\nfailed: [aio1] (item\u003d{\u0027name\u0027: \u0027admin\u0027, \u0027host\u0027: \u0027%\u0027, \u0027password\u0027: \u0027d7c6d77d71e3350d9a95d25664f382\u0027, \u0027priv\u0027: \u0027*.*:ALL,GRANT\u0027, \u0027state\u0027: \u0027present\u0027}) \u003d\u003e {\"ansible_loop_var\": \"item\", \"attempts\": 3, \"changed\": false, \"item\": {\"host\": \"%\", \"name\": \"admin\", \"password\": \"d7c6d77d71e3350d9a95d25664f382\", \"priv\": \"*.*:ALL,GRANT\", \"state\": \"present\"}, \"msg\": \"(1045, \\\"Access denied for user \u0027root\u0027@\u0027localhost\u0027 (using password: YES)\\\")\"}","accounts_in_message":[],"_revision_number":5},{"id":"a73073149d824f78910074e647ef4802c1932dc6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-22 15:15:45.000000000","message":"Patch Set 5:\n\nIs this because it\u0027s using root@localhost (tcp connection) rather than root/unix-socket connection?","accounts_in_message":[],"_revision_number":5},{"id":"b198bd6057d84a78cba68caed9ccb3cab7cfefe8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-22 15:25:03.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e Is this because it\u0027s using root@localhost (tcp connection) rather than root/unix-socket connection?\n\nI think yes, but... We\u0027re resetting root@localhost passowrd on master-1, so unix-socket connection should not work either...","accounts_in_message":[],"_revision_number":5},{"id":"386e79f9afc8c7d1a087d62b0005701e8d244327","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2021-03-23 16:59:51.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"f5287b7f76d64e61b8efae5f57c5faf2a3df05e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 18:23:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/743de2bd29dd45728b8f49ce8d200853 : SUCCESS in 6m 06s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/886e6863f8be4a6eae8377857407f115 : SUCCESS in 23m 11s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/85ccaadd30b74644b2d7a69d7db18149 : SUCCESS in 1h 11m 56s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/16cc661e6dda4a0384c7cf1b78cb0546 : SUCCESS in 1h 08m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c0d431545c145c8828b609a0b51af15 : SUCCESS in 1h 08m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e473830655034bf8a86d7d047142d5d1 : FAILURE in 1h 07m 19s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4e5f1ac7ab64a28960516b5e42957b2 : FAILURE in 1h 02m 29s","accounts_in_message":[],"_revision_number":6},{"id":"6dbfb8ab08f52fa0e90813dc9a2587d2da8a31b3","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2021-03-25 19:18:15.000000000","message":"Patch Set 6:\n\nhttps://jira.mariadb.org/browse/MDEV-25030?jql\u003dtext%20~%20%2210.5.9%22\nCurrently this is our line of thinking.\nThanks to Jon on the good spot for that one!\nHopefully can look at the workarounds listed in the bug soon.","accounts_in_message":[],"_revision_number":6},{"id":"55e7ceb1a9d09223ce1df0b7568f039eaef77e4b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-30 07:15:22.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6:\n\u003e \n\u003e https://jira.mariadb.org/browse/MDEV-25030?jql\u003dtext%20~%20%2210.5.9%22\n\u003e Currently this is our line of thinking.\n\u003e Thanks to Jon on the good spot for that one!\n\u003e Hopefully can look at the workarounds listed in the bug soon.\n\nYeah, agree. We probably need to grant permissions and create admin user in upgrade scripts.","accounts_in_message":[],"_revision_number":6},{"id":"bfffc88a7495b66798321d1368d250fd0878684a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-16 16:38:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fe74c4a935744f5672b53d73cf5a345cc8b466b5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-16 16:45:25.000000000","message":"Patch Set 7: Workflow-1\n\nwhile that\u0027s going to work on metal aio, I\u0027m not sure this is valid solution for other ones. Since we rely on `listen: Bootstrap cluster` and not sure that will be the case while upgrading multinode deployment.\n\nSo probably more work should be done in terms of the listening `Restart mysql (All)` and splitting that handler into independant stop/start to chime in in the meanwhile","accounts_in_message":[],"_revision_number":7},{"id":"e78ca75e66cbc213deb43cd57830cd837ed2174b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 18:16:50.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd4c628fdc934f4d8dc693c4205f87a7 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2a3351a729c4c1b9533aa5a57309cd7 : SUCCESS in 3m 46s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49f09ea99d3a44c8a891a9c1a0ce5f80 : SUCCESS in 18m 22s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/bfcbbc15d6a849b389a66cecd20d47d0 : SUCCESS in 1h 13m 02s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/8dfe123b77724c5a91d529a7975b8ee1 : SUCCESS in 1h 06m 55s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/175c34ef47064c5a89145457d54c63a9 : SUCCESS in 1h 05m 06s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/655e3a5fa427472d88bec41e6d7d221a : FAILURE in 1h 08m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54697d1e259b4d1dab4b5f9fb2244520 : FAILURE in 1h 00m 15s","accounts_in_message":[],"_revision_number":7},{"id":"e68baaedb63d7ab90961346ac4bc587c83eeb559","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-16 18:43:20.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"5900d0a169bf1c2b562f67262e9ef9204cd9e4d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 20:51:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b33b57dfce424ec2a9e561e320b4ea95 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8196435dea9a450cbdc7b9bb08d24843 : SUCCESS in 4m 37s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60f1647f6c0b416397217098fece49ec : SUCCESS in 16m 32s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/132aecb367744754b2803cd28b0cc55d : SUCCESS in 1h 11m 53s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/4d21ffad8892497795d4d86d2dd77418 : SUCCESS in 1h 03m 23s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86bdb61a66854e81a281afb4a6d7c646 : SUCCESS in 1h 02m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/47f7d061dc104aa0994a4711088c9e48 : SUCCESS in 2h 00m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6948b5af0f6a4d32bea1da1d3d140a8b : SUCCESS in 1h 40m 54s","accounts_in_message":[],"_revision_number":8},{"id":"75c5d146aa4eb5ffb4cde23f9c83fff79b8859f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 09:23:06.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"84d7d443fff193a462160bd3dd28c8b5ac65f217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 10:52:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c789abac2ef345a8a0a8134e90f529df : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8010df4e434840959006d5e09a4d2677 : SUCCESS in 4m 39s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5fa4bbc31384850aae70ec6fea84ab7 : SUCCESS in 16m 41s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/35d1df26a3c147c293b28de5ae40d80b : SUCCESS in 1h 06m 21s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/6b22082944944e3094a12eb4c008e246 : SUCCESS in 1h 07m 45s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75da956f727346118bb2a261b546c9e6 : SUCCESS in 1h 10m 39s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/aa9955dbd8bf457f8f9d3be261549ed9 : FAILURE in 1h 23m 49s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e79f7bef8f9c44d0a3bcd4137932d349 : SUCCESS in 1h 21m 13s","accounts_in_message":[],"_revision_number":9},{"id":"edfc4d057729b584ca4deb5874b77a0a36f8ada8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 14:34:19.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"58abdee517974e5707279439e718793c896858ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 14:35:28.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"63efbfbb2ba0797afbb281783d75059db4808552","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 16:04:33.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e13897927564c3f8eedd223e593335b : SUCCESS in 9m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0b7e66b4d9d46bfb93e50e8f7d4a8a0 : SUCCESS in 4m 52s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c2018b9e17c4b62b95e50017e7dc7fd : SUCCESS in 21m 15s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/271e623e30aa46e99d1184d873950a1f : SUCCESS in 1h 12m 50s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/a14184a60a1846cdad22c64e3237aef3 : SUCCESS in 1h 14m 31s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d147a46124824a44ad9023bf97c63f78 : SUCCESS in 1h 09m 30s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/3aaca5aeb10d458cae168401dc727589 : FAILURE in 1h 22m 05s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9bac23126574fcc9cd7dd298ad578d6 : SUCCESS in 1h 24m 31s","accounts_in_message":[],"_revision_number":11},{"id":"4d848514c4c98a4abf947856f61a4bf0121770e3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-21 06:50:31.000000000","message":"Patch Set 11: Code-Review+2\n\nWell, it feels like working for me. Let\u0027s try it out I guess?","accounts_in_message":[],"_revision_number":11},{"id":"1c03ecb1b02385435c83375a5a3731ca14356003","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2021-04-21 09:34:38.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"7c876c039a08e512df7d1815fb5b6e013ad71d72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 09:35:04.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"a83b46c61ede48ee767bb0f80112957e7ca87e71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 10:58:38.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd0aa8bf269f409daed4f944a57908b7 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8966adf67614467a3f6c8fef30203b4 : SUCCESS in 3m 48s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c0121c60f0d40f0a7ca4fa65ba29217 : SUCCESS in 16m 15s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/403c4217cda44450a49737d7b51152b4 : SUCCESS in 1h 11m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e00fb0c5f06a49118cbc5a7be83ed903 : SUCCESS in 1h 02m 59s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ad4c6ff828a4ca38996eb86d93e663a : SUCCESS in 1h 15m 13s","accounts_in_message":[],"_revision_number":11},{"id":"63764762e054d4cca410be6930364f4c28d59138","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 10:58:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"4adc31365ebc0a9084adb31c0301d5796e69b991","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 11:00:02.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0254145aa4e248c5bce4c7c73c652b4b : SUCCESS in 47s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"1db0ef2efbd22b3491e1cf57e172a3d87b6f62a9","revisions":{"8fa9c69e9a00c0f7766ed89d3aff252c7a7ec8cf":{"kind":"REWORK","_number":1,"created":"2021-02-23 10:26:30.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/88/777088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/777088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/1"}}},"commit":{"parents":[{"commit":"3d07a246d5d9300d2ee3e8befbebba454dc4bcf8","subject":"Install xinetd clustercheck after mariadb is installed and setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3d07a246d5d9300d2ee3e8befbebba454dc4bcf8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 10:23:38.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 10:26:19.000000000","tz":0},"subject":"Bump MariaDB version to 10.5.9","message":"Bump MariaDB version to 10.5.9\n\nChange-Id: Id28057c9b9043c9ef609f4ed6f40a8a21a2e6a8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8fa9c69e9a00c0f7766ed89d3aff252c7a7ec8cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8fa9c69e9a00c0f7766ed89d3aff252c7a7ec8cf"}]},"branch":"refs/heads/master"},"9ef3ec10ffc94d3543d6d51669e170d318a169ac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-23 16:19:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/777088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/777088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/2"}}},"commit":{"parents":[{"commit":"931f3c74a78774f319acfb6867ff9742b9b46e3d","subject":"Configure my.cnf for server when galera_root_user is not root","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/931f3c74a78774f319acfb6867ff9742b9b46e3d"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 10:23:38.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-23 16:19:21.000000000","tz":0},"subject":"Bump MariaDB version to 10.5.9","message":"Bump MariaDB version to 10.5.9\n\nChange-Id: Id28057c9b9043c9ef609f4ed6f40a8a21a2e6a8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9ef3ec10ffc94d3543d6d51669e170d318a169ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9ef3ec10ffc94d3543d6d51669e170d318a169ac"}]},"branch":"refs/heads/master","description":"Rebase"},"13dbb915d218490e1d6d6dac714f9affa1e4e281":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-02-23 16:20:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/777088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/777088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/3"}}},"commit":{"parents":[{"commit":"931f3c74a78774f319acfb6867ff9742b9b46e3d","subject":"Configure my.cnf for server when galera_root_user is not root","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/931f3c74a78774f319acfb6867ff9742b9b46e3d"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 10:23:38.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-23 16:20:33.000000000","tz":0},"subject":"Bump MariaDB version to 10.5.9","message":"Bump MariaDB version to 10.5.9\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/775684\nChange-Id: Id28057c9b9043c9ef609f4ed6f40a8a21a2e6a8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/13dbb915d218490e1d6d6dac714f9affa1e4e281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/13dbb915d218490e1d6d6dac714f9affa1e4e281"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1b42c73d53881f47249e1f70fc5c374a26167ef7":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-03-09 08:47:45.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/88/777088/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/777088/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/4"}}},"commit":{"parents":[{"commit":"931f3c74a78774f319acfb6867ff9742b9b46e3d","subject":"Configure my.cnf for server when galera_root_user is not root","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/931f3c74a78774f319acfb6867ff9742b9b46e3d"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 10:23:38.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-09 08:47:45.000000000","tz":0},"subject":"Bump MariaDB version to 10.5.9","message":"Bump MariaDB version to 10.5.9\n\nDepends-On: https://review.opendev.org/775684\nChange-Id: Id28057c9b9043c9ef609f4ed6f40a8a21a2e6a8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1b42c73d53881f47249e1f70fc5c374a26167ef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1b42c73d53881f47249e1f70fc5c374a26167ef7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7e75f21fec72c120619bbfa1213569befc9081b8":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-03-18 10:50:16.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/88/777088/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/777088/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/5"}}},"commit":{"parents":[{"commit":"931f3c74a78774f319acfb6867ff9742b9b46e3d","subject":"Configure my.cnf for server when galera_root_user is not root","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/931f3c74a78774f319acfb6867ff9742b9b46e3d"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 10:23:38.000000000","tz":0},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2021-03-18 10:49:54.000000000","tz":0},"subject":"Bump MariaDB version to 10.5.9","message":"Bump MariaDB version to 10.5.9\n\nDepends-On: https://review.opendev.org/775684\nDepends-On: https://review.opendev.org/781305\nChange-Id: Id28057c9b9043c9ef609f4ed6f40a8a21a2e6a8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7e75f21fec72c120619bbfa1213569befc9081b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7e75f21fec72c120619bbfa1213569befc9081b8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1f607cbdd124d914c99098792c6c5ad47a1ce290":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-03-23 16:59:51.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/88/777088/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/777088/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/6"}}},"commit":{"parents":[{"commit":"1bb878049915c64dc588a8b8ef0e51c70825359a","subject":"Merge \"Update galera_wsrep_sst_auth_user to use galera_root_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bb878049915c64dc588a8b8ef0e51c70825359a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 10:23:38.000000000","tz":0},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2021-03-23 16:59:51.000000000","tz":0},"subject":"Bump MariaDB version to 10.5.9","message":"Bump MariaDB version to 10.5.9\n\nDepends-On: https://review.opendev.org/775684\nDepends-On: https://review.opendev.org/781305\nChange-Id: Id28057c9b9043c9ef609f4ed6f40a8a21a2e6a8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1f607cbdd124d914c99098792c6c5ad47a1ce290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1f607cbdd124d914c99098792c6c5ad47a1ce290"}]},"branch":"refs/heads/master","description":"Rebase"},"61e7777afb7001d85e400901e9f7cd6c55b97890":{"kind":"REWORK","_number":7,"created":"2021-04-16 16:38:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/777088/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/777088/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/7"}}},"commit":{"parents":[{"commit":"1bb878049915c64dc588a8b8ef0e51c70825359a","subject":"Merge \"Update galera_wsrep_sst_auth_user to use galera_root_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bb878049915c64dc588a8b8ef0e51c70825359a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 10:23:38.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-16 16:37:11.000000000","tz":180},"subject":"Bump MariaDB version to 10.5.9","message":"Bump MariaDB version to 10.5.9\n\nWe also workaround known mariadb bug which make upgrades from previous\nversion to fail because of changing privileges bits which ends up\nin revoking some of the privileges from superusers.\n\nDepends-On: https://review.opendev.org/775684\nDepends-On: https://review.opendev.org/781305\nChange-Id: Id28057c9b9043c9ef609f4ed6f40a8a21a2e6a8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/61e7777afb7001d85e400901e9f7cd6c55b97890"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/61e7777afb7001d85e400901e9f7cd6c55b97890"}]},"branch":"refs/heads/master"},"4f50934aa04857b67b62d6e4eabaa774d523eba7":{"kind":"REWORK","_number":8,"created":"2021-04-16 18:43:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/777088/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/777088/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/8"}}},"commit":{"parents":[{"commit":"1bb878049915c64dc588a8b8ef0e51c70825359a","subject":"Merge \"Update galera_wsrep_sst_auth_user to use galera_root_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bb878049915c64dc588a8b8ef0e51c70825359a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 10:23:38.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-16 18:43:16.000000000","tz":0},"subject":"Bump MariaDB version to 10.5.9","message":"Bump MariaDB version to 10.5.9\n\nWe also workaround known mariadb bug which make upgrades from previous\nversion to fail because of changing privileges bits which ends up\nin revoking some of the privileges from superusers.\n\nDepends-On: https://review.opendev.org/775684\nDepends-On: https://review.opendev.org/781305\nChange-Id: Id28057c9b9043c9ef609f4ed6f40a8a21a2e6a8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4f50934aa04857b67b62d6e4eabaa774d523eba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4f50934aa04857b67b62d6e4eabaa774d523eba7"}]},"branch":"refs/heads/master"},"2b74e39cb8ec43640175aeb1218e2ded14a0fefb":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-04-19 09:23:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/777088/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/777088/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/9"}}},"commit":{"parents":[{"commit":"e336046b633ae79628643360909b77a4546197c7","subject":"Use infra LXC upgrade jobs instead of metal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e336046b633ae79628643360909b77a4546197c7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 10:23:38.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-19 09:23:06.000000000","tz":0},"subject":"Bump MariaDB version to 10.5.9","message":"Bump MariaDB version to 10.5.9\n\nWe also workaround known mariadb bug which make upgrades from previous\nversion to fail because of changing privileges bits which ends up\nin revoking some of the privileges from superusers.\n\nDepends-On: https://review.opendev.org/775684\nDepends-On: https://review.opendev.org/781305\nChange-Id: Id28057c9b9043c9ef609f4ed6f40a8a21a2e6a8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2b74e39cb8ec43640175aeb1218e2ded14a0fefb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2b74e39cb8ec43640175aeb1218e2ded14a0fefb"}]},"branch":"refs/heads/master","description":"Rebase"},"de7f206ce78e99d92fc2449aa9fcc549096cf5be":{"kind":"REWORK","_number":10,"created":"2021-04-19 14:34:19.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/777088/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/777088/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/10"}}},"commit":{"parents":[{"commit":"e336046b633ae79628643360909b77a4546197c7","subject":"Use infra LXC upgrade jobs instead of metal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e336046b633ae79628643360909b77a4546197c7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 10:23:38.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-19 14:34:16.000000000","tz":0},"subject":"Bump MariaDB version to 10.5.9","message":"Bump MariaDB version to 10.5.9\n\nWe also workaround known mariadb bug which make upgrades from previous\nversion to fail because of changing privileges bits which ends up\nin revoking some of the privileges from superusers.\n\nDepends-On: https://review.opendev.org/775684\nDepends-On: https://review.opendev.org/781305\nChange-Id: Id28057c9b9043c9ef609f4ed6f40a8a21a2e6a8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/de7f206ce78e99d92fc2449aa9fcc549096cf5be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/de7f206ce78e99d92fc2449aa9fcc549096cf5be"}]},"branch":"refs/heads/master"},"1db0ef2efbd22b3491e1cf57e172a3d87b6f62a9":{"kind":"REWORK","_number":11,"created":"2021-04-19 14:35:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/777088/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/88/777088/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/88/777088/11"}}},"commit":{"parents":[{"commit":"e336046b633ae79628643360909b77a4546197c7","subject":"Use infra LXC upgrade jobs instead of metal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e336046b633ae79628643360909b77a4546197c7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 10:23:38.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-19 14:35:25.000000000","tz":0},"subject":"Bump MariaDB version to 10.5.9","message":"Bump MariaDB version to 10.5.9\n\nWe also workaround known mariadb bug which make upgrades from previous\nversion to fail because of changing privileges bits which ends up\nin revoking some of the privileges from superusers.\n\nDepends-On: https://review.opendev.org/775684\nDepends-On: https://review.opendev.org/781305\nChange-Id: Id28057c9b9043c9ef609f4ed6f40a8a21a2e6a8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1db0ef2efbd22b3491e1cf57e172a3d87b6f62a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1db0ef2efbd22b3491e1cf57e172a3d87b6f62a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
