)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~781305","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I8f03ee7a7a144fa901caf7b6c1ed041e09f2ffc0","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"I8f03ee7a7a144fa901caf7b6c1ed041e09f2ffc0","subject":"Update galera_wsrep_sst_auth_user to use galera_root_user","status":"MERGED","created":"2021-03-18 10:44:38.000000000","updated":"2021-03-22 23:33:35.000000000","submitted":"2021-03-22 23:32:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781305","meta_rev_id":"16c6600a3839a76e117f7082893c9da0605a55f4","_number":781305,"virtual_id_number":781305,"owner":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-22 23:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2021-03-22 20:38:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2021-03-18 10:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2021-03-22 20:38:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 10:52:55.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-03-18 13:08:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 20:38:31.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"}],"messages":[{"id":"6fd93326b89bcde83dc6d24f2675bbc0782b45a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2021-03-18 10:44:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"330e94d33c4468f059e44d73777e012d0125fd6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2021-03-18 10:46:16.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8694fd9530618d8cd8bf74cdd0c657405ffd1567","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-18 10:52:55.000000000","message":"Patch Set 2: Code-Review+2\n\nOh, nice catch!","accounts_in_message":[],"_revision_number":2},{"id":"78980b51a1ebef716f061dfc30746bb1e1c38735","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 13:08:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef48c7617f0e4f46a6d425286bee0a63 : SUCCESS in 5m 34s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7521c83dc5c64aea8b69ca2981cd53fd : SUCCESS in 20m 02s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/ee312c040e4b4895ae1010616c2d7910 : SUCCESS in 1h 18m 20s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/16dd80208bd645f8806bd821fbb2765d : FAILURE in 1h 08m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/510b3cae7fa34a6aa9f8ce168f43b067 : SUCCESS in 1h 08m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c985cd3f51834c8d9f111e81dfcbdde1 : SUCCESS in 1h 50m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6e4b37c150146e39b0ed0221f9df68f : SUCCESS in 1h 37m 26s","accounts_in_message":[],"_revision_number":2},{"id":"8d0fc7cbebe206f0e68811ff6e3a2015dcca08ab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-22 12:05:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6881c2ca74aca75ac6ec4e233f7acaaa370c0202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 14:23:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf62ffe45394133a0945afed9c15517 : SUCCESS in 6m 25s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42967bfca89f4c4bac28bd18d2221fbf : SUCCESS in 18m 06s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/a05597454ae44909aa6be9d36c724d45 : SUCCESS in 1h 08m 03s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/79b12419c53446c28c81466ffce64003 : SUCCESS in 1h 07m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9952503e9d042f8b7b1b821ef589f41 : SUCCESS in 1h 11m 37s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1d3e93be0f764efd809339c8da5ed247 : SUCCESS in 2h 05m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a18241ee3194a06baa5fde40370ac67 : SUCCESS in 1h 30m 43s","accounts_in_message":[],"_revision_number":2},{"id":"a5a1a009e284609c64d6813f942f0f3ec157aaf3","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-22 20:38:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"17ecd43d940161ee46b5786c94316a2556332dc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 20:39:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c5fb434a08db19b0b38717b65d2f4ca051745125","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 21:08:26.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"54393e65dc8a17db0efd3951d0cf6a9aec201cf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 23:32:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bbf01a4c80e430893e7fa23270a2c45 : SUCCESS in 9m 04s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dd57a1da9c743e6ae218045db72c8e3 : SUCCESS in 19m 09s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b7e4101db1004924bf860c584adcafd2 : SUCCESS in 1h 41m 40s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c087e4d209f84ecd9a54288fa06ef29f : SUCCESS in 1h 42m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/558a49135e5045fd9d30c8953343e280 : SUCCESS in 1h 19m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cb67853b6e24bb1b3f8eba52f36f327 : SUCCESS in 2h 09m 59s","accounts_in_message":[],"_revision_number":2},{"id":"86d529fbbb1f302f9a6dcc001cc2f920c7e0775c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 23:32:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"16c6600a3839a76e117f7082893c9da0605a55f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 23:33:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4119b35764b94e829b3c3422784b47ac : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"84e9f9671268278eef1584cef885bf8aa85ed3a5","revisions":{"cbb83e5d4a4d6cfc3fcab27ff4831de992cd6f93":{"kind":"REWORK","_number":1,"created":"2021-03-18 10:44:38.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/05/781305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/781305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/781305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/781305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/781305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/781305/1"}}},"commit":{"parents":[{"commit":"49b7d9488b915addc21b28c60b1385b4efc42dd3","subject":"Remove references to unsupported operating systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/49b7d9488b915addc21b28c60b1385b4efc42dd3"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2021-03-18 10:35:12.000000000","tz":0},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2021-03-18 10:35:12.000000000","tz":0},"subject":"Update galera_wsrep_sst_auth_user to use galera_root_user","message":"Update galera_wsrep_sst_auth_user to use galera_root_user\n\nWith the changes to the root user and creation of the admin user in:\nhttps://review.opendev.org/c/openstack/openstack-ansible/+/775684/\ngalera_wsrep_sst_auth_user needs to use the overridden galera_root_user\nas the password is no longer set on the root user.\nOtherwise galera cannot sync and cluster properly as access is denied.\n\nChange-Id: I8f03ee7a7a144fa901caf7b6c1ed041e09f2ffc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cbb83e5d4a4d6cfc3fcab27ff4831de992cd6f93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cbb83e5d4a4d6cfc3fcab27ff4831de992cd6f93"}]},"branch":"refs/heads/master"},"84e9f9671268278eef1584cef885bf8aa85ed3a5":{"kind":"NO_CHANGE","_number":2,"created":"2021-03-18 10:46:16.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/05/781305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/781305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/781305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/781305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/781305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/781305/2"}}},"commit":{"parents":[{"commit":"49b7d9488b915addc21b28c60b1385b4efc42dd3","subject":"Remove references to unsupported operating systems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/49b7d9488b915addc21b28c60b1385b4efc42dd3"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2021-03-18 10:35:12.000000000","tz":0},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2021-03-18 10:46:05.000000000","tz":0},"subject":"Update galera_wsrep_sst_auth_user to use galera_root_user","message":"Update galera_wsrep_sst_auth_user to use galera_root_user\n\nWith the changes to the root user and creation of the admin user in:\nhttps://review.opendev.org/c/openstack/openstack-ansible/+/775684/\ngalera_wsrep_sst_auth_user needs to use the overridden galera_root_user\nas the password is no longer set on the root user.\nOtherwise galera cannot sync and cluster properly as access is denied.\n\nChange-Id: I8f03ee7a7a144fa901caf7b6c1ed041e09f2ffc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/84e9f9671268278eef1584cef885bf8aa85ed3a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/84e9f9671268278eef1584cef885bf8aa85ed3a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
