)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~784069","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I73e548ac208a96ddaa687a1b5fbb22cac20037d0","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"bug/1921861","hashtags":[],"change_id":"I73e548ac208a96ddaa687a1b5fbb22cac20037d0","subject":"add support for encryption","status":"MERGED","created":"2021-03-31 10:43:33.000000000","updated":"2021-04-20 16:15:24.000000000","submitted":"2021-04-20 16:13:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":36,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"784069-bug/1921861","meta_rev_id":"785945629d0b5fb87f160e1552db05ae70371155","_number":784069,"virtual_id_number":784069,"owner":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-20 16:12:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-04-20 09:42:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-20 14:43:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-20 14:43:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-31 11:11:06.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-03-31 12:50:38.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-03-31 16:20:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b47224120a35ca49cb681a5fe85d8ca47ab8b8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-03-31 10:43:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a884f6a60b48d628a165c4f592db8bd40680edfd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-31 11:11:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"23fa55e76d39bc3ab4932dbb60b29e74565e20d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-03-31 12:15:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ffe0bb2a5382659fa03e49de0d6e9575fdf18efe","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-03-31 12:18:48.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"96b6ee25f8b9d3b7231db944a674adace1a70301","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-03-31 12:22:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8543ea073b039603c1c4081b44b4e757716f6726","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-31 12:45:42.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9d8640bfae8ebdacf0ffe96a52642deb8d945ccc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-31 12:47:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"819e1cb437314a5e4bf1418c33b2fe0b6b25e989","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-31 12:49:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b20c6f1252f77cdeac5c367169b7cf0086263537","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-31 12:49:40.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"1cc726960333562b0114d2f7abb877bc2655a151","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-03-31 13:02:50.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"53912d55aab719e3cf471ed13997788fd3aa2a70","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-03-31 13:14:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4aff19023ed5170fb5669b3795960a1f45c1eb6c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-31 13:49:10.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8ddf6d5734416b2ebcf755f3fe648e2f4b4f4085","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-03-31 14:17:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a96a97e67288ea098c98013a1313ada7d0c4b536","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-03-31 14:20:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dda2e852894b0ecd213dac3ba2d495361d22bf63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 16:20:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f9169ab88641979e7a73ed356fedfb : SUCCESS in 6m 24s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af43d9a1a4ac48b4b4ad3c7b60e01c13 : FAILURE in 15m 16s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/44e3ad4c453b410fb406cb1e5766fdc7 : SUCCESS in 1h 09m 34s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/74c4c81206f343389be2e21a5b476279 : SUCCESS in 1h 05m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e055664702f44e0b834e63187a172f82 : SUCCESS in 1h 09m 05s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a4db2c56723549dbaf24b85e8cec4794 : SUCCESS in 1h 53m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bafa5851b1e44179ba836fe75ed9223 : SUCCESS in 1h 37m 31s","accounts_in_message":[],"_revision_number":4},{"id":"da2d1d3712a8422ff96d04d548d7e5bcdfe05b92","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-15 16:36:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1587cac63554687693c6b6dbb943540cfbafbd58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-04-19 07:22:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c03bc05c4044493c5ad09b64ff4bf7d00c3922b1","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-04-19 07:24:17.000000000","message":"Patch Set 5:\n\n(8 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0b5f405e477862973087177e3878ddec71b5f47b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 07:28:36.000000000","message":"Patch Set 5:\n\nWould you mind writing a release note for this feature?\nreno usage guide https://docs.openstack.org/reno/latest/user/usage.html","accounts_in_message":[],"_revision_number":5},{"id":"b3790916016a47fd89f0d18db437e06fd77cfd80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-04-19 07:55:30.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e2025bee3b0a73314973ff4c8ff0f1f9aee92bb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 09:55:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f98d44578a74840b1f017f5d82b2810 : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81d6392a600a496bbf0b017e431f9dce : FAILURE in 3m 47s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0386eed9b151480ba686b1136d479e9b : SUCCESS in 16m 16s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/1ab482e5ce0f4be8a068c478cad56496 : SUCCESS in 1h 14m 52s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/4a0eb36b8dc145018d58a75397ce75c9 : SUCCESS in 1h 04m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3737a910b374c4c893cb369ba3090fc : SUCCESS in 1h 07m 32s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d63cc6bc1a114dd3a473c577b566056b : SUCCESS in 1h 52m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1ed99d8611e42b9acf81d2588d73c9c : SUCCESS in 1h 42m 15s","accounts_in_message":[],"_revision_number":6},{"id":"ceb3aa514132498a4bb6e7cfc91888f70d9a827c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-04-19 10:02:27.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0345b94be2eb56a754a476e737b0d46d87979f49","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 10:22:58.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"da3e20506cd436eaffd232e54154472e127f4301","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 10:25:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e12fb75f06bb2079c7c500057502e5abeff5721e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 10:26:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c25e3dc10c0d87eb19bdb38fe5ff574e37497843","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-04-19 10:29:27.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"319f5a933c4d15fdaf50d41a5127b5c9b908aa55","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-04-19 10:30:02.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7351e7063c6813e3807228898b65407e0f3907be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 12:19:14.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f714e2c09ea47f9a67bfde5b8b4f321 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f014c6e2e65c400da0ad093f86f4b5ad : FAILURE in 5m 00s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d245bbcdd75848119c60abd2448ff4a9 : SUCCESS in 15m 13s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b8789b0a052142b58038f0135f0fa98e : SUCCESS in 1h 10m 36s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/23e1554d25da4c11be682f93acc9845c : SUCCESS in 1h 03m 31s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28fed194d0cf44d5ba7531921cad8fe7 : SUCCESS in 1h 08m 46s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/56f33776160b4927952147c560b3de0a : SUCCESS in 1h 26m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34c0f0a9b33d4e0d8c2bc51b94138819 : SUCCESS in 1h 40m 25s","accounts_in_message":[],"_revision_number":8},{"id":"fcd327f299dcd0303fa5c7bb5b13ea2f97f2e2fc","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-04-19 12:49:25.000000000","message":"Patch Set 8:\n\nI don\u0027t know if the failing CI test is caused by my changes..","accounts_in_message":[],"_revision_number":8},{"id":"09e15b020139e0db6bb14dc141c77773d5456b02","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 14:31:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3b58dc50489e8e0975a979f2c3e96a7697f64a38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-04-20 05:03:01.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"601e3b942fd3fb1508291332fe8ec0116ddb78cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-20 06:42:33.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"a0c5696ff1450bfae94564002b4bc0b32488b7e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 09:15:02.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0355a50b48b94340842709fa8221d4e1 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de8de84ea0664f7ab58cd22082025b3a : SUCCESS in 4m 42s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc7ee6d513b34491b1c20afc01997400 : SUCCESS in 18m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/9faed15c13a141569d5c43af95787d9c : SUCCESS in 1h 15m 25s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/839b3beb9cef43a5ac9989832bb85271 : SUCCESS in 1h 06m 24s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1389ec91f675499ab0c3610a28b3b12f : SUCCESS in 59m 51s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6a2a8f141aaf435792684886fe2c2370 : SUCCESS in 2h 23m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3754bae9b3e545ac87e734281c0105e6 : SUCCESS in 1h 32m 44s","accounts_in_message":[],"_revision_number":10},{"id":"f49412f03cf8beec35a9b4cf704931cb1f28b666","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-20 09:42:07.000000000","message":"Patch Set 10: Code-Review+2\n\nI think we can start from this and iterate later if needed","accounts_in_message":[],"_revision_number":10},{"id":"84020cc1f47cddd81c0311184a07a2e9d0bcc18d","author":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"date":"2021-04-20 09:49:20.000000000","message":"Patch Set 10:\n\n\u003e Patch Set 10: Code-Review+2\n\u003e \n\u003e I think we can start from this and iterate later if needed\n\nThanks!","accounts_in_message":[],"_revision_number":10},{"id":"bf74cba5f8e2807445532d4d7f803c8b90e92d81","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-20 14:43:29.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"a8307324e8c03d817c0c8a056c782bd089224063","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 14:43:53.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"1029a303ad73be29695ca16ac3f5ec64e06d1a2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 16:12:59.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/996767392d1d4d2295c3ccecbe05ecd5 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6149a1015ee94784b52a73ce19ed5ce1 : SUCCESS in 3m 56s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c6212864aa04376943ac16bcfab2022 : SUCCESS in 14m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/68196c683a7c490c8cf1f11a66760736 : SUCCESS in 1h 19m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51d9337b82684323a051a9296356391d : SUCCESS in 1h 03m 08s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bfc223ecd55406da86a1ff1b33a6cd8 : SUCCESS in 1h 15m 08s","accounts_in_message":[],"_revision_number":10},{"id":"b67469b200b0900b4a8ebdd34eb7c7a67e8f424b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 16:13:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"785945629d0b5fb87f160e1552db05ae70371155","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 16:15:24.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5666e7357824ad389c59e31e0893847 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e91c8be4492aee2327911ddea5b7decede388af9","revisions":{"11f706f24415a8d5a749671be7ae041688ae2791":{"kind":"REWORK","_number":1,"created":"2021-03-31 10:43:33.000000000","uploader":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"ref":"refs/changes/69/784069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/69/784069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/1"}}},"commit":{"parents":[{"commit":"1bb878049915c64dc588a8b8ef0e51c70825359a","subject":"Merge \"Update galera_wsrep_sst_auth_user to use galera_root_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bb878049915c64dc588a8b8ef0e51c70825359a"}]}],"author":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 10:32:37.000000000","tz":120},"committer":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 10:42:53.000000000","tz":120},"subject":"add support for encryption","message":"add support for encryption\n\nSigned-off-by: Sebastian Gumprich \u003csebastian.gumprich@t-systems.com\u003e\nChange-Id: I73e548ac208a96ddaa687a1b5fbb22cac20037d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/11f706f24415a8d5a749671be7ae041688ae2791"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/11f706f24415a8d5a749671be7ae041688ae2791"}]},"branch":"refs/heads/master"},"56ee28c4c412426c70d210e5ca64c2c015b0011e":{"kind":"REWORK","_number":2,"created":"2021-03-31 12:15:28.000000000","uploader":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"ref":"refs/changes/69/784069/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/69/784069/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/2"}}},"commit":{"parents":[{"commit":"1bb878049915c64dc588a8b8ef0e51c70825359a","subject":"Merge \"Update galera_wsrep_sst_auth_user to use galera_root_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bb878049915c64dc588a8b8ef0e51c70825359a"}]}],"author":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 10:32:37.000000000","tz":120},"committer":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 12:07:35.000000000","tz":120},"subject":"add support for encryption","message":"add support for encryption\n\nCloses-Bug: #1921861\n\nChange-Id: I73e548ac208a96ddaa687a1b5fbb22cac20037d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/56ee28c4c412426c70d210e5ca64c2c015b0011e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/56ee28c4c412426c70d210e5ca64c2c015b0011e"}]},"branch":"refs/heads/master"},"1a0627cf0bfd12590e0df16360a11ad8f4f1cf41":{"kind":"REWORK","_number":3,"created":"2021-03-31 14:17:27.000000000","uploader":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"ref":"refs/changes/69/784069/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/69/784069/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/3"}}},"commit":{"parents":[{"commit":"1bb878049915c64dc588a8b8ef0e51c70825359a","subject":"Merge \"Update galera_wsrep_sst_auth_user to use galera_root_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bb878049915c64dc588a8b8ef0e51c70825359a"}]}],"author":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 10:32:37.000000000","tz":120},"committer":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 14:17:03.000000000","tz":120},"subject":"add support for encryption","message":"add support for encryption\n\nCloses-Bug: #1921861\n\nChange-Id: I73e548ac208a96ddaa687a1b5fbb22cac20037d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1a0627cf0bfd12590e0df16360a11ad8f4f1cf41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1a0627cf0bfd12590e0df16360a11ad8f4f1cf41"}]},"branch":"refs/heads/master"},"198d34e323feb0d056f2e437f16d639245d78ce8":{"kind":"REWORK","_number":4,"created":"2021-03-31 14:20:44.000000000","uploader":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"ref":"refs/changes/69/784069/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/69/784069/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/4"}}},"commit":{"parents":[{"commit":"1bb878049915c64dc588a8b8ef0e51c70825359a","subject":"Merge \"Update galera_wsrep_sst_auth_user to use galera_root_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bb878049915c64dc588a8b8ef0e51c70825359a"}]}],"author":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 10:32:37.000000000","tz":120},"committer":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 14:20:35.000000000","tz":120},"subject":"add support for encryption","message":"add support for encryption\n\nCloses-Bug: #1921861\n\nChange-Id: I73e548ac208a96ddaa687a1b5fbb22cac20037d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/198d34e323feb0d056f2e437f16d639245d78ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/198d34e323feb0d056f2e437f16d639245d78ce8"}]},"branch":"refs/heads/master"},"e04971a3400472ffb0e7200fbbf1d4a302e782e4":{"kind":"REWORK","_number":5,"created":"2021-04-19 07:22:45.000000000","uploader":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"ref":"refs/changes/69/784069/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/69/784069/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/5"}}},"commit":{"parents":[{"commit":"1bb878049915c64dc588a8b8ef0e51c70825359a","subject":"Merge \"Update galera_wsrep_sst_auth_user to use galera_root_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bb878049915c64dc588a8b8ef0e51c70825359a"}]}],"author":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 10:32:37.000000000","tz":120},"committer":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-04-19 07:22:18.000000000","tz":120},"subject":"add support for encryption","message":"add support for encryption\n\nCloses-Bug: #1921861\n\nChange-Id: I73e548ac208a96ddaa687a1b5fbb22cac20037d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e04971a3400472ffb0e7200fbbf1d4a302e782e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e04971a3400472ffb0e7200fbbf1d4a302e782e4"}]},"branch":"refs/heads/master"},"a6a4fc9a68ef99732e3eba1ede36c4cc135d4412":{"kind":"REWORK","_number":6,"created":"2021-04-19 07:55:30.000000000","uploader":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"ref":"refs/changes/69/784069/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/69/784069/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/6"}}},"commit":{"parents":[{"commit":"1bb878049915c64dc588a8b8ef0e51c70825359a","subject":"Merge \"Update galera_wsrep_sst_auth_user to use galera_root_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bb878049915c64dc588a8b8ef0e51c70825359a"}]}],"author":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 10:32:37.000000000","tz":120},"committer":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-04-19 07:55:20.000000000","tz":120},"subject":"add support for encryption","message":"add support for encryption\n\nCloses-Bug: #1921861\n\nChange-Id: I73e548ac208a96ddaa687a1b5fbb22cac20037d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a6a4fc9a68ef99732e3eba1ede36c4cc135d4412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a6a4fc9a68ef99732e3eba1ede36c4cc135d4412"}]},"branch":"refs/heads/master"},"20d82a12093ef01aa7aec1a5898e661a841db939":{"kind":"REWORK","_number":7,"created":"2021-04-19 10:02:27.000000000","uploader":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"ref":"refs/changes/69/784069/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/69/784069/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/7"}}},"commit":{"parents":[{"commit":"1bb878049915c64dc588a8b8ef0e51c70825359a","subject":"Merge \"Update galera_wsrep_sst_auth_user to use galera_root_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bb878049915c64dc588a8b8ef0e51c70825359a"}]}],"author":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 10:32:37.000000000","tz":120},"committer":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-04-19 10:01:39.000000000","tz":120},"subject":"add support for encryption","message":"add support for encryption\n\nCloses-Bug: #1921861\n\nChange-Id: I73e548ac208a96ddaa687a1b5fbb22cac20037d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/20d82a12093ef01aa7aec1a5898e661a841db939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/20d82a12093ef01aa7aec1a5898e661a841db939"}]},"branch":"refs/heads/master"},"8d484d5c0f436ff1ded1276a5e8e759e7a73627c":{"kind":"REWORK","_number":8,"created":"2021-04-19 10:29:27.000000000","uploader":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"ref":"refs/changes/69/784069/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/69/784069/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/8"}}},"commit":{"parents":[{"commit":"1bb878049915c64dc588a8b8ef0e51c70825359a","subject":"Merge \"Update galera_wsrep_sst_auth_user to use galera_root_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bb878049915c64dc588a8b8ef0e51c70825359a"}]}],"author":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 10:32:37.000000000","tz":120},"committer":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-04-19 10:29:17.000000000","tz":120},"subject":"add support for encryption","message":"add support for encryption\n\nCloses-Bug: #1921861\n\nChange-Id: I73e548ac208a96ddaa687a1b5fbb22cac20037d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8d484d5c0f436ff1ded1276a5e8e759e7a73627c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8d484d5c0f436ff1ded1276a5e8e759e7a73627c"}]},"branch":"refs/heads/master"},"cd63ea85811b08499b556f26d3375a9e4f12d6b8":{"kind":"REWORK","_number":9,"created":"2021-04-20 05:03:01.000000000","uploader":{"_account_id":33299,"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","username":"zufallsheld"},"ref":"refs/changes/69/784069/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/69/784069/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/9"}}},"commit":{"parents":[{"commit":"1bb878049915c64dc588a8b8ef0e51c70825359a","subject":"Merge \"Update galera_wsrep_sst_auth_user to use galera_root_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bb878049915c64dc588a8b8ef0e51c70825359a"}]}],"author":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 10:32:37.000000000","tz":120},"committer":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-04-20 05:02:29.000000000","tz":120},"subject":"add support for encryption","message":"add support for encryption\n\nCloses-Bug: #1921861\n\nChange-Id: I73e548ac208a96ddaa687a1b5fbb22cac20037d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cd63ea85811b08499b556f26d3375a9e4f12d6b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cd63ea85811b08499b556f26d3375a9e4f12d6b8"}]},"branch":"refs/heads/master"},"e91c8be4492aee2327911ddea5b7decede388af9":{"kind":"REWORK","_number":10,"created":"2021-04-20 06:42:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/69/784069/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/69/784069/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/69/784069/10"}}},"commit":{"parents":[{"commit":"1bb878049915c64dc588a8b8ef0e51c70825359a","subject":"Merge \"Update galera_wsrep_sst_auth_user to use galera_root_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1bb878049915c64dc588a8b8ef0e51c70825359a"}]}],"author":{"name":"Sebastian Gumprich","email":"sebastian.gumprich@t-systems.com","date":"2021-03-31 10:32:37.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-20 06:42:28.000000000","tz":0},"subject":"add support for encryption","message":"add support for encryption\n\nCloses-Bug: #1921861\n\nChange-Id: I73e548ac208a96ddaa687a1b5fbb22cac20037d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e91c8be4492aee2327911ddea5b7decede388af9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e91c8be4492aee2327911ddea5b7decede388af9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
