)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~790426","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I1041865b1f488b05240722ff52773d51c151ad62","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"I1041865b1f488b05240722ff52773d51c151ad62","subject":"Prevent mariadb from OOM killing","status":"MERGED","created":"2021-05-10 12:12:42.000000000","updated":"2021-06-16 11:05:27.000000000","submitted":"2021-06-16 11:03:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790426","meta_rev_id":"08fd0ff8e1fadb5a3639ce1079aab7848a72c532","_number":790426,"virtual_id_number":790426,"owner":{"_account_id":27915,"name":"PerToft","email":"per@minfejl.dk","username":"PTO"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 11:03:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-06-11 16:10:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-14 21:37:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-14 21:37:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-10 13:33:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-11 16:10:17.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-06-14 21:37:01.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"ade978884f4c0c72894f0896fb585ca29b9a45e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27915,"name":"PerToft","email":"per@minfejl.dk","username":"PTO"},"date":"2021-05-10 12:12:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbcdd41fc8fda5a136cc84b9a5e3d87b6fe0a458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 13:33:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71b669abfdb84272a0751f1b292fa817 : SUCCESS in 5m 38s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83bb37d514b648af9652c977739cc582 : SUCCESS in 20m 51s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/914687ccf654416d85d9d0b3939e114b : SUCCESS in 1h 08m 28s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/5047fa7b934045e89f77d8e1b466ea65 : SUCCESS in 1h 05m 04s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/5fb0688916824e55a91937a193af0aee : SUCCESS in 58m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13a757d9d762418ebd259359d3075e77 : SUCCESS in 57m 18s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b4d4ba904cb64f3192ae858a6f6ac91e : SUCCESS in 1h 18m 10s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a27539d3c5b4b468db85df536ed4f00 : SUCCESS in 1h 16m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1201663cf46988eff3f258a95ea7997cce31703d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-11 14:21:38.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"dea9e5882718062342807b2b65674612c2f82fc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-11 16:09:46.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a31f8e1b9fe4e6418d6f4e31acc2ceb79ddf2c0d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-11 16:10:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4a6d07a2328b651a42ef6b7984ca0c268c0d9afd","author":{"_account_id":27915,"name":"PerToft","email":"per@minfejl.dk","username":"PTO"},"date":"2021-06-11 17:28:25.000000000","message":"Patch Set 3:\n\nI have just seen the situation i a real oom scenario and mariadb lived another day, but rabbitmq were shot dead. Maybe rabbit and memcache should have oom priority and have men heavy services like ceilometer moved to the frontline","accounts_in_message":[],"_revision_number":3},{"id":"342124895c8a53a43802cf14ed3956715ce251ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 20:25:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a1b465da91d40aaabf75fc298a01a40 : SUCCESS in 7m 37s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc8fce1011e5496e8169352542bb51dd : SUCCESS in 13m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/164c43a7fe26436d9cfe095cdd39a5f9 : SUCCESS in 1h 10m 24s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0520403782554464944ac0c0d31b5305 : SUCCESS in 1h 18m 38s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/ab13fb4993904ad98f244f4ddd3d01e8 : SUCCESS in 1h 05m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c6e2de39d46a4747bfd8f4d81fec3672 : SUCCESS in 1h 07m 46s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96756e43fe614c0ca2d5e7bf4a080ef0 : SUCCESS in 1h 02m 50s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/7a33991a11be43f1957cac4822a5f13b : SUCCESS in 1h 54m 57s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40d0d66722bd48ddb963e41b8d0ddafa : SUCCESS in 1h 17m 43s","accounts_in_message":[],"_revision_number":3},{"id":"40a4a6fd870cd9cda9997159a8a992146a8749f0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-12 05:42:31.000000000","message":"Patch Set 3:\n\nI think the most problematic part with mariadb being shot is that database will most likely be corrupted in this scenario. For rabbit/memcahed it\u0027s not really such big deal.","accounts_in_message":[],"_revision_number":3},{"id":"006661df802d8d9d2af110dc0405a7205aefebd0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-06-14 21:37:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4c3bebaeefe516fd6256e8f940ff3411754e1642","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 21:37:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cb5047ff5f2beea6c4a89e215dd4d719869bc16d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 00:43:54.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33e33870479b4fc88f761e34eb42ee04 : SUCCESS in 5m 41s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edf1ed83a70640d3b84ecd991d3c074e : SUCCESS in 22m 05s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/132d62fa0971478e8b92ea671f4e427f : SUCCESS in 1h 23m 33s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/486d45fbdd294f2fa9a705bc8983e69c : SUCCESS in 1h 17m 31s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/53127d57de8d40978a8d507898d915d2 : SUCCESS in 1h 07m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11fef4a7ddd14117b8533c90b6dad8c3 : SUCCESS in 1h 12m 59s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69136b2786a14bf382e359fd4f52a3c7 : TIMED_OUT in 3h 05m 29s","accounts_in_message":[],"_revision_number":3},{"id":"c1219058fdfce2d56bfe15bc807ba57f10c13536","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-15 07:28:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9192643a58433772b45818edbec7734462eda554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 10:44:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/257d1583a1c34c4b952dfcce43fe4667 : SUCCESS in 6m 08s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3fefe3420074767a749c8953d95697c : SUCCESS in 15m 04s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/f2e5d27488ba42e489bed8130be591ae : SUCCESS in 1h 11m 28s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bbced05f88ea41338526be6620d26b26 : SUCCESS in 1h 13m 54s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/9cb9e9f56dd64eb19169476785d05f5f : SUCCESS in 1h 05m 52s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/af123527003c4e8d87ae1092968e882e : SUCCESS in 1h 01m 58s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/205fc1a3c71f42f8a4891fb1d0b0127e : SUCCESS in 1h 06m 28s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/ddcf4dc526d94bc0802fc8ded87e679d : SUCCESS in 1h 38m 35s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e60b14479c6c45b3b8eb857339851b7b : TIMED_OUT in 3h 07m 06s","accounts_in_message":[],"_revision_number":3},{"id":"da2b78646826bb086365a7672576f00f0407a34b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-06-15 18:37:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"80018ffdf892fedfe809256e9a31ed81b63d61f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 20:13:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bba13e0556545068c14afc27cb57138 : SUCCESS in 7m 16s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e96491d022d40a2a1510667e0eb716b : SUCCESS in 13m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/e287e4df937a419ea501069ce451813b : SUCCESS in 1h 20m 01s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f89856961f654659ab54663fc5690790 : SUCCESS in 1h 15m 43s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/a8b9849516d3489c8bfd461ebe42633a : SUCCESS in 1h 22m 31s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/9b20de8e090145e7a702f60ba69c6614 : SUCCESS in 1h 03m 00s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a41a664cf8804a4aa59b1d3fa800ab86 : SUCCESS in 1h 06m 41s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/316671f27bcf4eaf92b9a67c06391bd6 : SUCCESS in 1h 31m 55s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d06bcad415b43f9999538e24b37921d : SUCCESS in 1h 28m 46s","accounts_in_message":[],"_revision_number":3},{"id":"945d4fa93a64e1ee02a47442e6435ae4ae551ce5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 20:15:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0b056444b92229881114e7a08c9b020ff202dab5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 21:45:35.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40806a13ecc240da9b0bf320bf18d516 : SUCCESS in 6m 36s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ada4474c2efb4e8180ed9355c10ad5d8 : SUCCESS in 13m 02s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/0e7e31baa39c4a12816f3e4e9ee4382e : SUCCESS in 1h 13m 19s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/89e7effb69e34fb3a54984011369dd29 : FAILURE in 1h 11m 00s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/b85f9d1eec9940958257ac65dd52f1ab : SUCCESS in 1h 12m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d6a60bdebca4e09b640fd3d9d766391 : SUCCESS in 1h 10m 04s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e833d9bfd9546d6982e81d908c9f839 : SUCCESS in 1h 28m 59s","accounts_in_message":[],"_revision_number":3},{"id":"a37ddc44368a21b83bbbb68e2baade03d5796198","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-16 08:03:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0640a770e960d49068bb9c4f370fae92dfbea4f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 09:34:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f48daf728d6e4133b66ed5e5776b2f1d : SUCCESS in 5m 59s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e313cf1e795247c8bc9c0248c7d72ae9 : SUCCESS in 14m 00s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/0486f31d4e7d4545b2481a4c34c407f4 : SUCCESS in 1h 11m 30s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4d71079133f442995a67ecc01bc1cf0 : SUCCESS in 1h 16m 21s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/ad262a0a5c07476e84b4d47b03d628bf : SUCCESS in 1h 05m 15s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/68d3d9f24ada4ada841272f0e3520c2d : SUCCESS in 1h 03m 45s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3239fb817b6e40908525d9c9991bf8cf : SUCCESS in 1h 09m 04s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/6c67c5b443604dd78227320f635ad0b7 : SUCCESS in 1h 26m 08s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c77372bfe844925a9118b9e56a50680 : SUCCESS in 1h 20m 11s","accounts_in_message":[],"_revision_number":3},{"id":"977b8a09bcd0c4fed17cfefedc96aa22263ed42c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 09:35:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4e6ce23112f7176bdf16349adfa08c2fe8663735","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 11:03:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5376df10be54a949806d84bf50a7246 : SUCCESS in 6m 35s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8f92f994de44d10a22171ade2d90dd1 : SUCCESS in 16m 03s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/23be7301f81e42dfae3a50e42a669670 : SUCCESS in 1h 21m 36s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a72a375dd6214cee94d2cd2bd4227770 : SUCCESS in 1h 23m 13s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c024c36a293e430883d7426db377da95 : SUCCESS in 1h 09m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b22ced01d3049da83d5d78835540171 : SUCCESS in 1h 13m 05s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9de30af52536473c919a75681b1e7bce : SUCCESS in 1h 18m 46s","accounts_in_message":[],"_revision_number":3},{"id":"0dac03cc4677f474e643a5765aa297092969826a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 11:03:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"08fd0ff8e1fadb5a3639ce1079aab7848a72c532","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 11:05:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b4a1fb1214f4bbb8a4af21cd4b70d51 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cd370e87e6acf3f77e0c1bd0a8a342039ec1d2d3","revisions":{"0e0492cfc2acf6f95f377adf43aa4fc9432d3e78":{"kind":"REWORK","_number":1,"created":"2021-05-10 12:12:42.000000000","uploader":{"_account_id":27915,"name":"PerToft","email":"per@minfejl.dk","username":"PTO"},"ref":"refs/changes/26/790426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/26/790426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/26/790426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/26/790426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/26/790426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/26/790426/1"}}},"commit":{"parents":[{"commit":"f6bd5c9e533bdd6364d658a3421fe3da2bd1afd9","subject":"Add galera devel packages installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f6bd5c9e533bdd6364d658a3421fe3da2bd1afd9"}]}],"author":{"name":"Per Abildgaard Toft","email":"per@minfejl.dk","date":"2021-05-10 12:10:18.000000000","tz":120},"committer":{"name":"Per Abildgaard Toft","email":"per@minfejl.dk","date":"2021-05-10 12:12:36.000000000","tz":120},"subject":"Added OOMScoreAdjust to systemd environment config to avoid mariadb galera to be killed in a oom situration Closes-Bug: #1927951","message":"Added OOMScoreAdjust to systemd environment config to avoid mariadb galera to be killed in a oom situration\nCloses-Bug: #1927951\n\nChange-Id: I1041865b1f488b05240722ff52773d51c151ad62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0e0492cfc2acf6f95f377adf43aa4fc9432d3e78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0e0492cfc2acf6f95f377adf43aa4fc9432d3e78"}]},"branch":"refs/heads/master"},"6b96eb76b453e193c90ff951b97e97f7ec8229ee":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-11 14:21:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/26/790426/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/26/790426/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/26/790426/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/26/790426/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/26/790426/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/26/790426/2"}}},"commit":{"parents":[{"commit":"f6bd5c9e533bdd6364d658a3421fe3da2bd1afd9","subject":"Add galera devel packages installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f6bd5c9e533bdd6364d658a3421fe3da2bd1afd9"}]}],"author":{"name":"Per Abildgaard Toft","email":"per@minfejl.dk","date":"2021-05-10 12:10:18.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-06-11 14:20:55.000000000","tz":0},"subject":"Prevent mariadb from OOM killing","message":"Prevent mariadb from OOM killing\n\nAdded OOMScoreAdjust to systemd environment config to avoid mariadb galera to be killed in a oom situration\n\nCloses-Bug: #1927951\nChange-Id: I1041865b1f488b05240722ff52773d51c151ad62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6b96eb76b453e193c90ff951b97e97f7ec8229ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6b96eb76b453e193c90ff951b97e97f7ec8229ee"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cd370e87e6acf3f77e0c1bd0a8a342039ec1d2d3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-06-11 16:09:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/26/790426/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/26/790426/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/26/790426/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/26/790426/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/26/790426/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/26/790426/3"}}},"commit":{"parents":[{"commit":"60de69f0ec7c4476fc7455e3a698f8d740cba6be","subject":"Merge \"Add support for Debian Bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/60de69f0ec7c4476fc7455e3a698f8d740cba6be"}]}],"author":{"name":"Per Abildgaard Toft","email":"per@minfejl.dk","date":"2021-05-10 12:10:18.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-06-11 16:09:46.000000000","tz":0},"subject":"Prevent mariadb from OOM killing","message":"Prevent mariadb from OOM killing\n\nAdded OOMScoreAdjust to systemd environment config to avoid mariadb galera to be killed in a oom situration\n\nCloses-Bug: #1927951\nChange-Id: I1041865b1f488b05240722ff52773d51c151ad62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cd370e87e6acf3f77e0c1bd0a8a342039ec1d2d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cd370e87e6acf3f77e0c1bd0a8a342039ec1d2d3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
