)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~796041","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I7b3b0f4da047f82a49266ef57fba2fbaa24cebdc","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"796041","hashtags":[],"change_id":"I7b3b0f4da047f82a49266ef57fba2fbaa24cebdc","subject":"Replace systemd unit overrides with role","status":"MERGED","created":"2021-06-11 16:08:10.000000000","updated":"2021-06-18 10:16:36.000000000","submitted":"2021-06-18 10:15:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796041-796041","meta_rev_id":"de2fabc17cd10790cdb25f65037bc614d4a7e2a0","_number":796041,"virtual_id_number":796041,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-18 10:15:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-06-18 08:40:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-17 13:50:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-06-18 08:40:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-11 17:10:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 18:37:09.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-06-18 08:40:35.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"1730df212a94ccedc660bccb947d429adbe009e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-11 16:08:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efc94fbc4578d5c28cff2f161d72b8136f67e49e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 17:10:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fba71f38eb6c47fa813c7752372c640b : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aca3e771324546dfbce6df99300a85f0 : SUCCESS in 4m 43s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b751bd5487244889a361e0bac75b864 : SUCCESS in 18m 40s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/09c8bd4eccfe41d7be8fd868f50b1c23 : FAILURE in 46m 57s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/476795b5ba5f4743b7588876dc72e7fd : FAILURE in 45m 42s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/b936d2cf27e74dba9528e1efb87bde7e : FAILURE in 34m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/528442cc77a84ea09a0889f9bd781bdd : LOST in 20m 01s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25c9fbfe2b0c407d9c49b45d9277381e : FAILURE in 36m 40s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/6e0d9c1aa3384067b550b48f9009ccab : FAILURE in 32m 49s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed2203096e1341e2997ff53ae3b04517 : FAILURE in 57m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0c7e234dbcfb000ae04a2baf10ad48c4d6a82202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-11 20:08:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2a4dd2f5a2e0faf3c0a352a409047c27a207aa90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 21:21:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d954f80556af4dfb99023dd2aa4b198f : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67269e13a1d040189d2ef1f31ac7e29b : SUCCESS in 3m 53s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36470060623a4f339ccfb8a0e31a9450 : SUCCESS in 12m 59s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/66cf249ede854d7aa79713baace9762c : FAILURE in 39m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1ab14aac3f94341859a2b0bdea073c3 : FAILURE in 39m 44s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/123f47a08f434b3484d39da6b55c7fc0 : FAILURE in 31m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/45ccc522273140079e72184e251679ba : FAILURE in 41m 06s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/591d75dd51e24e00a8dc35748cbe4102 : FAILURE in 35m 12s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/34870b22bd39402d8ed31582bcdd6fcc : FAILURE in 1h 07m 19s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4251e6782aae4f12be8d40403a28d85d : FAILURE in 1h 03m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d9ecbc8eafcd04b06afbc81d0a7bc938ffac5737","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-12 05:41:06.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6d2454e8ec66ec99ee80faedf76a09362bca3c0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-12 07:02:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69944c63c69944cebba33e884fca8e97 : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/331ea31ca8c44c929b4c1a929f5d8c21 : SUCCESS in 3m 37s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/050c8de38b214d9cbb68baf05c9cd924 : FAILURE in 11m 19s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/9f6ea40cee274c11899e747eea81b9cf : FAILURE in 41m 28s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7305fe0a178b4f32be6681affc205d47 : FAILURE in 43m 10s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/da846f4b82e5454eb2a4c922a01c56da : FAILURE in 33m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/920942c037324bcb87833fce2664d8d3 : FAILURE in 33m 10s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e893db5ca7f544bbbb960664c3e24b2e : FAILURE in 35m 14s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/61522d60b2604d28a54ff1fb8b4e7189 : FAILURE in 1h 16m 31s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11dbe621c2f548d08f1ae2b5b39682ed : FAILURE in 56m 36s","accounts_in_message":[],"_revision_number":3},{"id":"7c501ccd85dcfb262ef63fcfe166e457ff0c029b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-12 11:16:58.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d98a12991ac955c6e04e6a55422542b199e85927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-12 12:53:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e2a2ac5a7df4f08aca2a404963c748b : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4aa1369cc7264e9a9cc346c0dfccf6f8 : SUCCESS in 4m 25s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28549e825ff0459797bac97b7f3b9240 : SUCCESS in 16m 25s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/4c21bb0b7724459b862ec7bb1963346b : SUCCESS in 1h 01m 01s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/29165150f7af4508808505e638ae524b : SUCCESS in 1h 13m 32s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/88bf264b8e274f1a86aa117cb1854570 : SUCCESS in 1h 08m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/37b52c253d044900b92bd2d26fdbddbb : SUCCESS in 1h 22m 43s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/045123906bef43c887f16e1954227efb : SUCCESS in 1h 03m 24s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/af9425e3cd2047d4aafed92d9dc849fb : SUCCESS in 1h 31m 15s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2957183d6da42db963267523f0136b3 : SUCCESS in 1h 30m 10s","accounts_in_message":[],"_revision_number":4},{"id":"a0e95c7fcb6a4368f64d6de32b58fe3146d0246e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-12 18:49:23.000000000","message":"Patch Set 4:\n\nLooks okeyish https://zuul.opendev.org/t/openstack/build/88bf264b8e274f1a86aa117cb1854570/log/logs/etc/openstack/aio1_galera_container-04389e62/systemd/system/mariadb.service.d/override.conf.txt","accounts_in_message":[],"_revision_number":4},{"id":"afd5ca5bcc30448d9393af2a065a8d81803cebcc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-06-15 18:37:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8861d6ec8bf3f27add947960f61581101882e96c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-06-15 18:37:20.000000000","message":"Removed Code-Review+2 by Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"c5ac843956462b2ecbba049d855d055019080a0f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-06-15 18:38:01.000000000","message":"Patch Set 4: Code-Review+1\n\nMaybe needs rebasing on https://review.opendev.org/c/openstack/openstack-ansible-galera_server/+/790426 ?","accounts_in_message":[],"_revision_number":4},{"id":"cd967ae71466adda727e021f55f4ae61d4d2fd41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-06-16 12:40:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"45ef4eb778eb7f8c9b33376a2af1b88366793925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 14:27:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4578d6ce4d4f40ed88520fc6a2d21439 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b361d9e825bb4bb5885a1e7c8dd9cd6b : SUCCESS in 5m 02s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8f8b0e130574e228d4aae1356e063cd : SUCCESS in 18m 18s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/a38975647a77483ebe41ab35ee51171a : SUCCESS in 1h 24m 04s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/989d966820c94703906f209162432a0d : SUCCESS in 1h 16m 04s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/ce141407cc634eb5b95a8b607c30cf84 : SUCCESS in 1h 11m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e09573b7094a4181ae3c389c64c4bbbe : SUCCESS in 1h 07m 52s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f29d6148606b4d41a36b2f0e6fc47006 : SUCCESS in 1h 16m 59s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/2df59ed8eeaa41cd9b72876f9ae18b0a : SUCCESS in 1h 36m 13s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/773759577afc4c3bb48c683d05dfdb00 : SUCCESS in 1h 32m 02s","accounts_in_message":[],"_revision_number":5},{"id":"2d082a1991b4432999aa15404ad5ffe992c5f44e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-06-17 13:50:56.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"5c904c3e8cef629378886a4107086a9d24fd838a","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-06-18 08:40:35.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c90b6f3624776dd3fecfef189a2cad2bdd0cf5ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 08:41:29.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"60926fda30b6e0334dd927d977b8cbcec0d28949","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 10:15:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eea9fa4cdae74365b48f9ec5b47e3d47 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5521f27e2b3a4416976ed2c954585b10 : SUCCESS in 5m 03s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adadecad6b844bbb97d3194756b767b7 : SUCCESS in 17m 19s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/262d7c3109f642dab02cb47a3887a4f4 : SUCCESS in 1h 12m 28s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f9ad13c3764a4c58be8edbb58d7f0711 : SUCCESS in 1h 26m 01s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/50324069428e470f8ec9c5803554dc05 : SUCCESS in 1h 14m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25fc28e38b59458f8d3d7e5082db5a1b : SUCCESS in 1h 19m 51s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d58f96b84c68440fba53ea51769adc8d : SUCCESS in 1h 31m 06s","accounts_in_message":[],"_revision_number":5},{"id":"abc067c23f81bf8d963b11fbd69bc2eeac7c1254","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 10:15:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"de2fabc17cd10790cdb25f65037bc614d4a7e2a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 10:16:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c573027284540bbbdfb4b233801f967 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"818c5a71b04866445758b186bb51af7b7dcac7e0","revisions":{"587cfe26d3690389000822ace0aecfdccd63bc87":{"kind":"REWORK","_number":1,"created":"2021-06-11 16:08:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/796041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/41/796041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/1"}}},"commit":{"parents":[{"commit":"60de69f0ec7c4476fc7455e3a698f8d740cba6be","subject":"Merge \"Add support for Debian Bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/60de69f0ec7c4476fc7455e3a698f8d740cba6be"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-06-11 16:07:59.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-06-11 16:07:59.000000000","tz":180},"subject":"Replace systemd unit overrides with role","message":"Replace systemd unit overrides with role\n\nInstead of placing bunch of templates, we can use our systemd_role\nthat is capable of placing just overrides file, that will have same\nfunctionality but also provide ability to easily add required data into\nsystemd overrides.\n\nChange-Id: I7b3b0f4da047f82a49266ef57fba2fbaa24cebdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/587cfe26d3690389000822ace0aecfdccd63bc87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/587cfe26d3690389000822ace0aecfdccd63bc87"}]},"branch":"refs/heads/master"},"a468c50bdef52afa18c9816ea69cdc88a5196b46":{"kind":"REWORK","_number":2,"created":"2021-06-11 20:08:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/796041/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/41/796041/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/2"}}},"commit":{"parents":[{"commit":"60de69f0ec7c4476fc7455e3a698f8d740cba6be","subject":"Merge \"Add support for Debian Bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/60de69f0ec7c4476fc7455e3a698f8d740cba6be"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-06-11 16:07:59.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-06-11 20:08:51.000000000","tz":0},"subject":"Replace systemd unit overrides with role","message":"Replace systemd unit overrides with role\n\nInstead of placing bunch of templates, we can use our systemd_role\nthat is capable of placing just overrides file, that will have same\nfunctionality but also provide ability to easily add required data into\nsystemd overrides.\n\nChange-Id: I7b3b0f4da047f82a49266ef57fba2fbaa24cebdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a468c50bdef52afa18c9816ea69cdc88a5196b46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a468c50bdef52afa18c9816ea69cdc88a5196b46"}]},"branch":"refs/heads/master"},"45d690b4d2c55c2ec95c80fde6932f9b14d0d990":{"kind":"REWORK","_number":3,"created":"2021-06-12 05:41:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/796041/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/41/796041/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/3"}}},"commit":{"parents":[{"commit":"60de69f0ec7c4476fc7455e3a698f8d740cba6be","subject":"Merge \"Add support for Debian Bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/60de69f0ec7c4476fc7455e3a698f8d740cba6be"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-06-11 16:07:59.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-06-12 05:41:00.000000000","tz":0},"subject":"Replace systemd unit overrides with role","message":"Replace systemd unit overrides with role\n\nInstead of placing bunch of templates, we can use our systemd_role\nthat is capable of placing just overrides file, that will have same\nfunctionality but also provide ability to easily add required data into\nsystemd overrides.\n\nChange-Id: I7b3b0f4da047f82a49266ef57fba2fbaa24cebdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/45d690b4d2c55c2ec95c80fde6932f9b14d0d990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/45d690b4d2c55c2ec95c80fde6932f9b14d0d990"}]},"branch":"refs/heads/master"},"138faab1ec6b7af34ccddcf0e10c5fafc8a1962c":{"kind":"REWORK","_number":4,"created":"2021-06-12 11:16:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/796041/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/41/796041/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/4"}}},"commit":{"parents":[{"commit":"60de69f0ec7c4476fc7455e3a698f8d740cba6be","subject":"Merge \"Add support for Debian Bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/60de69f0ec7c4476fc7455e3a698f8d740cba6be"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-06-11 16:07:59.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-06-12 11:16:50.000000000","tz":0},"subject":"Replace systemd unit overrides with role","message":"Replace systemd unit overrides with role\n\nInstead of placing bunch of templates, we can use our systemd_role\nthat is capable of placing just overrides file, that will have same\nfunctionality but also provide ability to easily add required data into\nsystemd overrides.\n\nChange-Id: I7b3b0f4da047f82a49266ef57fba2fbaa24cebdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/138faab1ec6b7af34ccddcf0e10c5fafc8a1962c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/138faab1ec6b7af34ccddcf0e10c5fafc8a1962c"}]},"branch":"refs/heads/master"},"818c5a71b04866445758b186bb51af7b7dcac7e0":{"kind":"REWORK","_number":5,"created":"2021-06-16 12:40:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/796041/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/41/796041/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/796041/5"}}},"commit":{"parents":[{"commit":"dd47f7247eb6dba65127f7d7e31868af10c28477","subject":"Merge \"Prevent mariadb from OOM killing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dd47f7247eb6dba65127f7d7e31868af10c28477"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-06-11 16:07:59.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-06-16 12:40:29.000000000","tz":180},"subject":"Replace systemd unit overrides with role","message":"Replace systemd unit overrides with role\n\nInstead of placing bunch of templates, we can use our systemd_role\nthat is capable of placing just overrides file, that will have same\nfunctionality but also provide ability to easily add required data into\nsystemd overrides.\n\nChange-Id: I7b3b0f4da047f82a49266ef57fba2fbaa24cebdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/818c5a71b04866445758b186bb51af7b7dcac7e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/818c5a71b04866445758b186bb51af7b7dcac7e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
