)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~804371","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Iedd3cba5f1ccb6bdc4990d839a86e23825480d2f","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"Iedd3cba5f1ccb6bdc4990d839a86e23825480d2f","subject":"Partial Revert \"Bump MariaDB version to 10.5.9\"","status":"ABANDONED","created":"2021-08-12 09:45:03.000000000","updated":"2021-08-16 11:46:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6a33c265670e98cdec219f5dd8444841274e6b64","_number":804371,"virtual_id_number":804371,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-12 14:39:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-08-16 07:03:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-12 11:17:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-16 07:03:43.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"569334840b7a2f655826e0f58d77ba11837eaadd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-08-12 09:45:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5836f3225df7cfffaba00277778b634d080a4199","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 11:17:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70214f2bf2ee4b03ab211a0e9ad97cea : SUCCESS in 5m 23s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/285413b89e4448379bdf973d631ec499 : SUCCESS in 14m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/2a13d72573344e4aa3331321f24f201a : SUCCESS in 1h 14m 27s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/caa2323250214941b62bc3eaf9e0864e : SUCCESS in 1h 13m 39s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e379963856124f40980829e4948d66af : SUCCESS in 1h 07m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d78b0d3f65364793a64241c9e3edc226 : FAILURE in 34m 00s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0716e939f59a4c90a45e7c433e2b3651 : SUCCESS in 1h 08m 38s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b1f088ab9ae447c18266703dcaff9d7a : SUCCESS in 1h 27m 09s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0134e468e0f440f1a59965ae0defa665 : SUCCESS in 1h 23m 46s","accounts_in_message":[],"_revision_number":1},{"id":"71a5eba2b234f970907379ead4b0ec0b855b3d82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-08-12 12:50:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c66019c1f3b5dc56a9c2d4680d04dd394b8f994d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 14:39:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/731a698581ef4a9ba7e5b5b3e59b8193 : SUCCESS in 7m 18s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/503a89f914de4b54bf3ec019ba37ab78 : SUCCESS in 14m 14s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b7c59e54b4f348ffafe9c4baa25485a3 : SUCCESS in 1h 44m 50s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/13df4bcc722b40808842404b8c3c983d : SUCCESS in 1h 25m 00s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/fff7426be3404552bc4ae2eb7f9a0a77 : SUCCESS in 1h 07m 13s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fea2d46680a94c4e8f41f099eb1fba8e : SUCCESS in 1h 07m 43s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0fad717430b4f0d867ed458ac5427e9 : SUCCESS in 1h 06m 52s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/a73765a27fc747429a010ff3e69f891d : SUCCESS in 1h 41m 07s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f52c1b9e5b9c4015ad75a85836d4f50d : SUCCESS in 1h 25m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9c98d2d31dc80c677a11c7e19e87a14320b92812","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-16 07:03:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6a33c265670e98cdec219f5dd8444841274e6b64","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-16 11:46:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d5d77b33b14697bcbecd992f55d7081ec62b744","revisions":{"dd448cd4bc2d057a68aba73b201587dbca5d728b":{"kind":"REWORK","_number":1,"created":"2021-08-12 09:45:03.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/71/804371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/71/804371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/71/804371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/71/804371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/71/804371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/71/804371/1"}}},"commit":{"parents":[{"commit":"debc89a21737d9c2f9cac083b63d886c74988390","subject":"Update galera to 10.6.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/debc89a21737d9c2f9cac083b63d886c74988390"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-08-11 17:53:15.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-08-12 09:44:10.000000000","tz":60},"subject":"Partial Revert \"Bump MariaDB version to 10.5.9\"","message":"Partial Revert \"Bump MariaDB version to 10.5.9\"\n\nThis revert removes the workarounds needed for 10.5.9 due to\nroot user losing ability to grant privs.\n\nThis reverts commit 1db0ef2efbd22b3491e1cf57e172a3d87b6f62a9.\n\nChange-Id: Iedd3cba5f1ccb6bdc4990d839a86e23825480d2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dd448cd4bc2d057a68aba73b201587dbca5d728b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dd448cd4bc2d057a68aba73b201587dbca5d728b"}]},"branch":"refs/heads/master"},"8d5d77b33b14697bcbecd992f55d7081ec62b744":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-08-12 12:50:52.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/71/804371/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/71/804371/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/71/804371/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/71/804371/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/71/804371/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/71/804371/2"}}},"commit":{"parents":[{"commit":"083b571b7c884f29ecf2aa34ccf56583fae83dfc","subject":"Update galera to 10.6.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/083b571b7c884f29ecf2aa34ccf56583fae83dfc"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-08-11 17:53:15.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-08-12 12:50:52.000000000","tz":0},"subject":"Partial Revert \"Bump MariaDB version to 10.5.9\"","message":"Partial Revert \"Bump MariaDB version to 10.5.9\"\n\nThis revert removes the workarounds needed for 10.5.9 due to\nroot user losing ability to grant privs.\n\nThis reverts commit 1db0ef2efbd22b3491e1cf57e172a3d87b6f62a9.\n\nChange-Id: Iedd3cba5f1ccb6bdc4990d839a86e23825480d2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8d5d77b33b14697bcbecd992f55d7081ec62b744"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8d5d77b33b14697bcbecd992f55d7081ec62b744"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
