)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~807717","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I72738e4f8bd2233dedbed4428baafd4436de84b5","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"osa/galera_pki","hashtags":[],"change_id":"I72738e4f8bd2233dedbed4428baafd4436de84b5","subject":"Use ansible-role-pki to generate SSL certificates","status":"MERGED","created":"2021-09-07 12:55:30.000000000","updated":"2021-10-07 12:00:12.000000000","submitted":"2021-10-07 11:58:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"807717-osa/galera_pki","meta_rev_id":"4073c7b519607f919117afed498332bdb78426bd","_number":807717,"virtual_id_number":807717,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-07 11:58:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2021-09-29 16:22:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2021-10-05 17:28:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":1,"date":"2021-10-05 17:28:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2021-10-07 10:10:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":31749,"name":"James Gibson","email":"james.gibson@bbc.co.uk","username":"jamesgibo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-07 15:51:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-07 19:41:40.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2021-09-07 19:41:45.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-09-29 16:22:59.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2021-10-05 17:28:07.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2021-10-07 10:04:30.000000000","updated_by":{"_account_id":31749,"name":"James Gibson","email":"james.gibson@bbc.co.uk","username":"jamesgibo"},"reviewer":{"_account_id":31749,"name":"James Gibson","email":"james.gibson@bbc.co.uk","username":"jamesgibo"},"state":"CC"}],"messages":[{"id":"122148c73934414666670d1a6865c46b991ed48f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-07 12:55:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"feafbb88e5ae24d4f8615a176f45fd43596e4640","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-07 13:20:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a23662b3d4435f1a52e99236ec1daf1769220f11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 15:51:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2411af78436423d801043ae9d436d07 : SUCCESS in 6m 25s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/745a0e9ee49d41f0883bc5673bc7db25 : SUCCESS in 21m 53s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/19cccf84ebe849a9a67a0f50be9ca06a : SUCCESS in 1h 18m 33s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/260d903e1a09408e90313fdabf9f24b0 : SUCCESS in 1h 18m 34s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/ac4a05d07f824374bfdb285cfd9f1765 : SUCCESS in 1h 30m 52s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/26369f7ea6ba4f7db03e0b150e7f63da : SUCCESS in 1h 14m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efb93a1bc71b4bf889c861d934d4a319 : SUCCESS in 1h 22m 23s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b52e543c754d4f79bc690e4b1a1fed87 : SUCCESS in 2h 14m 16s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cc478b123834bc9bde1f941cd98612e : SUCCESS in 1h 57m 39s","accounts_in_message":[],"_revision_number":2},{"id":"bb59903a6996dfd4b1212bd6ec6656e6a199f578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-07 16:15:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d7f899d4819fd33ec86e3b22d9fed59a90d80a14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-07 16:34:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dd19a6e6506a43ffac8699fc339a996c19b4ee0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 18:15:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/204ef4766e494197b19b3d3e91f534cf : SUCCESS in 5m 51s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0393deb13027436aad10daa9aab114dd : SUCCESS in 20m 39s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/f85c06ed53cd472c818fb73fd7eeb374 : SUCCESS in 1h 23m 41s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9970d772ed0a4903a1a85130626d1bf2 : SUCCESS in 1h 23m 13s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/7ac589265a114a83a47e44c9012ce151 : SUCCESS in 1h 12m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/2497cdac841a46f4bdf8e2bafc280da4 : SUCCESS in 1h 12m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f5006d261b94ca0b57751a4e6129578 : SUCCESS in 1h 17m 13s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1933e81f920495b933fa74533d23489 : SUCCESS in 1h 32m 32s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3494f32ed3614a598dfbb85a0ce073a9 : SUCCESS in 1h 26m 56s","accounts_in_message":[],"_revision_number":4},{"id":"bee17f016fd5d34504269fc8461c874a85471e09","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-09-07 19:41:40.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nrabbit -\u003e galera/mariadb as needed","accounts_in_message":[],"_revision_number":4},{"id":"24f0ceddd0463aa29b90c972c3712018b462bcb1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-09-07 19:41:45.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"79ab215695d0caca475aa014363071df5aceed71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-08 14:33:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a8b03416e9515cb47a8a1b2f0e1807a1bdbf379a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 16:11:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b5514b36c0a4e7db1392193fc45e315 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1feba86fc764fe9a2f258a9a4daf45d : FAILURE in 5m 06s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d5b2ccfd9ef4f1ea255d5aa8206337f : SUCCESS in 23m 34s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/365aed34228740caa409f186e1faf815 : FAILURE in 1h 18m 14s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/819bffb1004e48529590e6976f628ac9 : FAILURE in 1h 22m 52s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c3051a8fa1ee45ada3e900f636ea4165 : FAILURE in 1h 14m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/bb880e8adf474e5a82c77cbf036eb791 : FAILURE in 1h 07m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a09b78950c564c8a8f736850bc2e079e : FAILURE in 1h 09m 57s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/34cfd9fea2b64c90b4278ff42112a82b : FAILURE in 1h 34m 20s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/122bbe055995447e96baa96527b16e65 : FAILURE in 1h 09m 46s","accounts_in_message":[],"_revision_number":5},{"id":"a61a432d542225ccfcde06472e5c0cd983b0bc20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-09 10:18:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"238d36114c4acf57917caf51cee5b413a2f70f56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 12:38:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/966146eeecfb45dab5802e9f2369f11d : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3bacc543bd548bd988f7629001a246e : FAILURE in 4m 28s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ad34695f6e74bcca61bf74d5134d5ee : SUCCESS in 11m 44s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/3dc5dee4dcf94ab9ac337b2db53fe2d6 : SUCCESS in 1h 19m 37s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/fdddc944581a41fd8afb5080d699350e : SUCCESS in 1h 11m 54s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/fbf5316a6bf543d2ad7d33181f071892 : SUCCESS in 1h 04m 57s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/371b7cf0dd724375870f4736cd077b0e : SUCCESS in 59m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee6d0789502841eb8c2edb1d374bbfcf : SUCCESS in 1h 03m 38s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2287580358f0431892452912ba898d4d : SUCCESS in 2h 12m 11s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d422bb4db5ce434b9518561043bb98b9 : SUCCESS in 1h 27m 32s","accounts_in_message":[],"_revision_number":6},{"id":"617adca9fe9db54dcca20a3791f9f78e557b8073","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-09 14:47:50.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"e271461e49f8891b85a6afed011d18a82f14a9f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-09 15:21:43.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"89a9ca49502645e022d294a7a1d84f0bcd9264eb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-09 16:29:22.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"7f760524c84776ac23eb2ed425266a07f51b71a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 17:01:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a72c504559f495fb3391ff40f13c03a : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5f141c39a7e4b78966d094af27bc4e7 : SUCCESS in 4m 58s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27e73e8932e04186aedb71d7afcc01a1 : SUCCESS in 17m 30s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/cd966fd8d58a431b9a480e8e8fa1b3bd : SUCCESS in 1h 21m 54s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/832ddd6d38a64d70a2c3290f6e8c2815 : SUCCESS in 1h 30m 54s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/0ce8f4cae6be4881be46a8210fbade84 : SUCCESS in 1h 02m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b2404a3039b043c1aa66ca9c32f87941 : SUCCESS in 1h 02m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/255281d23592427e8b9269a2596d3aea : SUCCESS in 1h 08m 36s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2275db0e930844d38589d98a10de3818 : SUCCESS in 1h 28m 16s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c239617304f4e878100de887f28f1ac : SUCCESS in 1h 27m 58s","accounts_in_message":[],"_revision_number":8},{"id":"25a394a5bd2285d0dd58dd0849de2560b77ae2b4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-14 16:50:49.000000000","message":"Patch Set 8: -Workflow\n\nin this shape it\u0027s known to work.\n\nhttps://review.opendev.org/c/openstack/openstack-ansible/+/807880 passes LXC tests, nd fails metal because of the concept overall. we would need to refactor roles to get this working in easy manner everywhere.","accounts_in_message":[],"_revision_number":8},{"id":"a1db561233c55eaa4aab2674b8aea75155798de1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-17 09:51:55.000000000","message":"Topic 807717 removed","accounts_in_message":[],"_revision_number":8},{"id":"6c88923fed6f913a1df6e6d5df575683c53e7e4a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-17 09:52:52.000000000","message":"Topic set to osa/galera_pki","accounts_in_message":[],"_revision_number":8},{"id":"554cbd3949700e1379419cf00f6640f41747dda9","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-09-29 16:22:59.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"a92da5729c15de0ea14210d6f576988e92e28255","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-10-05 17:28:07.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"390e17159980e26dc1ca77eade45048fabc4e1bc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-10-07 10:10:49.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"482e3781f48c2ad9a5bb1830e33628c5e1e5d9fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 10:16:52.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"dfa66c00453d0c0f3bf09ae54861ae7c88e771ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 11:58:16.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60441532bb9640d3ac765b311d3d0731 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1922ab6fc14c423297725b35d77def0f : SUCCESS in 5m 26s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9da9423c3d3149b59d4147653f9688e7 : SUCCESS in 19m 13s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/c3d9e18cc27848b798e98d1fb19831a5 : SUCCESS in 1h 18m 11s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/765d6de6657e440db617938049a2f7cd : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/1294b6cf952a4db2ad8f852d303aee98 : SUCCESS in 1h 15m 44s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8a76eaf85a46478f9af48364213f1d89 : SUCCESS in 1h 09m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/573dfea4b89b45a88df4e6741bd43d4e : SUCCESS in 1h 10m 50s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2731f38692ea404c969df9d2da3d5a4e : SUCCESS in 1h 37m 09s","accounts_in_message":[],"_revision_number":8},{"id":"7ffc12fda356548d0a51ff798a5807984a7ab703","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 11:58:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"4073c7b519607f919117afed498332bdb78426bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 12:00:12.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bcba9abae1e4d10b399f7f69a14c6f0 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6bc6929d09a31e91ef962519bf5284620747e441","revisions":{"15bf3d661d624831ec0f1a7c46a0e156e6dc654f":{"kind":"REWORK","_number":1,"created":"2021-09-07 12:55:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/807717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/17/807717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/1"}}},"commit":{"parents":[{"commit":"4b66580dadea0a45419095c0a488399d0a3d7125","subject":"Revert \"Update galera to 10.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4b66580dadea0a45419095c0a488399d0a3d7125"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-07 12:55:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-07 12:55:20.000000000","tz":180},"subject":"Use ansible-role-pki to generate SSL certificates","message":"Use ansible-role-pki to generate SSL certificates\n\nSupports two scenarios:\n\n1) variables defined in defaults/main.yml are sufficient to create\na root/intermediate CA certificate for rabbitmq when this role\nis used outside openstack-ansible.\n\n2) when:\n\nopenstack_pki_dir\nopenstack_pki_setup_host\nopenstack_pki_authorities\nopenstack_pki_service_intermediate_cert_name\n\nare defined, an external CA already created on the deploy host\nwith a previous run of ansible-role-pki will be used as the CA.\n\nServer certificates for the rabbitmq instances are created from the\ndata in galera_pki_certificates in both situations\n\nChange-Id: I72738e4f8bd2233dedbed4428baafd4436de84b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/15bf3d661d624831ec0f1a7c46a0e156e6dc654f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/15bf3d661d624831ec0f1a7c46a0e156e6dc654f"}]},"branch":"refs/heads/master"},"17ac5189f538edd16c53cc3864316286b15e53c4":{"kind":"REWORK","_number":2,"created":"2021-09-07 13:20:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/807717/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/17/807717/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/2"}}},"commit":{"parents":[{"commit":"4b66580dadea0a45419095c0a488399d0a3d7125","subject":"Revert \"Update galera to 10.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4b66580dadea0a45419095c0a488399d0a3d7125"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-07 12:55:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-07 13:19:54.000000000","tz":180},"subject":"Use ansible-role-pki to generate SSL certificates","message":"Use ansible-role-pki to generate SSL certificates\n\nSupports two scenarios:\n\n1) variables defined in defaults/main.yml are sufficient to create\na root/intermediate CA certificate for rabbitmq when this role\nis used outside openstack-ansible.\n\n2) when:\n\nopenstack_pki_dir\nopenstack_pki_setup_host\nopenstack_pki_authorities\nopenstack_pki_service_intermediate_cert_name\n\nare defined, an external CA already created on the deploy host\nwith a previous run of ansible-role-pki will be used as the CA.\n\nServer certificates for the rabbitmq instances are created from the\ndata in galera_pki_certificates in both situations\n\nChange-Id: I72738e4f8bd2233dedbed4428baafd4436de84b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/17ac5189f538edd16c53cc3864316286b15e53c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/17ac5189f538edd16c53cc3864316286b15e53c4"}]},"branch":"refs/heads/master"},"1d6e8b28e02e28feaf125e32318e90454b9f3bb3":{"kind":"REWORK","_number":3,"created":"2021-09-07 16:15:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/807717/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/17/807717/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/3"}}},"commit":{"parents":[{"commit":"4b66580dadea0a45419095c0a488399d0a3d7125","subject":"Revert \"Update galera to 10.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4b66580dadea0a45419095c0a488399d0a3d7125"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-07 12:55:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-07 16:15:07.000000000","tz":180},"subject":"Use ansible-role-pki to generate SSL certificates","message":"Use ansible-role-pki to generate SSL certificates\n\nSupports two scenarios:\n\n1) variables defined in defaults/main.yml are sufficient to create\na root/intermediate CA certificate for rabbitmq when this role\nis used outside openstack-ansible.\n\n2) when:\n\nopenstack_pki_dir\nopenstack_pki_setup_host\nopenstack_pki_authorities\nopenstack_pki_service_intermediate_cert_name\n\nare defined, an external CA already created on the deploy host\nwith a previous run of ansible-role-pki will be used as the CA.\n\nServer certificates for the rabbitmq instances are created from the\ndata in galera_pki_certificates in both situations\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/807771\nChange-Id: I72738e4f8bd2233dedbed4428baafd4436de84b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1d6e8b28e02e28feaf125e32318e90454b9f3bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1d6e8b28e02e28feaf125e32318e90454b9f3bb3"}]},"branch":"refs/heads/master"},"390d7ab1c4ea5ca8d2565347ba35b00b64ac3a05":{"kind":"REWORK","_number":4,"created":"2021-09-07 16:34:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/807717/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/17/807717/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/4"}}},"commit":{"parents":[{"commit":"4b66580dadea0a45419095c0a488399d0a3d7125","subject":"Revert \"Update galera to 10.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4b66580dadea0a45419095c0a488399d0a3d7125"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-07 12:55:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-07 16:34:17.000000000","tz":180},"subject":"Use ansible-role-pki to generate SSL certificates","message":"Use ansible-role-pki to generate SSL certificates\n\nSupports two scenarios:\n\n1) variables defined in defaults/main.yml are sufficient to create\na root/intermediate CA certificate for rabbitmq when this role\nis used outside openstack-ansible.\n\n2) when:\n\nopenstack_pki_dir\nopenstack_pki_setup_host\nopenstack_pki_authorities\nopenstack_pki_service_intermediate_cert_name\n\nare defined, an external CA already created on the deploy host\nwith a previous run of ansible-role-pki will be used as the CA.\n\nServer certificates for the rabbitmq instances are created from the\ndata in galera_pki_certificates in both situations\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/807771\nChange-Id: I72738e4f8bd2233dedbed4428baafd4436de84b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/390d7ab1c4ea5ca8d2565347ba35b00b64ac3a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/390d7ab1c4ea5ca8d2565347ba35b00b64ac3a05"}]},"branch":"refs/heads/master"},"e44ff9624e475749314c54051742fc834a61ab77":{"kind":"REWORK","_number":5,"created":"2021-09-08 14:33:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/807717/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/17/807717/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/5"}}},"commit":{"parents":[{"commit":"4b66580dadea0a45419095c0a488399d0a3d7125","subject":"Revert \"Update galera to 10.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4b66580dadea0a45419095c0a488399d0a3d7125"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-07 12:55:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-08 14:33:14.000000000","tz":180},"subject":"Use ansible-role-pki to generate SSL certificates","message":"Use ansible-role-pki to generate SSL certificates\n\nSupports two scenarios:\n\n1) variables defined in defaults/main.yml are sufficient to create\na root/intermediate CA certificate for mariadb when this role\nis used outside openstack-ansible.\n\n2) when:\n\nopenstack_pki_dir\nopenstack_pki_setup_host\nopenstack_pki_authorities\nopenstack_pki_service_intermediate_cert_name\n\nare defined, an external CA already created on the deploy host\nwith a previous run of ansible-role-pki will be used as the CA.\n\nServer certificates for the galera instances are created from the\ndata in galera_pki_certificates in both situations\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/807771\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/807880\nChange-Id: I72738e4f8bd2233dedbed4428baafd4436de84b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e44ff9624e475749314c54051742fc834a61ab77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e44ff9624e475749314c54051742fc834a61ab77"}]},"branch":"refs/heads/master"},"6060d413f9ba764ee2231273aa0de18da84509a0":{"kind":"REWORK","_number":6,"created":"2021-09-09 10:18:56.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/807717/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/17/807717/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/6"}}},"commit":{"parents":[{"commit":"4b66580dadea0a45419095c0a488399d0a3d7125","subject":"Revert \"Update galera to 10.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4b66580dadea0a45419095c0a488399d0a3d7125"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-07 12:55:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-09 10:18:36.000000000","tz":180},"subject":"Use ansible-role-pki to generate SSL certificates","message":"Use ansible-role-pki to generate SSL certificates\n\nSupports two scenarios:\n\n1) variables defined in defaults/main.yml are sufficient to create\na root/intermediate CA certificate for mariadb when this role\nis used outside openstack-ansible.\n\n2) when:\n\nopenstack_pki_dir\nopenstack_pki_setup_host\nopenstack_pki_authorities\nopenstack_pki_service_intermediate_cert_name\n\nare defined, an external CA already created on the deploy host\nwith a previous run of ansible-role-pki will be used as the CA.\n\nServer certificates for the galera instances are created from the\ndata in galera_pki_certificates in both situations\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/807771\nChange-Id: I72738e4f8bd2233dedbed4428baafd4436de84b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6060d413f9ba764ee2231273aa0de18da84509a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6060d413f9ba764ee2231273aa0de18da84509a0"}]},"branch":"refs/heads/master"},"026adfce192a71b9004e6de33301822b6ebd08e1":{"kind":"REWORK","_number":7,"created":"2021-09-09 14:47:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/807717/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/17/807717/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/7"}}},"commit":{"parents":[{"commit":"4b66580dadea0a45419095c0a488399d0a3d7125","subject":"Revert \"Update galera to 10.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4b66580dadea0a45419095c0a488399d0a3d7125"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-07 12:55:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-09 14:47:47.000000000","tz":0},"subject":"Use ansible-role-pki to generate SSL certificates","message":"Use ansible-role-pki to generate SSL certificates\n\nSupports two scenarios:\n\n1) variables defined in defaults/main.yml are sufficient to create\na root/intermediate CA certificate for mariadb when this role\nis used outside openstack-ansible.\n\n2) when:\n\nopenstack_pki_dir\nopenstack_pki_setup_host\nopenstack_pki_authorities\nopenstack_pki_service_intermediate_cert_name\n\nare defined, an external CA already created on the deploy host\nwith a previous run of ansible-role-pki will be used as the CA.\n\nServer certificates for the galera instances are created from the\ndata in galera_pki_certificates in both situations\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/807771\nChange-Id: I72738e4f8bd2233dedbed4428baafd4436de84b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/026adfce192a71b9004e6de33301822b6ebd08e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/026adfce192a71b9004e6de33301822b6ebd08e1"}]},"branch":"refs/heads/master"},"6bc6929d09a31e91ef962519bf5284620747e441":{"kind":"REWORK","_number":8,"created":"2021-09-09 15:21:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/807717/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/17/807717/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/17/807717/8"}}},"commit":{"parents":[{"commit":"4b66580dadea0a45419095c0a488399d0a3d7125","subject":"Revert \"Update galera to 10.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4b66580dadea0a45419095c0a488399d0a3d7125"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-07 12:55:20.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-09 15:21:38.000000000","tz":0},"subject":"Use ansible-role-pki to generate SSL certificates","message":"Use ansible-role-pki to generate SSL certificates\n\nSupports two scenarios:\n\n1) variables defined in defaults/main.yml are sufficient to create\na root/intermediate CA certificate for mariadb when this role\nis used outside openstack-ansible.\n\n2) when:\n\nopenstack_pki_dir\nopenstack_pki_setup_host\nopenstack_pki_authorities\nopenstack_pki_service_intermediate_cert_name\n\nare defined, an external CA already created on the deploy host\nwith a previous run of ansible-role-pki will be used as the CA.\n\nServer certificates for the galera instances are created from the\ndata in galera_pki_certificates in both situations\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/807771\nChange-Id: I72738e4f8bd2233dedbed4428baafd4436de84b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6bc6929d09a31e91ef962519bf5284620747e441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6bc6929d09a31e91ef962519bf5284620747e441"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
