)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~819425","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I9450912ec7960a8ab713517532164cab52628b30","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"db-pooling","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2021-12-02 17:23:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9450912ec7960a8ab713517532164cab52628b30","subject":"Inherit default galera_wait_timeout value from openstack_db_connection_recycle_time","status":"MERGED","created":"2021-11-26 11:10:48.000000000","updated":"2021-12-02 17:25:11.000000000","submitted":"2021-12-02 17:23:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"819425-db-pooling","meta_rev_id":"e4f12fbc8c06d30a01d4e677250ea54fb7770cf9","_number":819425,"virtual_id_number":819425,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-02 17:23:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-12-02 11:36:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-02 15:54:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-02 15:54:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-26 11:59:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-02 10:49:30.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2021-12-02 11:36:14.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-12-02 15:54:24.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"659874d764284cf6eccb36a3da386be81d8e8051","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-11-26 11:10:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd79abe2f35bc730e238bb3c915855ff36b87abf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-11-26 11:25:44.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"44a09b6d797d1fbe6f801d3aa78dbc5536b550d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 11:59:45.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"76c4d8f7a2d5349d83029556f7489c26825b511a","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-12-02 10:49:30.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"89f58af8ea6a804e62b352dc0a961e0651ff2417","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-02 11:15:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"060d644eb75e93f3a9c59b15468c7e64ccd069e3","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-12-02 11:16:41.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"a96ae7d5c718de69986f4a19a89540c374853950","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-02 11:36:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"14756226b4b4cdcf5b9dd0a5c03ac82ad1124d4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 13:01:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1700d0f37f5b48deba7cfb0163a2435e : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07f3f70472cf4712bf99d614703c55ae : SUCCESS in 4m 12s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51833e5b1bfa41aab029f6285f6be0a6 : SUCCESS in 14m 41s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/2fedcbd33eda44b793f7480aa8d9334a : SUCCESS in 1h 17m 18s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f3c03f01dca4285939b6ec2f53d4d9b : SUCCESS in 1h 21m 41s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/f25f2dae24da45db95eaa4219bd07874 : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/41bb151a3f9844829cdf517e89564a99 : SUCCESS in 1h 07m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37aba85cd1c147d4b5202b7a6862f027 : SUCCESS in 1h 12m 00s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f487120034c0442e85a930a4d5808303 : SUCCESS in 1h 42m 31s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bb0ccc167fe4f0d942a4f3c17ee9a39 : SUCCESS in 1h 30m 35s","accounts_in_message":[],"_revision_number":3},{"id":"9c3343510b58e2490dae98fe589af8b4081ecd0c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-02 15:54:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8dcaa5e47b1efb987aa2c4e0fd84e1cf7c05c155","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 15:57:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d0cc7c7a70d2e9affdd8c3429754bdf07298dbf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 17:23:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d64956e87b48f1af08285f99b82a9c : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b9a6c9a54e643bca0d91f35b9e7e7b3 : SUCCESS in 3m 47s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d691596a91dd4b63b46cf46221d5405a : SUCCESS in 25m 37s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/994162d3f2694c9ca793346e0c86d2ad : SUCCESS in 1h 06m 51s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/011d384a3b8a46d9bbba163ca6d3dd0d : SUCCESS in 1h 24m 29s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/b089cf7e69df4a288f847ed77b3f403d : SUCCESS in 1h 03m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b5bbb18c22c74023b55cb87da50783c4 : SUCCESS in 55m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1edf8bd688b42f88cb778a12d121162 : SUCCESS in 1h 07m 24s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfc695110d684fc3935b5bf9e89c5990 : SUCCESS in 1h 11m 01s","accounts_in_message":[],"_revision_number":3},{"id":"49a87444142b6ddfe65d18ae94b0e0580d36e615","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 17:23:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e4f12fbc8c06d30a01d4e677250ea54fb7770cf9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 17:25:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/212bc7d353b1467da14e5272132ec9d4 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b79574382bf9a65803c03b4b68b8b8055893acc2","revisions":{"a3d71802c853c44f0d79ada90a0f0b56c584bbbb":{"kind":"REWORK","_number":1,"created":"2021-11-26 11:10:48.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/25/819425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/25/819425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/25/819425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/25/819425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/25/819425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/25/819425/1"}}},"commit":{"parents":[{"commit":"d9a21b14331fa4533114bc94f3525fbce62402c6","subject":"Update mariadb to 10.6.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d9a21b14331fa4533114bc94f3525fbce62402c6"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-11-26 11:10:37.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-11-26 11:10:37.000000000","tz":60},"subject":"Inherit default galera_wait_timeout value from openstack_db_connection_recycle_time","message":"Inherit default galera_wait_timeout value from openstack_db_connection_recycle_time\n\nSet a new default value for ``galera_wait_timeout`` which is inherited from global ``openstack_db_connection_recycle_time``.\n\nThese variables are directly related, it would cause errors when ``galera_wait_timeout`` is lower than ``openstack_db_connection_recycle_time``.\nFrom the other hand, I don\u0027t see any reason for ``galera_wait_timeout`` to be higher than ``openstack_db_connection_recycle_time`` in most cases.\n\nChange-Id: I9450912ec7960a8ab713517532164cab52628b30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a3d71802c853c44f0d79ada90a0f0b56c584bbbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a3d71802c853c44f0d79ada90a0f0b56c584bbbb"}]},"branch":"refs/heads/master"},"dac9a6f5a33341e00499436a5609793224d981a2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-26 11:25:44.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/25/819425/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/25/819425/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/25/819425/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/25/819425/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/25/819425/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/25/819425/2"}}},"commit":{"parents":[{"commit":"d9a21b14331fa4533114bc94f3525fbce62402c6","subject":"Update mariadb to 10.6.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d9a21b14331fa4533114bc94f3525fbce62402c6"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-11-26 11:10:37.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","date":"2021-11-26 11:25:44.000000000","tz":0},"subject":"Inherit default galera_wait_timeout value from openstack_db_connection_recycle_time","message":"Inherit default galera_wait_timeout value from openstack_db_connection_recycle_time\n\nSet a new default value for ``galera_wait_timeout`` which is inherited from global ``openstack_db_connection_recycle_time``.\n\nThese variables are directly related, it would cause errors when ``galera_wait_timeout`` is lower than ``openstack_db_connection_recycle_time``.\nFrom the other hand, I don\u0027t see any reason for ``galera_wait_timeout`` to be higher than ``openstack_db_connection_recycle_time`` in most cases.\n\nChange-Id: I9450912ec7960a8ab713517532164cab52628b30\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/819424\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dac9a6f5a33341e00499436a5609793224d981a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dac9a6f5a33341e00499436a5609793224d981a2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b79574382bf9a65803c03b4b68b8b8055893acc2":{"kind":"REWORK","_number":3,"created":"2021-12-02 11:15:17.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/25/819425/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/25/819425/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/25/819425/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/25/819425/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/25/819425/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/25/819425/3"}}},"commit":{"parents":[{"commit":"d9a21b14331fa4533114bc94f3525fbce62402c6","subject":"Update mariadb to 10.6.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d9a21b14331fa4533114bc94f3525fbce62402c6"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-11-26 11:10:37.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2021-12-02 11:15:06.000000000","tz":60},"subject":"Inherit default galera_wait_timeout value from openstack_db_connection_recycle_time","message":"Inherit default galera_wait_timeout value from openstack_db_connection_recycle_time\n\nSet a new default value for ``galera_wait_timeout`` which is inherited from global ``openstack_db_connection_recycle_time``.\n\nThese variables are directly related, it would cause errors when ``galera_wait_timeout`` is lower than ``openstack_db_connection_recycle_time``.\nFrom the other hand, I don\u0027t see any reason for ``galera_wait_timeout`` to be higher than ``openstack_db_connection_recycle_time`` in most cases.\n\nChange-Id: I9450912ec7960a8ab713517532164cab52628b30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b79574382bf9a65803c03b4b68b8b8055893acc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b79574382bf9a65803c03b4b68b8b8055893acc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
