)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~822911","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I0380782a034d291fd2173c2ae59fdfb8f4468bf0","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-12-28 13:04:08.000000000","reason":"Dmitriy Rabotyagov replied on the change"}},"hashtags":[],"change_id":"I0380782a034d291fd2173c2ae59fdfb8f4468bf0","subject":"Fix galera_force_bootstrap behaviour","status":"MERGED","created":"2021-12-24 14:50:05.000000000","updated":"2021-12-28 16:04:52.000000000","submitted":"2021-12-28 16:03:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822911","meta_rev_id":"b38fb8dbd9ff07344ad680e13a50e19b519fbe67","_number":822911,"virtual_id_number":822911,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-28 16:03:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":2,"date":"2021-12-26 12:57:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-28 10:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-28 10:52:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-24 16:04:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-25 05:22:26.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"CC"},{"updated":"2021-12-26 12:57:15.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2021-12-28 10:52:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"fd6b18916f9d876ef9c5890a52ad537e63046302","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-24 14:50:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c95fca8616aac75960126e64581d185f24d5569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 16:04:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47a8f369efc34c419da4bf3d00a874dd : SUCCESS in 5m 29s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb7f58afaa4740128f3a4f6d27b46c43 : SUCCESS in 17m 24s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/0cca7c1dd9f249f2bcd30bec6676c265 : SUCCESS in 1h 05m 21s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/20d89f96e6f1468d95aa466f67507d33 : SUCCESS in 1h 07m 02s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6feacddd80684e27bfa219d0eb3da949 : FAILURE in 44m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c6be31be16a46139e699932e3e0fade : FAILURE in 50m 44s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf5f99eb912b488893bda7c8b86f9af7 : SUCCESS in 1h 11m 24s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e10d7e8557834599972d24650c2a5545 : FAILURE in 55m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b481f208556830b9cda9ac38f281a503bbfe853e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-24 16:29:35.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2e81f877d4d471f569a4811aea7edc11d8985a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 21:51:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7667f1f62fc646d99fbc4f3ad476f6dc : SUCCESS in 6m 21s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/951b018319644894aeece183ff645936 : SUCCESS in 19m 01s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/393eb1e96aaa4a1aa2c763b7959a1efe : TIMED_OUT in 3h 06m 20s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b12dfc02e03f4bd7ab523defce59ca4c : SUCCESS in 1h 11m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c6dac966d6894cdd99b0c20c8dc11efb : FAILURE in 42m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c45a175fb2f443cc838d050880012275 : FAILURE in 47m 49s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2abe2da2c6f4ce2979ffa0a0b8d951f : SUCCESS in 1h 19m 20s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f204a73ddeee4b7b95f71b640114daf6 : FAILURE in 1h 10m 44s","accounts_in_message":[],"_revision_number":2},{"id":"66104533d95abed19ae2ee32451aa921015fb23a","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-12-25 05:22:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78753bdb6591f4315528df10436fd025a2e459c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-25 07:03:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19a239fac9b749b2ab5f9c5d928d4eda : SUCCESS in 6m 11s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c13e495c0e7b4c14999ceb9ce35d5ccd : SUCCESS in 12m 17s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/a9c579a453ed458b8ee00e75ebe195e9 : SUCCESS in 1h 03m 25s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/93afc9160adb457497851ea2da2dbe9c : SUCCESS in 1h 09m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/2d8fd959678a47959f3349554b6a28e6 : SUCCESS in 58m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0808826de774d6bad9671f2370c8551 : SUCCESS in 46m 10s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/738f12c016ea452d88ef8252cf3d9234 : SUCCESS in 1h 34m 17s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/078aae7cbaad4664a58386e41b9f7c21 : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":2},{"id":"3ce71a9a25de9a7aba73e7a4b4d42f68d2741d2c","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-12-26 12:57:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b7ffa421408071f98f1b901d2627b0fe0308defc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-28 10:52:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1a79625f570bc677a99595b0d1a78c69278e3726","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 10:55:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7dd20c904f46842ba1dc36e83beeb3906b0d2f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 12:28:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c29986a22bf450f9a0097f9f2f76221 : SUCCESS in 5m 00s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e80b47d85954290be3da3a6f461d70d : SUCCESS in 15m 11s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/2f4c3102abd340c7ad628cdf82cf982d : SUCCESS in 52m 46s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e015fde94764b57b60f7f7ac3ceaaac : SUCCESS in 1h 07m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1640134195f14ea582b2202b23b0e74c : FAILURE in 33m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61cc8e1f3ef44a9fbd0bf5e94a842079 : FAILURE in 47m 14s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a44ceb54b4c4c66919b32d51fede9b7 : SUCCESS in 1h 32m 51s","accounts_in_message":[],"_revision_number":2},{"id":"2489a12d568b6fa7ebd3184ad0c987e794dd9b65","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-28 13:04:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8d5127020d1c059e8a88796cd8b915d49a907398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 14:36:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb90a9cf1334455f99e0752a1e9f0244 : SUCCESS in 6m 09s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a373fd6b0c3441ca9700f3d9664ddc85 : SUCCESS in 14m 12s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/1717add81f724639af8cf52fa3fa00d2 : SUCCESS in 59m 32s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/01c7ad6159704d28b58fb8217476d9ae : SUCCESS in 1h 03m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/9679741f315948eebca4da7c96f94a45 : SUCCESS in 50m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d794d911e5f249b9a794216d6af0f7b1 : SUCCESS in 1h 04m 59s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac77c16f3f5946e8a7d9f0ffb8332e1b : SUCCESS in 1h 29m 05s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a698205935947f88a2ae316f827082a : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":2},{"id":"42f4108b52d20a63d71641a9b3d8cf88adac1fbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 14:39:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4194c41069b89f3a88e037387d2b22ef2026db61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 16:03:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e9e8c3d3e945e4970bec7e74f291f3 : SUCCESS in 6m 30s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99f266a6c67c445cb99875a7c9ffbe0f : SUCCESS in 15m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/e6c7cbb907c3442b8710fc6be50142dc : SUCCESS in 1h 07m 51s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1daa13508e0f47f7bc4858fd6b4b4002 : SUCCESS in 1h 04m 43s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b11c18aee37f4b4baf6de0bc7735ff22 : SUCCESS in 1h 00m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08c492b2e2c0430daddc0e7e581c2cb4 : SUCCESS in 51m 43s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9b02d73f6044ca48de6324e7f633198 : SUCCESS in 1h 24m 09s","accounts_in_message":[],"_revision_number":2},{"id":"d8ce384e66e5f613ced20dcb44d183ecea979292","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 16:03:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b38fb8dbd9ff07344ad680e13a50e19b519fbe67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 16:04:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/898b6b86e9db496d93b8d1268d556192 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"47d708e1612cfae7a4d0fe6dbef71f58a165d166","revisions":{"70a13b0f0657e5ef7ebb7068c7ed6cd1730c15dc":{"kind":"REWORK","_number":1,"created":"2021-12-24 14:50:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/822911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/11/822911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/822911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/822911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/822911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/822911/1"}}},"commit":{"parents":[{"commit":"c998a14d82d23bec7c9a0dfca884fe6af89c281b","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c998a14d82d23bec7c9a0dfca884fe6af89c281b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-24 14:49:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-24 14:49:56.000000000","tz":120},"subject":"Fix galera_force_bootstrap behaviour","message":"Fix galera_force_bootstrap behaviour\n\nCurrently, ``galera_force_bootstrap`` fuctionality is eqivalent to\n``galera_ignore_cluster_state`` and not really functional.\nWe add extra conditions to cluster bootstrap handlers that allow\nto force bootstrap cluster when it\u0027s already running.\n\nThis could also be leveraged while stretching galera cluster.\n\nChange-Id: I0380782a034d291fd2173c2ae59fdfb8f4468bf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/70a13b0f0657e5ef7ebb7068c7ed6cd1730c15dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/70a13b0f0657e5ef7ebb7068c7ed6cd1730c15dc"}]},"branch":"refs/heads/master"},"47d708e1612cfae7a4d0fe6dbef71f58a165d166":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-24 16:29:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/822911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/11/822911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/822911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/822911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/822911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/11/822911/2"}}},"commit":{"parents":[{"commit":"c998a14d82d23bec7c9a0dfca884fe6af89c281b","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c998a14d82d23bec7c9a0dfca884fe6af89c281b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-24 14:49:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-24 16:29:34.000000000","tz":0},"subject":"Fix galera_force_bootstrap behaviour","message":"Fix galera_force_bootstrap behaviour\n\nCurrently, ``galera_force_bootstrap`` fuctionality is eqivalent to\n``galera_ignore_cluster_state`` and not really functional.\nWe add extra conditions to cluster bootstrap handlers that allow\nto force bootstrap cluster when it\u0027s already running.\n\nThis could also be leveraged while stretching galera cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-python_venv_build/+/821415\nChange-Id: I0380782a034d291fd2173c2ae59fdfb8f4468bf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/47d708e1612cfae7a4d0fe6dbef71f58a165d166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/47d708e1612cfae7a4d0fe6dbef71f58a165d166"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
