)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~822943","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fwallaby~I0380782a034d291fd2173c2ae59fdfb8f4468bf0","project":"openstack/openstack-ansible-galera_server","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-01-04 13:23:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0380782a034d291fd2173c2ae59fdfb8f4468bf0","subject":"Fix galera_force_bootstrap behaviour","status":"MERGED","created":"2021-12-28 17:31:13.000000000","updated":"2022-01-04 13:25:47.000000000","submitted":"2022-01-04 13:23:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822943","cherry_pick_of_change":822942,"cherry_pick_of_patch_set":1,"meta_rev_id":"819bf0099131e3a4b926664f8a7ad8140b53b4a9","_number":822943,"virtual_id_number":822943,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-04 13:23:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-01-04 11:41:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-31 16:01:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-01-04 11:41:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-28 19:02:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-30 22:44:37.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2021-12-31 16:01:22.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-01-04 11:41:06.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"637162a20c7777d4c0820699492b90f9ff84363b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-28 17:31:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"999e1284ae6a42c189d01d7e61189a7bdadfb4b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 19:02:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c65bd79f11041a0bf6b5f7ae940c57c : SUCCESS in 5m 15s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc7e1054a11341e4a91a9cf06c775e3e : SUCCESS in 20m 11s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/711fd5b34593418d83a38f4ec60338fd : SUCCESS in 1h 04m 43s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d4bcf118a914444aebb270283062f2b : SUCCESS in 1h 10m 57s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/7a9f313701fc487e8b711e3aa578a06e : SUCCESS in 56m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a04da92110b04cf18d142805b5cb4e8f : SUCCESS in 58m 09s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70fc792b942046fda21748af74f6a05e : FAILURE in 37m 46s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/11261fbde5364ac38c326fbda13ed47a : SUCCESS in 1h 20m 44s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/283ec8002746476bab1b788ff1ef7681 : SUCCESS in 1h 26m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e7d2725f60c9997708ea4a672be7ef054f608f40","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-30 22:44:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d6601264adcfab638622a50a6334892ceca58da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 00:25:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9bbcc1d2cc54f36839adbc81dcd480f : SUCCESS in 5m 29s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6a02c88034441fdafe47b7cd72c4fbf : SUCCESS in 14m 05s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/60b07d3af6074c569086ef6713835b8b : SUCCESS in 1h 23m 22s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/11be75ed29f04b84b2d22c525a77c2c4 : SUCCESS in 1h 16m 45s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/6a88e583d7714699b74a0ebe9a6a9e68 : SUCCESS in 1h 07m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/598f89821a89432ebf2ec12802c5cd7c : SUCCESS in 1h 08m 07s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f480a29381d45768ad22f7a85e9fefc : SUCCESS in 1h 22m 58s\n- openstack-ansible-upgrade-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/19e67d3f782c45a39ffbf5edf0f37fba : SUCCESS in 1h 37m 45s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5f0c69773ce4fd1b559e85bfba13b19 : SUCCESS in 1h 14m 57s","accounts_in_message":[],"_revision_number":1},{"id":"63ef3896e5722ac995b11f0f84b1d1164abb0c62","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-31 16:01:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b7d9a0c47c7c52d67f9fc9b7b590c58b738047e9","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-04 11:41:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66f3d01898aa264970f9236b05143eca8b4a9e68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 11:45:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c1f68097748dbc094376d93f49902c74240a9ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 13:23:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6981fb80d2754bb59aa781fa9476fbee : SUCCESS in 5m 59s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c8451147dad4f7c824c7de59f8fced0 : SUCCESS in 21m 06s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/1ae6ab46a8e54dad9fd0784f07f4423a : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/deb96363198746b6949b6fc9875cb589 : SUCCESS in 1h 23m 05s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/8a17792d4f1a4e9e943c992e13360c0b : SUCCESS in 1h 13m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fb326a9b05849b69ab47fe012f424fe : SUCCESS in 1h 31m 20s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e10e0ba2f2c24ba9b488b91d66746237 : SUCCESS in 1h 23m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a088754070ce50bd1f52c8aaae63924c082a4ccf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 13:23:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"819bf0099131e3a4b926664f8a7ad8140b53b4a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 13:25:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9f7c5246ff74976a67acccd89a832f9 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c74c86e8a6afaae40135fc06e77d7012485b4257","revisions":{"c74c86e8a6afaae40135fc06e77d7012485b4257":{"kind":"REWORK","_number":1,"created":"2021-12-28 17:31:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/43/822943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/43/822943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/43/822943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/43/822943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/43/822943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/43/822943/1"}}},"commit":{"parents":[{"commit":"65d462703cfaa84c16f41eb12f86cd415dbbcc15","subject":"Fix mysql_upgrade possible race condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/65d462703cfaa84c16f41eb12f86cd415dbbcc15"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-24 14:49:56.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-28 17:31:13.000000000","tz":0},"subject":"Fix galera_force_bootstrap behaviour","message":"Fix galera_force_bootstrap behaviour\n\nCurrently, ``galera_force_bootstrap`` fuctionality is eqivalent to\n``galera_ignore_cluster_state`` and not really functional.\nWe add extra conditions to cluster bootstrap handlers that allow\nto force bootstrap cluster when it\u0027s already running.\n\nThis could also be leveraged while stretching galera cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-python_venv_build/+/821415\nChange-Id: I0380782a034d291fd2173c2ae59fdfb8f4468bf0\n(cherry picked from commit 47d708e1612cfae7a4d0fe6dbef71f58a165d166)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c74c86e8a6afaae40135fc06e77d7012485b4257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c74c86e8a6afaae40135fc06e77d7012485b4257"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
