)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~823983","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I41cdbb82b6ef82ea7acb4d22ca29aac84efebe30","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"osa/centos-9","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-05-25 23:35:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I41cdbb82b6ef82ea7acb4d22ca29aac84efebe30","subject":"Add support for centos-9","status":"MERGED","created":"2022-01-10 13:29:05.000000000","updated":"2022-05-25 23:36:34.000000000","submitted":"2022-05-25 23:35:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823983-osa/centos-9","meta_rev_id":"27eaa70dbce30ff8d7ad136f6e7e06ead802a9fd","_number":823983,"virtual_id_number":823983,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-25 23:35:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-05-23 08:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2022-05-20 06:58:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-05-23 08:39:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-10 14:47:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-20 06:58:09.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-05-23 08:39:00.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"8b26af3a52133c73d983743c9a8cc7a0f3cda906","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-10 13:29:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca7db2ad70f45fc9ce0dc0909989b607038d9ce8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 14:47:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/402fafe4a0db4990aa124968b3428012 : SUCCESS in 5m 16s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83883de268614bf185f81602168f6189 : FAILURE in 14m 21s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/0ebd88f78e8c4a9489a89fbe9531e7ae : FAILURE in 56m 02s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2cab50b1aa8409d9cb245df57a222b4 : FAILURE in 1h 14m 43s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8d59ec88adaa4f01b116126722b95465 : FAILURE in 54m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f53b95e212114dd2860a42c4b0852add : FAILURE in 1h 04m 38s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9feb1e6e027347349ac7f90ba1c9a776 : FAILURE in 1h 13m 29s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ed59c8dd100405eb6cd5612e2ffd354 : FAILURE in 1h 02m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b30578bd0185aff4af5f9eab4e194527d221cc5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-11 08:30:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5cc25e46de348e545aeaf829bf0145fda33c2c7d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-11 09:00:57.000000000","message":"Topic set to osa/centos-9","accounts_in_message":[],"_revision_number":2},{"id":"81a720bcb919109eb78b2fd087f1e964c49bf4a6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-11 09:01:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aac7bf605c1a97bf03be8ab6d517f2dc5898d7d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 10:00:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f58437e838c0428fafb265d0038bb60f : SUCCESS in 6m 14s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8002461908a04de58b273098132e3db6 : FAILURE in 17m 58s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/1fb042d401ea4c17959c4cfb30a5a93e : FAILURE in 1h 07m 43s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/86a3a6a1006b44c7be25cc247487929e : FAILURE in 1h 00m 31s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/78c0e1b9edf64e50970e6b85686d5040 : SUCCESS in 1h 23m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f94056c28274495f8f667756876bf7ed : SUCCESS in 1h 14m 32s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4ec2e5429564eb08f8ee019615ca86b : FAILURE in 1h 20m 01s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fac266e322c45ffbb0a466c67bd7b85 : FAILURE in 1h 01m 13s","accounts_in_message":[],"_revision_number":2},{"id":"81e1ed254e50e1edcab3ae6d877d07de433edb7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-26 12:05:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"20f81d261db2f8a82ca1198b9d5fa53a2357f714","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-26 12:41:54.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"39f8257fbddbb00cd1b2f1038f7f75ed3112aefa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 14:14:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c496e512536496a92f163628fa157e1 : SUCCESS in 4m 55s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0aec6d26e99491ab59d987ac3717fb2 : SUCCESS in 11m 55s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f132b81b171469f9afeab12b9d22ab0 : SUCCESS in 1h 23m 34s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5d5503d470204e07aaf0036c7e3ca78f : SUCCESS in 1h 10m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/374bec5c43fa411ebc91e4dc3507aea6 : SUCCESS in 1h 19m 10s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/69f1afa488184971813b487f1b9f9e39 : FAILURE in 1h 24m 13s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7934917003324fa5a389a2ba1ecd3f79 : FAILURE in 1h 04m 40s","accounts_in_message":[],"_revision_number":4},{"id":"c7f063426d91f8b91a981824cb22f85b3d7312a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-31 11:33:13.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"2a0c052445032b2d16d4689a01b89fb7dae8508d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 13:02:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cba6ce4f6024409a8726bd176e90d0b : SUCCESS in 5m 24s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1dc5f13ebd1648e589a276f2f1a5bd72 : SUCCESS in 12m 00s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/20e1695b17dc4ed4aab05e8737ee75cf : FAILURE in 49m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/63a9a991c6d74154b2597b775cecba00 : FAILURE in 39m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c56d5401b89e4cf7849eb03924d8a52e : FAILURE in 40m 04s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e73d77487e8f478cac035140a43e5921 : FAILURE in 1h 19m 27s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d13b5f933cc94e2f904d78751803bae7 : FAILURE in 1h 04m 14s","accounts_in_message":[],"_revision_number":5},{"id":"7b860e309ff66ba4161affa78aaefee877590a1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-12 13:20:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"214d02afc4d11b8d3cacf484887a9013ccfbaeca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 15:00:19.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46ebb6b7f5134cf08a9a99b6e756ef57 : SUCCESS in 4m 38s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dedf158be2e541829fc717227d67a77b : SUCCESS in 15m 53s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ebab88d0ffb4275a5f9e3ce4995579d : FAILURE in 1h 03m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/07aea2cd8fec47ebab2cb944328a5f0e : FAILURE in 49m 31s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/6f74aac865134c13906cb0535a0dc985 : FAILURE in 1h 04m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcf3b260aa1e41b9924674789e9e8bd0 : FAILURE in 53m 23s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/196e0cd5999340139c9b90fa2971a6a8 : FAILURE in 1h 34m 40s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6e0dbae0836497490fc8b2984ddaf0c : FAILURE in 1h 25m 13s","accounts_in_message":[],"_revision_number":6},{"id":"5a4a950ef0e63a96425f84b4565a4a8d03bce183","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-16 11:02:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a07f3bbcb3322e550ad772e5f7a71ec0a4931dc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 13:00:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8680f21272304a2a90427e2e8283c63d : SUCCESS in 5m 50s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d517c5e259234229b55162fc31a8a7d9 : SUCCESS in 14m 35s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/efc5b10c83454c73b40ac0243fae6cd0 : SUCCESS in 1h 24m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1c5f023f91384eb6aedf717c855049b1 : SUCCESS in 1h 00m 24s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b4b609732ef24600a36fc3c741d1c21c : SUCCESS in 1h 17m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d7dde59d7824afd8ce99208879b55ed : SUCCESS in 1h 16m 40s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9609bf159a5b4284b8631cfc6cba6bd6 : SUCCESS in 1h 50m 05s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2984cc72ca4c4e1699b597b5c633e4bf : SUCCESS in 1h 30m 32s","accounts_in_message":[],"_revision_number":6},{"id":"2eabcf3363839484626b5b0a3d37161a4cae41db","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-20 06:58:09.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"85731e607ba284de8442851ad3672df0ca684af2","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-05-23 08:39:00.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"49467e2bdfe3e766a4183e610b08990e152191cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 08:41:25.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"de3b0dd08cc8eddd73af5c5a42d70b4dbabf87ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 10:10:33.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4e09a30a92c4b40ab114ad5b88be416 : SUCCESS in 6m 04s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22a1ad3b66d248fea832681dd03950b2 : SUCCESS in 16m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a9862c9f4464230a6423f8ef88828f3 : SUCCESS in 1h 20m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8fabf66fcd9b4bc0b19318d79be85735 : SUCCESS in 1h 09m 56s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d6828e723eb4453099329aea110ad165 : SUCCESS in 1h 23m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73bfc13f7ac347bd8e612b8de9370fb6 : SUCCESS in 1h 10m 37s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5e381b612054bfab322e874cba89b35 : FAILURE in 1h 00m 44s","accounts_in_message":[],"_revision_number":6},{"id":"4369b0c074a25e5c16dbbe526b55a63a97bcd6b4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-25 19:56:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d1042860907d0922bb9a4ec399511843c2d9212a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 21:49:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a938579c14c6478f82c19486fed03bfb : SUCCESS in 6m 32s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d5ef5bf53f54d1c8c1eb75903b68407 : SUCCESS in 15m 00s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/93959c5d6cc84bd297114aaa69fb68e2 : SUCCESS in 1h 29m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1950ec82d72e492c99d184ad582b6f4d : SUCCESS in 1h 15m 27s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/fb6ebe50e67a4cf78c57ed7b07593602 : SUCCESS in 1h 43m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12bd46196d0f4802b3c7e5adf8c80329 : SUCCESS in 1h 14m 47s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f21973a4f80d4b999d7685f22e27c5f6 : FAILURE in 1h 47m 07s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48216ec08f8f4f9b8cfe170d5a216780 : SUCCESS in 1h 40m 07s","accounts_in_message":[],"_revision_number":6},{"id":"6394e97c2e05d4c3628d311684667455c4091939","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 21:51:31.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"103b08806e0e2dd6e4c7a206229c252f9b0388df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 23:35:23.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c5540831ee4f44a918221f0681ef00 : SUCCESS in 5m 54s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a952529b89f64ab4a7a200b7e94faab2 : SUCCESS in 21m 17s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/36f04269189d44f9848fff6dff1c3d86 : SUCCESS in 1h 31m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/723c4284705d448cb7e9db3b9fea141f : SUCCESS in 1h 05m 57s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/1bd7677a08b346d68b079497876ce887 : SUCCESS in 1h 39m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b5b1417fb7d4a9b8749d9077907e68c : SUCCESS in 1h 19m 58s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad5afb5185f746bcb61eeefb0bea434f : SUCCESS in 1h 33m 56s","accounts_in_message":[],"_revision_number":6},{"id":"fc4136b2302641b73af497463660e28629b6815c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 23:35:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"27eaa70dbce30ff8d7ad136f6e7e06ead802a9fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 23:36:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/191eee016ee8493bab3fe8ce6b4de2e6 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d083461eb064e8aaf4c0378cea3f55aebc4e6144","revisions":{"fc0c1565239475735532f14055d6398352be4ec8":{"kind":"REWORK","_number":1,"created":"2022-01-10 13:29:05.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/83/823983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/83/823983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/1"}}},"commit":{"parents":[{"commit":"47d708e1612cfae7a4d0fe6dbef71f58a165d166","subject":"Fix galera_force_bootstrap behaviour","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/47d708e1612cfae7a4d0fe6dbef71f58a165d166"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-10 13:26:09.000000000","tz":-300},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-10 13:28:58.000000000","tz":-300},"subject":"DNM - Add support for centos-9","message":"DNM - Add support for centos-9\n\nThis uses the distro package for mariadb so should not be merged until\nthere is an official package from mariadb.org and this patch is updated\n\nChange-Id: I41cdbb82b6ef82ea7acb4d22ca29aac84efebe30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fc0c1565239475735532f14055d6398352be4ec8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fc0c1565239475735532f14055d6398352be4ec8"}]},"branch":"refs/heads/master"},"80fbc5a8eb65299b8ccbb85c29c334f8bd155168":{"kind":"REWORK","_number":2,"created":"2022-01-11 08:30:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/83/823983/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/83/823983/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/2"}}},"commit":{"parents":[{"commit":"6d8417583d7a031093e47e34aaae1e9dd0f658ef","subject":"Convert xinetd clustercheck to systemd socket service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6d8417583d7a031093e47e34aaae1e9dd0f658ef"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-10 13:26:09.000000000","tz":-300},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-11 08:29:39.000000000","tz":-300},"subject":"DNM - Add support for centos-9","message":"DNM - Add support for centos-9\n\nThis uses the distro package for mariadb so should not be merged until\nthere is an official package from mariadb.org and this patch is updated\n\nChange-Id: I41cdbb82b6ef82ea7acb4d22ca29aac84efebe30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/80fbc5a8eb65299b8ccbb85c29c334f8bd155168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/80fbc5a8eb65299b8ccbb85c29c334f8bd155168"}]},"branch":"refs/heads/master"},"9b9b1a3d72f45531aa9930b115a99f39a99d147f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-26 12:05:57.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/83/823983/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/83/823983/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/3"}}},"commit":{"parents":[{"commit":"c2b586e8749f99ea37df98019965730a55255571","subject":"Convert xinetd clustercheck to systemd socket service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c2b586e8749f99ea37df98019965730a55255571"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-10 13:26:09.000000000","tz":-300},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-26 12:05:57.000000000","tz":0},"subject":"DNM - Add support for centos-9","message":"DNM - Add support for centos-9\n\nThis uses the distro package for mariadb so should not be merged until\nthere is an official package from mariadb.org and this patch is updated\n\nChange-Id: I41cdbb82b6ef82ea7acb4d22ca29aac84efebe30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9b9b1a3d72f45531aa9930b115a99f39a99d147f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9b9b1a3d72f45531aa9930b115a99f39a99d147f"}]},"branch":"refs/heads/master","description":"Rebase"},"792c31c48ca8e356e85baac46600a7c137bd1cc3":{"kind":"REWORK","_number":4,"created":"2022-01-26 12:41:54.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/83/823983/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/83/823983/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/4"}}},"commit":{"parents":[{"commit":"c2b586e8749f99ea37df98019965730a55255571","subject":"Convert xinetd clustercheck to systemd socket service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c2b586e8749f99ea37df98019965730a55255571"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-10 13:26:09.000000000","tz":-300},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-26 12:41:54.000000000","tz":0},"subject":"DNM - Add support for centos-9","message":"DNM - Add support for centos-9\n\nThis uses the distro package for mariadb so should not be merged until\nthere is an official package from mariadb.org and this patch is updated\n\nChange-Id: I41cdbb82b6ef82ea7acb4d22ca29aac84efebe30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/792c31c48ca8e356e85baac46600a7c137bd1cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/792c31c48ca8e356e85baac46600a7c137bd1cc3"}]},"branch":"refs/heads/master"},"a2cac59b8cce05186529d71396bb993e6485c529":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-01-31 11:33:13.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/83/823983/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/83/823983/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/5"}}},"commit":{"parents":[{"commit":"f12f66242c3c45bd1ee4213ccc11d7a0b73f86e8","subject":"Convert xinetd clustercheck to systemd socket service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f12f66242c3c45bd1ee4213ccc11d7a0b73f86e8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-10 13:26:09.000000000","tz":-300},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-31 11:33:13.000000000","tz":0},"subject":"DNM - Add support for centos-9","message":"DNM - Add support for centos-9\n\nThis uses the distro package for mariadb so should not be merged until\nthere is an official package from mariadb.org and this patch is updated\n\nChange-Id: I41cdbb82b6ef82ea7acb4d22ca29aac84efebe30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a2cac59b8cce05186529d71396bb993e6485c529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a2cac59b8cce05186529d71396bb993e6485c529"}]},"branch":"refs/heads/master","description":"Rebase"},"d083461eb064e8aaf4c0378cea3f55aebc4e6144":{"kind":"REWORK","_number":6,"created":"2022-05-12 13:20:48.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/83/823983/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/83/823983/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/823983/6"}}},"commit":{"parents":[{"commit":"090d3cc63ec9de019175a62c8482f4a9f49e1cdd","subject":"Add support for Ubuntu 22.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/090d3cc63ec9de019175a62c8482f4a9f49e1cdd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-10 13:26:09.000000000","tz":-300},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-12 13:20:19.000000000","tz":60},"subject":"Add support for centos-9","message":"Add support for centos-9\n\nThis uses the distro package for mariadb as there is no release from\nmariadb.org yet.\n\nChange-Id: I41cdbb82b6ef82ea7acb4d22ca29aac84efebe30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d083461eb064e8aaf4c0378cea3f55aebc4e6144"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d083461eb064e8aaf4c0378cea3f55aebc4e6144"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
