)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~824350","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I703079f9ba98ca4c0c825bd36746280d91dd4a5b","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"osa/include_vars","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-01-12 15:22:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I703079f9ba98ca4c0c825bd36746280d91dd4a5b","subject":"Refactor use of include_vars","status":"MERGED","created":"2022-01-12 08:50:15.000000000","updated":"2022-01-12 15:26:36.000000000","submitted":"2022-01-12 15:22:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824350-osa/include_vars","meta_rev_id":"3e98ba0a609c4f1ad5c794bd19e9773d1fdb56aa","_number":824350,"virtual_id_number":824350,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-12 15:22:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-01-12 13:02:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2022-01-12 11:10:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-01-12 13:02:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 10:51:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-12 11:10:36.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-01-12 13:02:46.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"aee45e0f529efb46ceee845588de0a2a99790c66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-12 08:50:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac94dced9bf7116d51f8e34e93658480f25537a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 10:51:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68a39b7a342b454ead7613c601f098ec : SUCCESS in 6m 35s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2430bfab8d1468f9bbc12cedd2cfd73 : SUCCESS in 22m 41s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b826ec4190024d41a5c787fd7c5ff196 : SUCCESS in 1h 24m 24s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4573e728d6a34dbe828f5ba27c34457d : SUCCESS in 1h 24m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7ed1155da3e1494299409481d72b66d2 : SUCCESS in 1h 12m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a22f261d8cf4b72bd822f04c55b915a : SUCCESS in 1h 16m 47s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5e61d9b2ec1452d843b2c5fe13341b3 : SUCCESS in 1h 39m 57s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d87facd51b904f5d9a09936cb90fc220 : SUCCESS in 1h 34m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f626607aa241f53cdbdebc02f0c2b4fcc1293ac4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-12 11:10:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7e382590c82f41805f98ae2f534b49848921985f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-12 11:13:27.000000000","message":"Topic set to osa/include_vars","accounts_in_message":[],"_revision_number":1},{"id":"008d3be825f3e5d4d4d6a5d8002d00292d8b309a","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-12 13:02:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cdbf63de8fcc0c99e08fd0a6a47e2bfa2d98649b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 13:16:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8676fe051414c31b1df57b71324b8387fc0a752","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 15:22:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8338ef0395f94141822ec08d2f1846d3 : SUCCESS in 5m 33s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee063bdc0192440ab8283ac4a04a5553 : SUCCESS in 22m 06s\n- openstack-ansible-deploy-infra_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/90743ddbd7114802a238a81d260f3f77 : SUCCESS in 1h 23m 55s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b457e4ca61fc4b1488bee8258e966b3e : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b10c7b19f5084b6dad9416ce6954a0e0 : SUCCESS in 1h 19m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/060c660da4cb4c728df8008bf832cd4c : SUCCESS in 1h 17m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44ca4a2ec8134cd682a25ba98a3b1eb7 : SUCCESS in 1h 30m 06s","accounts_in_message":[],"_revision_number":1},{"id":"44749f9e9f444efcc8cab55de06c269fd0da3169","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 15:22:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3e98ba0a609c4f1ad5c794bd19e9773d1fdb56aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 15:26:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e3d5261f71a49e0bf759c13b07f3d16 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f924870f67e459ab77ed8178f130c2404721f1f","revisions":{"5f924870f67e459ab77ed8178f130c2404721f1f":{"kind":"REWORK","_number":1,"created":"2022-01-12 08:50:15.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/50/824350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/50/824350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/824350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/824350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/824350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/824350/1"}}},"commit":{"parents":[{"commit":"b816b4774f0ee899ebbdb386c560e78d1dbb0770","subject":"Merge \"Allow galera_address to be FQDN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b816b4774f0ee899ebbdb386c560e78d1dbb0770"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 08:49:57.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 08:49:57.000000000","tz":0},"subject":"Refactor use of include_vars","message":"Refactor use of include_vars\n\nUse a first_found lookup instead of a with_first_found loop so that\nthe \u0027paths\u0027 parameter can be used.\n\nThis ensures that only vars from the role are included, and not vars\nfrom a parent calling role. This can happen when a parent role has\na higher priority vars file available for inclusion than the role\nit calls.\n\nChange-Id: I703079f9ba98ca4c0c825bd36746280d91dd4a5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5f924870f67e459ab77ed8178f130c2404721f1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5f924870f67e459ab77ed8178f130c2404721f1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
