)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~826503","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ibe1bc6f5cee30ab0682078dfe3ce5464336cf822","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"826503","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-03-21 10:11:03.000000000","reason":"Dmitriy Rabotyagov replied on the change"}},"hashtags":[],"change_id":"Ibe1bc6f5cee30ab0682078dfe3ce5464336cf822","subject":"Clean-up systemd overrides removal task","status":"MERGED","created":"2022-01-26 17:51:30.000000000","updated":"2022-03-21 13:53:25.000000000","submitted":"2022-03-21 13:51:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826503-826503","meta_rev_id":"ebf0fe189f55c82542167657ae18111c208b5a1b","_number":826503,"virtual_id_number":826503,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-21 13:51:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-03-16 13:34:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-13 10:21:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-03-16 13:34:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-26 19:41:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-13 10:21:29.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-03-16 13:34:14.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"5dace83cff9494de00031848c8b1e7f5827c4b26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-26 17:51:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66654de40b7b50636fa9a6e8bdba0d8b8ea9cb5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 19:41:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4183fb5f53574e5c801cc0cf973094e8 : SUCCESS in 5m 46s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79d94c5c61ec4bd09c2dec07917269d0 : SUCCESS in 13m 44s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/51f4750cd19c40aca53e9eb552b7af55 : SUCCESS in 1h 08m 15s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/caf3efdf4f3b4144898fb7464b68d715 : SUCCESS in 1h 03m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e41282fd11a34c99a482310655258e61 : SUCCESS in 1h 15m 01s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e75e80ec96d479188607fd50c2b2312 : FAILURE in 1h 41m 14s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/564e4ec4cdcd4297bab73ef7a8469a38 : FAILURE in 1h 25m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2dbd2885e5f4e1cfd0768dae1e5b83335160423b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-28 14:39:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71af6b9456292b9f04db63a87a4e1406105f8b31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 16:39:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a67ad8f1e54fd191dcc54c5ae9f2f4 : SUCCESS in 7m 16s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d605d4580717444ba75895bd8b50737e : SUCCESS in 21m 01s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0f31044a8b64cfd9d4a95cc3d44863b : FAILURE in 1h 36m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/08599b29a1594eb5aa78388be6fd31c5 : SUCCESS in 1h 16m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f68a0adc76e4d6aa8fc7fd225a8ef20 : SUCCESS in 1h 13m 31s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d998482befc45049c79d2f14d79d275 : SUCCESS in 1h 52m 21s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7440b3f06eb469a9c963dfbf6334798 : SUCCESS in 1h 33m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ae257e8fc1c9544b06e38e0bc661edbcadcd2dd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-10 15:50:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8598df76f6d181f636b15f99a580f75490f76815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 17:22:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70faf8cc21b0423190abcc6e4ac51e91 : SUCCESS in 6m 55s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18f5d3eaf1e84f588c01415084c6e306 : SUCCESS in 16m 38s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/62539a66b28a4f5e97ebe7838a58157c : FAILURE in 1h 15m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3e922901eadf41529dd52ce92f4a51a1 : FAILURE in 52m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33af427cc0954b63b083914a3a219de1 : FAILURE in 1h 06m 36s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdcd4c568c0b4e628182dd30484d6f09 : FAILURE in 1h 14m 09s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ee5fbbb2de54fcaa1a4cbb3cb049a1a : FAILURE in 1h 24m 35s","accounts_in_message":[],"_revision_number":2},{"id":"ab7d049a3ee67f8e851e5cddfb9fd2cf300cf001","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-10 20:31:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"061aa9c219c3bece4c28cd5f54e102e7684c80e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 22:11:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b97a2ccf01b04d44b7034477168b5025 : SUCCESS in 7m 10s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d908b7aa25b4aaebf30b6e4305660ac : SUCCESS in 22m 57s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/fcc8a2a5b4654131b6fc43f838a1d1f2 : FAILURE in 1h 10m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/9da8964078784b329fe51d8770b941a6 : FAILURE in 1h 05m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91d8ec7dff2f4083b160ece1aa0bf99e : FAILURE in 1h 04m 12s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a5bf6f241b54f44adb273408a5ac7cc : FAILURE in 1h 33m 41s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cda859e31b1d408c88aa09b25890759a : FAILURE in 1h 18m 39s","accounts_in_message":[],"_revision_number":2},{"id":"1a369d39d55cb1e4af2cdeebaa2e4076d308a74e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-11 09:20:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d76901917ebd50ad518059e1950a8838da96dd19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 11:09:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eabb79f4fd0448e1852d8112af86695e : SUCCESS in 6m 55s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb966fa91b70466ea43ab64099306058 : SUCCESS in 16m 27s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ec15fea211d42d891ad0335b7a4d102 : SUCCESS in 1h 27m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1520661557944650b4a4822b6f035760 : SUCCESS in 1h 09m 58s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/1437fb9eebe545ba9031564e4a8f8d9a : SUCCESS in 1h 24m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0067dd86afc413fa40047b8c7ea8325 : SUCCESS in 1h 25m 54s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/45ba8734c83c489882d1dd25b9604cfe : SUCCESS in 1h 40m 52s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27ac0f295ff84bd491523dcbd1569893 : SUCCESS in 1h 36m 32s","accounts_in_message":[],"_revision_number":2},{"id":"bdd2cb302df0e2efbe3ea99f9f7346bb1a323e4e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-03-13 10:21:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"62b82706a8cd147a5bafae5a5302cbf11e0ae6ff","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-03-16 13:34:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f96f095c12201aaa25749c2f3c2005f5112a7833","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 13:36:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a26f028ecf25126c48f3d6dcbeb6aac047fe45c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 15:26:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4070eae809a3481e80889b64e304e90a : SUCCESS in 7m 39s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f904777c0124b208f5b63380eb64121 : SUCCESS in 18m 53s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d6454d6a6ec4628b3829ae605a53e08 : SUCCESS in 1h 31m 43s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/545bc379c658492fb145b4cd56b13cf7 : SUCCESS in 1h 17m 50s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/69cda11c379c4d0eabdd81405fa43ab1 : SUCCESS in 1h 30m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dec9845f40649f9b90d029392f992e2 : SUCCESS in 1h 29m 35s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d265b0c176345f5bdbbe02c0de94f5f : FAILURE in 1h 38m 59s","accounts_in_message":[],"_revision_number":2},{"id":"e04a73fc267cf5d45f778a1b5a0e2d8e3462c311","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-17 14:35:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b76489b18e058a6ecdb37b0800f1651d3dfaffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 16:52:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a99177c3ced439e8883a8deaa25a8fe : SUCCESS in 7m 36s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/172557dfaae349bc9e4411d4b2fbe179 : SUCCESS in 18m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3fc28e8d3f14877b892620e2e8936e6 : SUCCESS in 1h 30m 11s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a57c8e1a59ae49eda24e47d0875a0758 : FAILURE in 1h 09m 40s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/36d7a6d6ec464400ae902a7005d16a36 : FAILURE in 1h 15m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9c86de52ee940ea8b6b95e5370ccee1 : FAILURE in 1h 13m 29s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/24b8546ce4324f9bbc4859c2dea2f6c0 : SUCCESS in 2h 09m 31s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f15d89956bfe4c04b19c9a48def63863 : FAILURE in 54m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1cf0439ac73562a2fa59b94c499a351060a8f0c4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-17 17:28:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9df324bcf1a8b2bb7f09755bd041d3fb1940dee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 19:14:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c1cc55f77140dd963bc9eec8c50cd9 : SUCCESS in 7m 17s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80b24e3507a645d9badb6d391a3a17e1 : SUCCESS in 15m 15s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f08512b04eec428fa1e15275c6f77680 : SUCCESS in 1h 18m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1d13373ccdbd4ff0820312509f74bedf : SUCCESS in 1h 08m 01s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/8882dc6fdf23485081a2235f5698ae2f : FAILURE in 1h 13m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12c52d201ff94b70b88eb0dab31c597b : FAILURE in 1h 04m 43s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4dcc20906e8497fbfdbfacaaf5fd02e : SUCCESS in 1h 38m 10s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a26f180c0ca7453d824476ed1193d725 : FAILURE in 52m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1bd6946b098856346466e8d95f6baf0628562c32","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-18 08:14:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d00c9e183abc222e1252ca71ff0333e31ac61f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 10:17:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a67058bb665f42759e1ea3af26fa8cbd : SUCCESS in 6m 36s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e2b5f4849344d9c8d27290840f6a9b1 : SUCCESS in 12m 40s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e2f26d100814d5cad211d3cfa2d3dbd : SUCCESS in 1h 24m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/065cb2558e3d48c6854d6f8d53010aa8 : SUCCESS in 1h 10m 05s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4aede0be28d24c03b44a1740a4be391c : SUCCESS in 1h 21m 23s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f40115aaf6b457c81c657064f9e03af : SUCCESS in 1h 12m 50s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/5df9a921aa13495fa49ddb666dee9469 : SUCCESS in 1h 52m 41s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03ed1b27eb0f42659ae242f6647771d6 : SUCCESS in 1h 36m 50s","accounts_in_message":[],"_revision_number":2},{"id":"089152f2b9ee647aadfd4343021ada1316d6f25b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 10:22:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3a00d9b359cf108006759919b3fdf74de95534f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 11:55:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d4d70024364c189291c56a405b788d : SUCCESS in 6m 52s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d199cfe2ca749989f509e49010d5c7d : SUCCESS in 15m 27s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d40550ade4484841b59ccd04b6da508f : SUCCESS in 1h 16m 59s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ddd4495e161546a38a8c2c4d762370d2 : FAILURE in 51m 08s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/5ce9802a70b04ec5ab9a6504c419ea3c : SUCCESS in 1h 16m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a83eb3281a64a41be7337b717ecc135 : SUCCESS in 1h 11m 02s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f149c5ea47b14f098ee3f664b98c0012 : SUCCESS in 1h 25m 39s","accounts_in_message":[],"_revision_number":2},{"id":"4bfa115c8062cea8a6b0fd884003cea6d68e0988","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-19 09:04:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1baa1575891683ce741df88cf5313a323873d262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 10:40:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d457155674f74f628021c6d9401ee233 : SUCCESS in 8m 41s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f09d8c0996c4d2f99f9a8ead88e9e18 : SUCCESS in 19m 19s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/697dd5283e894287b0cb44c006457a6c : SUCCESS in 1h 15m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5c736c73149641f1bf29840131d6a8e9 : SUCCESS in 1h 07m 51s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/398663c54a8247b4bb1d5da12d8b0916 : SUCCESS in 1h 18m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c10585033ff941958b5952414e5fa436 : SUCCESS in 52m 04s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/872f94d6da374409b8b3923fa1a0b40a : SUCCESS in 1h 31m 04s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd1c63028aa94f1895489e2ddf33c8cd : FAILURE in 1h 09m 25s","accounts_in_message":[],"_revision_number":2},{"id":"70e7829b529aa838afabfb6539428d1017830cf9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-21 10:11:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"562d313335c3716fa9d219884a0e917acdb82c8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 12:17:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4db8b758f634727bb3e562b6962d9e6 : SUCCESS in 7m 52s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/175d1e156f974213aa293cb78dd8bb97 : SUCCESS in 17m 39s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/3dc1f253d3be44f09708fc5b6e7c4cf8 : SUCCESS in 1h 26m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/0c3a7fdbb87e4543baac4f7e73890d31 : SUCCESS in 1h 04m 44s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4ced9db6041646e693b37869cba90fcc : SUCCESS in 1h 23m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cacc6f468f94377a1838d7c95ce55d5 : SUCCESS in 1h 13m 52s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/284a85d9732841fda1dca9c1d3f318b4 : SUCCESS in 1h 54m 47s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42eafd4e3e9f49c5969817f7e9ca3cf1 : SUCCESS in 1h 22m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f905944deb5d0eed9ffe756762540af5bd38a5ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 12:19:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0ef7cd34d68b914bf20e2215db368e0d6e8a6474","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 13:51:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3736188e20354a5a9bfdd5ccd2373295 : SUCCESS in 6m 53s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d0b43c1824347b5b839a75810f8998c : SUCCESS in 13m 44s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e54fbba94774eb995fd4c2718738fe3 : SUCCESS in 1h 26m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/2158e3fc381f4ae7b632671208e1ddb9 : SUCCESS in 1h 09m 39s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/a87f1ec98d0242e7aecf6066ae542821 : SUCCESS in 1h 21m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05366923fee942eead5ee5ec8de07e8d : SUCCESS in 1h 15m 48s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70b5c1fc4fe0423a85e4927072b9d900 : SUCCESS in 1h 27m 57s","accounts_in_message":[],"_revision_number":2},{"id":"ceb335087fb5909643618ceec877a37b73340602","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 13:51:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ebf0fe189f55c82542167657ae18111c208b5a1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 13:53:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b16ce80421fb4e018f2325a0826777e1 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"45263ac6213420cc54702ea84d76a742cd1da51d","revisions":{"1b4146a8d0d10e3258e349db774e58fa9481e36d":{"kind":"REWORK","_number":1,"created":"2022-01-26 17:51:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/826503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/826503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/826503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/826503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/826503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/826503/1"}}},"commit":{"parents":[{"commit":"18d8de22dc6925777463656d292b5686b9ce384a","subject":"Merge \"Refactor use of include_vars\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/18d8de22dc6925777463656d292b5686b9ce384a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-26 17:51:22.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-26 17:51:22.000000000","tz":120},"subject":"Clean-up systemd overrides removal task","message":"Clean-up systemd overrides removal task\n\nOnce upgrade is done and release is branched, we can cleanup\ntask that was added for upgrade purposes only\n\nChange-Id: Ibe1bc6f5cee30ab0682078dfe3ce5464336cf822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1b4146a8d0d10e3258e349db774e58fa9481e36d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1b4146a8d0d10e3258e349db774e58fa9481e36d"}]},"branch":"refs/heads/master"},"45263ac6213420cc54702ea84d76a742cd1da51d":{"kind":"REWORK","_number":2,"created":"2022-03-10 15:50:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/826503/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/826503/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/826503/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/826503/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/826503/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/826503/2"}}},"commit":{"parents":[{"commit":"3c57409bdfd0c48e08daf34c60ca401f0861eb56","subject":"Use unix socket while granting access for the backup service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3c57409bdfd0c48e08daf34c60ca401f0861eb56"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-26 17:51:22.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-10 15:50:35.000000000","tz":60},"subject":"Clean-up systemd overrides removal task","message":"Clean-up systemd overrides removal task\n\nOnce upgrade is done and release is branched, we can cleanup\ntask that was added for upgrade purposes only\n\nChange-Id: Ibe1bc6f5cee30ab0682078dfe3ce5464336cf822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/45263ac6213420cc54702ea84d76a742cd1da51d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/45263ac6213420cc54702ea84d76a742cd1da51d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
