)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~829260","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fwallaby~I71bc866aedaa6fba3cc86d2a2a99ed32f0727c54","project":"openstack/openstack-ansible-galera_server","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2022-02-21 10:57:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I71bc866aedaa6fba3cc86d2a2a99ed32f0727c54","subject":"Use unix socket while granting access for the backup service","status":"MERGED","created":"2022-02-15 18:48:46.000000000","updated":"2022-02-21 10:59:09.000000000","submitted":"2022-02-21 10:57:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829260","cherry_pick_of_change":829259,"cherry_pick_of_patch_set":1,"meta_rev_id":"b9fe226e4f7769eea7a91c8005bf962e89cb2e9b","_number":829260,"virtual_id_number":829260,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-21 10:57:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-02-16 08:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-15 20:35:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-02-16 08:35:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 20:24:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-15 20:35:55.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-02-16 08:35:25.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2022-02-18 14:55:01.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"}],"messages":[{"id":"ab8fb9cd5184dc69ba1207afd0c7330bd70d368c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-02-15 18:48:46.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"698fc0e52fd864fff1176f9f986cea76c76d0d22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 20:24:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e2f4f45939c456d9daf487919e9909e : SUCCESS in 6m 33s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bdab57afab14d88a77881de55367c60 : SUCCESS in 17m 19s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/90e6d2202b2b4b8a86ff15c2c9533169 : SUCCESS in 1h 27m 13s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/1a2e0a110e1247838cc8be99b42c6190 : SUCCESS in 1h 23m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/742e5bc00214493ebbde99a959c5d68a : SUCCESS in 1h 11m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f96bea713f244aa58cb24a98ec7c7b69 : SUCCESS in 1h 19m 11s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/580b72f57b484cc0b21d71ca5bb66445 : FAILURE in 28m 10s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cfa6b7223e247a7a45c79518c8adb3c : SUCCESS in 1h 31m 02s","accounts_in_message":[],"_revision_number":1},{"id":"12778a6a784b1b75ab134e55bc55581ac1eee341","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-15 20:35:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0ba6949d17281b8ab051fcf2656f2764f19872a","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-02-16 08:35:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"503908d52a6188a91477679898d3559b535d6490","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 08:38:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7d03c1cc15268e1a7a32c0d1553855ad9800164","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:10:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/321f37385eac42bdbf74d05d6f4a4660 : SUCCESS in 5m 38s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52d3fdae30c040d4ab8ab604a933f611 : SUCCESS in 18m 22s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d7b9b41186ed4ce3ba81e9ec2f4e2c70 : SUCCESS in 1h 16m 15s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/f50caf5901e840f6b4132e1ad2471d16 : SUCCESS in 1h 14m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5637ec1a208f41a1a81e6a8c20d3b904 : SUCCESS in 1h 12m 36s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ec33e81805d403d9729f20adcc41a7f : FAILURE in 21m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1503ce0038ccf7444a2ca333a8c8db4cc4550fd0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-17 07:20:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea1e3469b55db7d118809f16030b5d18d5c613c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 09:57:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a17bb5422c974a71a7c4192c81a349b7 : SUCCESS in 7m 11s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/330a67352cd641418eb36c7f48ca94f5 : SUCCESS in 24m 06s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e12f6961670478c9229be4425917f8d : POST_FAILURE in 2h 28m 32s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/39a1c38f416d4388ac3f885bb53b3bbf : SUCCESS in 1h 03m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4219c69d3a334402ab1e8aae010e7124 : SUCCESS in 1h 11m 32s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aba5b7a153e249afabb4fce8224ec75c : SUCCESS in 1h 20m 28s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d6f5529cde9f4eb989de2778bcdb2c4a : FAILURE in 29m 00s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/586e12279ffa4105a3b3ed4a9a44be10 : FAILURE in 18m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e7aebde860cc91677a4462056fb1854b3576c6ff","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-18 14:55:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"027043067a05bb034c1aac0f22d4ccc96794481b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 16:52:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8aa614f7c9f43e0aacf759323984e5a : SUCCESS in 6m 06s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9c6a70dc0b1412eaca13a64c94a4b56 : SUCCESS in 14m 27s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b7cd5b9cc2040528b8cd759f175e174 : SUCCESS in 1h 53m 39s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/f275c01568af4349b90aeb9cc71f87de : SUCCESS in 1h 07m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ed4969a33d8f4521a8d857f4f84f80ae : SUCCESS in 1h 05m 37s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e73390d6526b4704b5367a06b4be770c : SUCCESS in 1h 05m 08s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea53d1d6b25d45c5848f15f8c39d443e : FAILURE in 17m 50s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a65c8c3c0e64ce893d961f33430739c : FAILURE in 15m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4449b4ba1c94ca61807a14b856f823ea328304f4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-19 15:31:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86320287a969992c8835f016bd9d19a1418df16d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-19 18:46:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de9648908344f35a4347591a77fd8f8 : SUCCESS in 6m 05s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1945fab3e590408d818738426121ba0c : SUCCESS in 14m 02s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/38638abde8d340eba5a2aadafbc0a180 : TIMED_OUT in 3h 10m 58s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/69c085720e3d4a3e96e8b1e6c3ae4b86 : SUCCESS in 1h 03m 34s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e067602d6f254ecaa3f695b97d53cc37 : SUCCESS in 56m 23s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d16eea2a0c9b428897efd0bae8976e1c : SUCCESS in 1h 00m 35s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/333a9e8300a448bbbc5e3eed6f49027b : FAILURE in 18m 52s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42de14726e2a48d68e0a33f574767c99 : SUCCESS in 1h 15m 05s","accounts_in_message":[],"_revision_number":1},{"id":"31dde094e2ae0225f705570fe433dc57fd96376a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-21 07:45:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a60ddbe1383f8fbf83aadad764d18771d2cfe32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 09:26:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/100ad20ff75f4cbd96787d1acc99ef62 : SUCCESS in 4m 57s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cd3eebec4a248b0b57c22e7f8d90044 : SUCCESS in 15m 32s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/84cada065f5d47dea94f21dcf6d8bdc9 : SUCCESS in 1h 31m 54s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/b25eeee9649347259c1bc3a3d3938907 : SUCCESS in 1h 17m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b1d0c7c105ef4aaaa8440a071a5a9d7f : SUCCESS in 1h 15m 41s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe9efb6a0bf042228a399a42534e1df6 : SUCCESS in 1h 20m 26s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b9a4b4fbe944085957fd30b7071ee5e : FAILURE in 26m 04s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/442b4f5a20ff4163a65295a03a2c5569 : SUCCESS in 1h 28m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9ba7649b9dd81b1131402d656318b48ee02850c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 09:30:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f00a7c29d6452eabb86ea54e3b179ecc6ab630cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 10:57:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02860278e9c34a7fb5b64b6f6217e88d : SUCCESS in 6m 21s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a549ed248444b989e1691bf583808e4 : SUCCESS in 22m 05s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/347bb992d65e4eaa89fafac94cfb41d8 : SUCCESS in 1h 25m 16s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c46684fd8a8b4d838adfbd088d0b8c69 : SUCCESS in 1h 15m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b435561ed154c749672c96554250713 : SUCCESS in 1h 19m 23s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54f48968543c4b1ab280682cc23bf169 : SUCCESS in 1h 23m 31s","accounts_in_message":[],"_revision_number":1},{"id":"800a18486969d8c5f7f0a8ea108683f4bb323a4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 10:57:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b9fe226e4f7769eea7a91c8005bf962e89cb2e9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 10:59:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0569638da4c452eb838100b582a261d : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"661d1379d954791c2af8f78199f8d2b17201ac72","revisions":{"661d1379d954791c2af8f78199f8d2b17201ac72":{"kind":"REWORK","_number":1,"created":"2022-02-15 18:48:46.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/60/829260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/60/829260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/60/829260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/60/829260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/60/829260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/60/829260/1"}}},"commit":{"parents":[{"commit":"c74c86e8a6afaae40135fc06e77d7012485b4257","subject":"Fix galera_force_bootstrap behaviour","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c74c86e8a6afaae40135fc06e77d7012485b4257"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2022-02-13 17:56:59.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","date":"2022-02-15 18:48:46.000000000","tz":0},"subject":"Use unix socket while granting access for the backup service","message":"Use unix socket while granting access for the backup service\n\n/root/.my.cnf is not stored on galera container anymore, so it\u0027s not\npossible to run mysql_user via TCP connection.\nUnix socket should be used instead.\n\nChange-Id: I71bc866aedaa6fba3cc86d2a2a99ed32f0727c54\n(cherry picked from commit 3c57409bdfd0c48e08daf34c60ca401f0861eb56)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/661d1379d954791c2af8f78199f8d2b17201ac72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/661d1379d954791c2af8f78199f8d2b17201ac72"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
