)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~831550","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ia110dd9ed09b04f6bb7a0a3adf5a808966558507","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-04-25 20:14:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia110dd9ed09b04f6bb7a0a3adf5a808966558507","subject":"Use separate tmp directory","status":"MERGED","created":"2022-03-02 15:57:48.000000000","updated":"2022-04-25 20:15:55.000000000","submitted":"2022-04-25 20:14:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"831550","meta_rev_id":"c8448bc0cd3d8f17452e1408ba56737ebf8b2be4","_number":831550,"virtual_id_number":831550,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-25 20:14:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-04-25 18:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2022-04-14 20:57:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-13 11:40:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2022-04-25 18:35:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2022-04-25 20:15:55.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-02 17:03:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-19 13:22:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-04-14 20:57:28.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2022-04-25 18:35:16.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"310404ead35b9cf844c329fd8003d7b562bfd7f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-02 15:57:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"401f731564b0476e771121ed59b55ea23175de33","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-02 16:00:38.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"fe49fc069a2366762ae869f81aec3f5b2692834a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 17:03:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53882425a384493faf50ffaa61c318be : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42fb514a74084810a1bb5709d5d22305 : SUCCESS in 6m 00s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fdea5e91b9549afaee706f6ef1714ad : SUCCESS in 18m 51s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb158b05bd9046ab87a00ebe5bac6e4c : RETRY_LIMIT in 14m 54s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/03a3ed10e7c0474698fd4fc85e7636a7 : RETRY_LIMIT in 13m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95700f58971d4cb58350d5850847c0fc : RETRY_LIMIT in 12m 27s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ffe0e5a0a8a46468f171241456cb1c5 : FAILURE in 16m 02s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/842647d169494057903dcf79c8451361 : FAILURE in 12m 01s","accounts_in_message":[],"_revision_number":1},{"id":"889ada989634e60798a39b8d660aa1c50b1ede79","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-10 15:55:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9d5f249d47484c3c875c4fe27baec4671552b9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 17:22:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d837e02a67664ee8964ff2d3154cb76a : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d6f0d6858d04a228efabdf3cd25d9ca : SUCCESS in 5m 09s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fced8dfc733948ce860bb8d102ba48eb : SUCCESS in 14m 51s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/388c71f1725e40be8db560be685f2946 : FAILURE in 44m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c74a34f387604038b76c97363b57972e : FAILURE in 36m 21s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0f5813ef1184fa2a96ea9f511704b69 : FAILURE in 35m 27s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9eaff7488674bf3b0e8ced7aede7cc9 : FAILURE in 1h 19m 13s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/344f62a20f834e80bc41e7999db36d59 : FAILURE in 1h 02m 17s","accounts_in_message":[],"_revision_number":1},{"id":"61e720a7d2c36352875bf6bfd26304d6c7b3af0d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-10 20:32:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02c5b1889ac60d913aacc681b348965fe2a04d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 21:50:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbac2dc85f044abaa84db7be699d68df : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f28c195f0b9c4e2f8a84ed304bcae3fb : SUCCESS in 6m 12s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/126e6ea744f1442fb55d29ffbf9a50f7 : SUCCESS in 15m 52s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/041d25fbb2614909b0f6dc96b0565971 : FAILURE in 43m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8f9b08239d13423f8edeee9840ea7cae : FAILURE in 42m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80a54684a4394d7bbe1a32d38f00337a : FAILURE in 35m 44s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1cf326ffcb134b13a9753f4b6680c415 : FAILURE in 1h 12m 11s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3648df8772134c7a87e79234ec12b4f5 : FAILURE in 1h 01m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6c1580c395cbbc1215c2f459c6cfc4b613a343f2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-11 10:52:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75ef08a233d2097639c675394be7fffb6355a3df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 12:43:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/059449c08c884dfd933a5b1233c9fbc0 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ffacc965dde44f5aa094609ac857cdd : SUCCESS in 5m 17s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad6bdbcb668a43678c0345fb94c0f56f : SUCCESS in 16m 27s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/64a340fbbb5a4d66b03b56b4ce8c9d0a : FAILURE in 37m 13s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/bf7e436c045a45abadcd6a089ba01b87 : FAILURE in 34m 04s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/530b42868b4642ba976968bdcc6430dd : FAILURE in 42m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6884dcea995048588efd996c9d7e3272 : FAILURE in 36m 33s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7aaee6eec8ea44edad6e2a51c60d923e : FAILURE in 1h 18m 38s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/165d6eb9654146b7a39ad7d2d0380d46 : FAILURE in 1h 33m 33s","accounts_in_message":[],"_revision_number":1},{"id":"83c038c2dfec7afcb90cd5666f3baf8449698ff2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-15 11:26:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7f583358296d98c1c0d91cacb759580eff516107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 14:41:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/174c6eee481745f7a4e9a380270a7287 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8957e7b8a3374dedb241ea43a04e2a50 : SUCCESS in 5m 05s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/160e0e150b034347b58af9411ca325ab : SUCCESS in 15m 42s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd1f9d7f62ea45bc8960b7960a6375d0 : TIMED_OUT in 3h 10m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4364f55fd69f4d20b06b9576a8ee3c6a : SUCCESS in 1h 18m 02s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/443f2b035c5346ae9266b19bbd636991 : SUCCESS in 1h 24m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e089a3b84b0b4cb3b72cff1c40eaa616 : SUCCESS in 1h 16m 32s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/31087e0c0ce6422baf01e37fd36b5d8c : SUCCESS in 1h 49m 49s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce6fef823b584890a0a345d49fa5cb44 : SUCCESS in 1h 44m 22s","accounts_in_message":[],"_revision_number":2},{"id":"8891b42e64c3af4e4bbc8dd56e182f6e9bc61ab8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-15 14:58:15.000000000","message":"Patch Set 2: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"31c3c771f10524855e7f524f506ec3915b91ef11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 16:44:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/649a22b12a7845f68542b45838cccb01 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c99ec1e13c3c4b6bb3f5dd04bcad2afb : SUCCESS in 4m 54s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82d0465f274347b581995e577d258aa8 : SUCCESS in 14m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/29ab3b834edb4429bc3470ad9c7e740d : SUCCESS in 1h 18m 45s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/147af896961a47ed9741bb8e954d0ef7 : SUCCESS in 1h 08m 54s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/f2bff219e3fd45da8b352fe7d178d59f : SUCCESS in 1h 22m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5ffb397fd85421282834175854fb3c1 : SUCCESS in 1h 15m 35s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a38b252c68d436e9e3ff9a6cca45ee7 : SUCCESS in 1h 38m 59s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61efb7ee3d2b4d1da74c811a769f98b3 : SUCCESS in 1h 34m 59s","accounts_in_message":[],"_revision_number":2},{"id":"05d79842dc922fc06df88941f690c92bd74073b9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-03-19 13:22:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90a64ffe2abe01845620656455ba95d6d49dbcfa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-24 16:02:55.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ab83a8a879aa8655bc409cc9ada0b3ad5127c57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-11 15:24:29.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3a4291bc8713280dccf411fc387be3a63502fcf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 17:07:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74ba6f50d5fc40f9a33bd3d24272444f : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/561aecbf68bb4459bf2fa9ab978da9d8 : SUCCESS in 3m 53s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a45f7b2d8e4942b99509224537566cb7 : SUCCESS in 14m 40s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a104b1e07c74813a79d0943d0ed8d3f : SUCCESS in 1h 31m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/dca682b8cb8a444baaaf19fa00469ce7 : SUCCESS in 1h 04m 20s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/8f7421b47757496293447e67c4da1110 : SUCCESS in 1h 33m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2070262df2414d4b9064475af859282a : SUCCESS in 1h 11m 40s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/943d3f00eb464b05bd26918b6dfb9439 : FAILURE in 1h 17m 34s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb402343245d4cec8e310eed10d5d3ec : FAILURE in 1h 03m 24s","accounts_in_message":[],"_revision_number":3},{"id":"e9bb4c7ee29b2a8a072268a12f3cf8804313d388","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-12 07:18:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8992171aea09dbac326f166cd77dfb9c248ef4bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 09:15:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b7124b6ed3243a8ae5f5445b97d84e3 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a72e022b6f304c45b7d160806b36921d : SUCCESS in 3m 48s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd20b10f10a5426b85475fbd34712899 : SUCCESS in 14m 39s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/26583afe4631461bbc7b165d028d5254 : SUCCESS in 1h 11m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d7119707d6e04519baa1f7e9144513a1 : SUCCESS in 1h 08m 27s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/61346ffc018146e8815afd26c3662894 : SUCCESS in 1h 24m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfc5dafdaa034d95b85a68de204371e4 : SUCCESS in 1h 07m 09s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a66c85b511e44767bfa7d5ba73ddde07 : FAILURE in 1h 18m 41s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7c5c9df21cd4b4ab42b85b080e6a59f : SUCCESS in 1h 53m 30s","accounts_in_message":[],"_revision_number":3},{"id":"19e139fb16f6bc449b8567d5a4700c772498d2a9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-13 11:40:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0ee3e626871c7316e848e675129e813bbfc123ac","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-04-14 20:57:28.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"05c787c040e9402a4e291879f1b1d250786a8ff2","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-25 18:35:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"79627c9c27136435e75611a2e6d0b2ee8939b539","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 18:38:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3f6a8a0aabe0ee8d81f87893bf85b78a0ac1a120","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 20:14:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/979072793a79451ca7817b15e3a97527 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d4edc35dabb4fe48fc677ebc59888f5 : SUCCESS in 3m 30s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/089f24207b3e453688668e603b66c544 : SUCCESS in 16m 02s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bcfad11239174dae97b93f186f5c13d6 : SUCCESS in 1h 20m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/152b89a91b324694ad1c8e6c2786ea63 : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/38ddc79c439740c49144482cd52752ee : SUCCESS in 1h 19m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53d16c39097b4ae4944c761af4eb48cd : SUCCESS in 1h 18m 04s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e24eec583dd04651a9f531fc23e811ed : SUCCESS in 1h 35m 46s","accounts_in_message":[],"_revision_number":3},{"id":"7c93824231ddf43bf3d3c17254748313383e2dd9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 20:14:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"347b660dd1609faa996feb28abf2d29f3454dd40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 20:15:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a1489d6b2024fdba02d245a55d22b82 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ebc0417919fcedd924fa5a21107055a433eca6f6","revisions":{"cdd9bb2229002f7eabcfe4ffb23ceae2f2a03c47":{"kind":"REWORK","_number":1,"created":"2022-03-02 15:57:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/831550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/50/831550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/831550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/831550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/831550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/831550/1"}}},"commit":{"parents":[{"commit":"3c57409bdfd0c48e08daf34c60ca401f0861eb56","subject":"Use unix socket while granting access for the backup service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3c57409bdfd0c48e08daf34c60ca401f0861eb56"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-02 15:57:40.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-02 15:57:40.000000000","tz":60},"subject":"Use separate tmp directory","message":"Use separate tmp directory\n\nDuring upgrades or cluster repairments, temporary directories are created\ninside /var/lib/mysql and treated as databases. This results in errors\nduring mysqlcheck like:\n\n`Got error: 1102: Incorrect database name \u0027#mysql50#tmp.stLr46FBlt\u0027`\n\nPath outside of datadir is not choosen since it could be separate mount\npoint and it\u0027s important for replication\nto survive reboots.\n\nChange-Id: Ia110dd9ed09b04f6bb7a0a3adf5a808966558507\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cdd9bb2229002f7eabcfe4ffb23ceae2f2a03c47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cdd9bb2229002f7eabcfe4ffb23ceae2f2a03c47"}]},"branch":"refs/heads/master"},"41756ef0912582336e0927ec1d145dab97ef8077":{"kind":"REWORK","_number":2,"created":"2022-03-15 11:26:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/831550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/50/831550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/831550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/831550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/831550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/831550/2"}}},"commit":{"parents":[{"commit":"3c57409bdfd0c48e08daf34c60ca401f0861eb56","subject":"Use unix socket while granting access for the backup service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3c57409bdfd0c48e08daf34c60ca401f0861eb56"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-02 15:57:40.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-15 11:26:10.000000000","tz":0},"subject":"Use separate tmp directory","message":"Use separate tmp directory\n\nDuring upgrades or cluster repairments, temporary directories are created\ninside /var/lib/mysql and treated as databases. This results in errors\nduring mysqlcheck like:\n\n`Got error: 1102: Incorrect database name \u0027#mysql50#tmp.stLr46FBlt\u0027`\n\nPath outside of datadir is not choosen since it could be separate mount\npoint and it\u0027s important for replication\nto survive reboots.\n\nChange-Id: Ia110dd9ed09b04f6bb7a0a3adf5a808966558507\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/41756ef0912582336e0927ec1d145dab97ef8077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/41756ef0912582336e0927ec1d145dab97ef8077"}]},"branch":"refs/heads/master"},"ebc0417919fcedd924fa5a21107055a433eca6f6":{"kind":"REWORK","_number":3,"created":"2022-04-11 15:24:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/831550/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/50/831550/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/831550/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/831550/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/831550/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/50/831550/3"}}},"commit":{"parents":[{"commit":"3c57409bdfd0c48e08daf34c60ca401f0861eb56","subject":"Use unix socket while granting access for the backup service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3c57409bdfd0c48e08daf34c60ca401f0861eb56"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-02 15:57:40.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-04-11 15:24:29.000000000","tz":0},"subject":"Use separate tmp directory","message":"Use separate tmp directory\n\nDuring upgrades or cluster repairments, temporary directories are created\ninside /var/lib/mysql and treated as databases. This results in errors\nduring mysqlcheck like:\n\n`Got error: 1102: Incorrect database name \u0027#mysql50#tmp.stLr46FBlt\u0027`\n\nPath outside of datadir is not choosen since it could be separate mount\npoint and it\u0027s important for replication\nto survive reboots.\n\nChange-Id: Ia110dd9ed09b04f6bb7a0a3adf5a808966558507\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ebc0417919fcedd924fa5a21107055a433eca6f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ebc0417919fcedd924fa5a21107055a433eca6f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
