)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~831552","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I6fcefe216236ffea60da5fee42aad47c6f7da133","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"831552","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-04-25 20:17:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6fcefe216236ffea60da5fee42aad47c6f7da133","subject":"Add galera_data_dir variable","status":"MERGED","created":"2022-03-02 16:07:18.000000000","updated":"2022-04-25 20:18:50.000000000","submitted":"2022-04-25 20:17:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831552-831552","meta_rev_id":"62761091b096d59fe3df4b91fd6d4e989cb35337","_number":831552,"virtual_id_number":831552,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-25 20:17:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-04-25 20:18:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-13 11:39:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-13 11:39:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-02 17:03:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-15 15:46:45.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-03-16 13:30:26.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"7b066e819b088741a550be3220a7fe4d7acd2f58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-02 16:07:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d295f023a54ff06d07b51123f94295e38acfd673","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 17:03:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3e52e2e5af4a3796046ce6d9829c2a : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91e98f8f350d4e208bf4c45b5d1055d8 : SUCCESS in 5m 11s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d5a80bd0d7a44759779a1606b15d262 : SUCCESS in 16m 16s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8398dacf03d44b189a46e88bac966f5e : RETRY_LIMIT in 13m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/af7fd43773b34be3a88892eb9bf23899 : RETRY_LIMIT in 12m 53s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfda7fd19a0641c0bf9602c0e265aa66 : RETRY_LIMIT in 12m 50s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/59d98bd897b542bdafdd2a98fb344208 : FAILURE in 13m 17s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78690ab640af4a0cad3727fe45837253 : FAILURE in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"2e5c2e908ef2141dc3ec5c70d21da18942e835f8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-10 15:55:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"857948c61ad3d4e0345e4bd6c899e23a5f16af4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 17:28:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/712dafb5ef124193b5fabecdb45c8e23 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9190f055b5ba45d4b4a2f4512d1f3819 : SUCCESS in 5m 54s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/218b92c3321b484e944611da7c0773bc : SUCCESS in 22m 14s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/77e3fdaf498e4306942d1214a25226da : FAILURE in 39m 59s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/cac64b058e674c75839783ddd21ab7b0 : FAILURE in 41m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c62d1ffcc8904156a310563d84ff1397 : FAILURE in 40m 08s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d91116effe81407ea7383fa3e32b922e : FAILURE in 1h 20m 58s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f245e0f120b4113ab8ae98b265e7d53 : FAILURE in 1h 02m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0d8df10f28f5efd9dfb880a73a13bbdbe5dd355b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-10 20:32:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a125297f92a5cef8eb7c8486dac1e1061e67eac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 21:58:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6fcc4215cd84bfeb17c2dcc8cbb8c8a : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/614c05387fd542f59447e3bb4a080611 : SUCCESS in 5m 39s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd9f2257860e4f4c9d771ce1ca62bafc : SUCCESS in 16m 01s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/af42f59d2e09438f805c10c91ad7f6ca : FAILURE in 55m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6cd1ce01400c40c4be1ce87f25d29850 : FAILURE in 27m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21dbaa726c094ed7b8b205b22c914529 : FAILURE in 31m 56s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c50be003ffb4cad925923267e369308 : FAILURE in 1h 07m 50s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6a0acef56a140919b182584ac765010 : FAILURE in 1h 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"43011bbcf99bcaae4df47a6cd2ffd8f2c14eee52","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-11 10:52:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"772cfea67ad1b3361c89548e089ae96c4e7c72d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 12:18:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8355e283727b447eac09a11a0358604d : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50ec24c63a09435f8a72655d41946455 : SUCCESS in 5m 36s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3043ddd6e2314413a0e374d3bfc620e6 : SUCCESS in 19m 47s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a25d9725725948a383a2cdb9a4ee8c39 : FAILURE in 43m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/dc54e50957a94815b8bce51a61eb005f : FAILURE in 32m 24s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/beaf1223b05f4d06b80fc050b0ab0933 : FAILURE in 39m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/965508c91a1249c0a4c47f77b666fb80 : FAILURE in 38m 34s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c856eeacf1d2499da6f0ec5378f494ab : FAILURE in 1h 13m 51s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/851d5a2c241247ed938a8dcbd3927f5b : FAILURE in 1h 07m 41s","accounts_in_message":[],"_revision_number":1},{"id":"179e2348be4f5f1dea0084b833b90d231d9b2d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-15 11:28:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e5c2c5f3fa61ce080578b6594ef700e0fd2c483a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-15 12:27:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f8de53b446831c5c9667e5a2ae9ff8284bba2701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 14:10:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f7169e9cdf4147ba4750bcb37caef4 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54097221b8134d79b6843f5fc9661018 : SUCCESS in 5m 18s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1716b477bce64f418782bfced0c246b1 : SUCCESS in 15m 29s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/60039ee566194bf285b8d4dd3a54f706 : SUCCESS in 1h 28m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/81a21724538d403aaa013b69709bc524 : SUCCESS in 1h 22m 22s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/abc03f19d2b44228a1fdb8c47e8706b8 : SUCCESS in 1h 16m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6d5d59fa03a4a35ad6543e7a0b0b7b0 : SUCCESS in 1h 07m 41s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/961a6ee42c3d43608df182af52fff61a : SUCCESS in 1h 35m 46s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa20fa881cd3452f87554dd5c2ad1632 : SUCCESS in 1h 24m 52s","accounts_in_message":[],"_revision_number":3},{"id":"d2259ecdea524e4a902185bd4301b4f84a7a12c5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-03-15 15:46:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dae1d3ab9073aa6c9b37cd76605a14a51dab57d8","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-03-16 13:30:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3bd85dcd47d108d038d5256bce2024a9ea4f97ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-12 11:27:03.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"8c8e7d433da75763f8613797fc4c25d5ad2651b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 13:41:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31223486ccca4361a943279d85b7adb7 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d02c7466a8a4c029ac3f30f17d26730 : SUCCESS in 4m 21s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5492a85702ac4dfe85cc0c1297354f2f : SUCCESS in 18m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad30c30c18e4447d8c95395056a8a18a : SUCCESS in 1h 24m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b0d88e750ebc4eef9ed9ec630abfcefa : SUCCESS in 1h 12m 34s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e6cf13e9d80448b2a35220fe61d4c73c : SUCCESS in 1h 15m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc6a15b26a004c64aae49280707415c1 : SUCCESS in 1h 23m 10s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/15f55ca0de0b450693b187024a8a1569 : SUCCESS in 1h 56m 50s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/572081afc8fa4a75a55fad4b60467b1f : SUCCESS in 1h 30m 00s","accounts_in_message":[],"_revision_number":4},{"id":"d78c3f3dbf56a2a801af9952a44b1aee44d29e6c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-13 11:39:24.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d1af28bcb4d85a459a42863e2045d1a5e0e8e679","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 18:38:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"27fc334c79c54ccb9955e2224f80b9e3e813ad1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 20:17:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cdd7c29cda54d74b86a7f3f3ae31c55 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf2d1ce1d4624f6689d23f23453a3e4a : SUCCESS in 3m 55s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b28f7e7ab81948a8b37a8efecdfcf523 : SUCCESS in 14m 12s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf101d2487bc49a0bb8bd88cf757232a : SUCCESS in 1h 16m 52s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/44c7f59dd4e149c5b71558f9b20812f1 : SUCCESS in 55m 00s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d9a45ae882244939bdba9d4d290e81d1 : SUCCESS in 1h 33m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8529d6361cff4b9cae30027ccb4a5dab : SUCCESS in 1h 10m 04s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86782ccc4a30494e923668f2129e8bf8 : SUCCESS in 1h 24m 02s","accounts_in_message":[],"_revision_number":4},{"id":"940aea31687041a4c66a57a34484a40672878aaa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 20:17:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f407a2c3970277317d1e22ad7263c067f7db2d0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 20:18:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8dddfb8a88f412c94b70d3e38871028 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5884ee918e96c8656be04d052ae1b152e8053f58","revisions":{"1a1a17bb21700488a54b318ba478794ddcce510d":{"kind":"REWORK","_number":1,"created":"2022-03-02 16:07:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/52/831552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/52/831552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/1"}}},"commit":{"parents":[{"commit":"cdd9bb2229002f7eabcfe4ffb23ceae2f2a03c47","subject":"Use separate tmp directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cdd9bb2229002f7eabcfe4ffb23ceae2f2a03c47"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-02 16:07:06.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-02 16:07:06.000000000","tz":60},"subject":"Add galera_data_dir variable","message":"Add galera_data_dir variable\n\nControl mysql datadir with variable. Decrease code dublication since path\nis heavily used in different places. If path needs to be changed\noverriding config won\u0027t be enough.\n\nChange-Id: I6fcefe216236ffea60da5fee42aad47c6f7da133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1a1a17bb21700488a54b318ba478794ddcce510d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1a1a17bb21700488a54b318ba478794ddcce510d"}]},"branch":"refs/heads/master"},"697c0711c0a1c3253478961a4674a0aadf000497":{"kind":"REWORK","_number":2,"created":"2022-03-15 11:28:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/52/831552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/52/831552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/2"}}},"commit":{"parents":[{"commit":"cdd9bb2229002f7eabcfe4ffb23ceae2f2a03c47","subject":"Use separate tmp directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cdd9bb2229002f7eabcfe4ffb23ceae2f2a03c47"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-02 16:07:06.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-15 11:28:03.000000000","tz":0},"subject":"Add galera_data_dir variable","message":"Add galera_data_dir variable\n\nControl mysql datadir with variable. Decrease code dublication since path\nis heavily used in different places. If path needs to be changed\noverriding config won\u0027t be enough.\n\nChange-Id: I6fcefe216236ffea60da5fee42aad47c6f7da133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/697c0711c0a1c3253478961a4674a0aadf000497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/697c0711c0a1c3253478961a4674a0aadf000497"}]},"branch":"refs/heads/master"},"f2efc69c197fdc1011ae02eff455a773a46ed2fe":{"kind":"REWORK","_number":3,"created":"2022-03-15 12:27:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/52/831552/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/52/831552/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/3"}}},"commit":{"parents":[{"commit":"41756ef0912582336e0927ec1d145dab97ef8077","subject":"Use separate tmp directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/41756ef0912582336e0927ec1d145dab97ef8077"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-02 16:07:06.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-15 12:27:34.000000000","tz":60},"subject":"Add galera_data_dir variable","message":"Add galera_data_dir variable\n\nControl mysql datadir with variable. Decrease code dublication since path\nis heavily used in different places. If path needs to be changed\noverriding config won\u0027t be enough.\n\nChange-Id: I6fcefe216236ffea60da5fee42aad47c6f7da133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f2efc69c197fdc1011ae02eff455a773a46ed2fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f2efc69c197fdc1011ae02eff455a773a46ed2fe"}]},"branch":"refs/heads/master"},"5884ee918e96c8656be04d052ae1b152e8053f58":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-04-12 11:27:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/52/831552/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/52/831552/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/52/831552/4"}}},"commit":{"parents":[{"commit":"ebc0417919fcedd924fa5a21107055a433eca6f6","subject":"Use separate tmp directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ebc0417919fcedd924fa5a21107055a433eca6f6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-03-02 16:07:06.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-04-12 11:27:03.000000000","tz":0},"subject":"Add galera_data_dir variable","message":"Add galera_data_dir variable\n\nControl mysql datadir with variable. Decrease code dublication since path\nis heavily used in different places. If path needs to be changed\noverriding config won\u0027t be enough.\n\nChange-Id: I6fcefe216236ffea60da5fee42aad47c6f7da133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5884ee918e96c8656be04d052ae1b152e8053f58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5884ee918e96c8656be04d052ae1b152e8053f58"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
