)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~847435","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fxena~I90286b01b5578df04447155c69c92f823388185a","project":"openstack/openstack-ansible-galera_server","branch":"stable/xena","topic":"mariadb-bump-stable/xena","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-07-20 09:27:44.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-07-20 09:27:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I90286b01b5578df04447155c69c92f823388185a","subject":"Bump mariadb version.","status":"MERGED","created":"2022-06-24 08:27:06.000000000","updated":"2022-07-20 09:28:55.000000000","submitted":"2022-07-20 09:27:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847435-mariadb-bump-stable/xena","cherry_pick_of_change":842106,"cherry_pick_of_patch_set":1,"meta_rev_id":"a825b6a1fc010b9b41412d41cd432aeff14e9507","_number":847435,"virtual_id_number":847435,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-20 09:27:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-07-20 07:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-19 17:42:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":1,"date":"2022-07-20 07:56:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-07-20 07:47:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 08:27:06.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"CC"},{"updated":"2022-06-24 10:32:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-18 11:03:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2022-07-19 17:42:36.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-07-20 07:47:26.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2022-07-20 07:56:04.000000000","updated_by":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"reviewer":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"state":"REVIEWER"}],"messages":[{"id":"b0f44ea821094437b27c49cdf07bbbd48da81d70","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-24 08:27:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* defaults/main.yml","accounts_in_message":[],"_revision_number":1},{"id":"8b9dda3cd8f16f2e516e13980d5eff39fec96e90","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-24 08:27:29.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1618d55825acbf144d02e5a5fd3759fff1216e1a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-24 08:29:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"07aa0f6266d4f75bb434ae72393717a740c70eee","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-24 08:30:02.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"f56f673be9d52c49b6ce1df7e4f790c849ff6e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 10:32:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e88f903efcd3416c922e182023d9c7b3 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ad7fe545678469495a807604bacf67f : SUCCESS in 4m 45s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dac49e5f54014a169fc71f7ac43b7603 : SUCCESS in 20m 18s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/23789032a7d641cda7a1f0a7f7dab6ed : POST_FAILURE in 1h 58m 55s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/5d883c538c2348d6a2e6bb58e48199a6 : SUCCESS in 1h 17m 06s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8f6cd76967c24c6b88848c15e32eaf48 : POST_FAILURE in 1h 34m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6d815f5290a41c5a09f2e6af6adb02b : SUCCESS in 1h 21m 18s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1116885475ba419599471b6b9ab6c728 : SUCCESS in 1h 55m 02s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6199ace6ce8d40b8a9e5beaf7b3fdaa7 : SUCCESS in 1h 35m 02s","accounts_in_message":[],"_revision_number":3},{"id":"b9e7bc68e8ad8ba14732d8ce5cc6b6530258b2ff","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-24 12:23:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6e34fd8ef28699b830c556c2809464e70cd17cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 14:35:03.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4078c1ca3b8d419694590aaa0d0c84ed : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c8c1758acca46b289d35fd6a4e94912 : SUCCESS in 3m 37s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/139b652ae98e4b35898d200d39b98f77 : SUCCESS in 14m 39s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b97cd1e2903c4bab904d8de5bcfade0c : SUCCESS in 1h 19m 47s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/eba2b1a041b840d59f9dc2fb95f660e3 : FAILURE in 1h 05m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1430e8d40dd9479b82e7bb5f3e080fa1 : SUCCESS in 1h 10m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4997a184cff49ceb8afc86d46d52802 : SUCCESS in 1h 10m 47s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/82bf390888424126a257d8047e1d17d1 : SUCCESS in 1h 21m 14s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e414ada2d5794d5c94819fe4a6a00e18 : SUCCESS in 1h 33m 21s","accounts_in_message":[],"_revision_number":3},{"id":"cd4d63c00bbaa95a2f3a8eb13a12e429184d876f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-14 06:10:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a7bb58fa2c5f0a3bed6ce5fe38e2a5edc99f6675","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-14 06:12:02.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"24639eddc2ebbf4e70223c70e8609f8a7000e345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-14 06:13:25.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"5d80f2bd65d24ef5649f1bf32ce5e723f0f6e4f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 07:56:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/698515bf11e3480c8d9a5a758003540e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c86e04439e74f9ba35af3fdc54cb055 : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cb4eaf9986d4d72b54493d98609efe2 : SUCCESS in 6m 09s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/701259020821456aac46d94e1f63c2f3 : SUCCESS in 38m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/65e1a4a00e0d407c8b0bcc7868937574 : SUCCESS in 1h 34m 52s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c3db727d781641eb964fc0804f5e1e95 : SUCCESS in 1h 33m 10s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e96f5bd60c56455bb86a316150291f0c : SUCCESS in 1h 26m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d947a79739334d6aae717f28e5bed0a2 : SUCCESS in 1h 36m 18s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/adb5707c1ea4427286a1b40e6f6e8599 : FAILURE in 1h 27m 00s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fcb8d3b970446698175d4c149b10877 : FAILURE in 1h 03m 16s","accounts_in_message":[],"_revision_number":5},{"id":"c04a0823ed32f35eaccc0dd1c64cb00b22d6880f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-18 11:03:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0cae7335c5bdde8366170b5f4b873bf6cae5a18c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 12:44:03.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/139877d9606f46f687d0a07a4efcfe8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc40580c85840bbbd78838661aac9a4 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9940fc93b13f40029eb97e8bc126179a : SUCCESS in 4m 39s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39de67bc73d24f6a8ab446ea1008f933 : SUCCESS in 22m 06s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/525a4c4cd9a848f7974ac8409f583102 : FAILURE in 1h 04m 06s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/d96e07c659ba4df59b51315af24f0f77 : SUCCESS in 1h 22m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/2b21a41dc62f4d68b75a56c170be669b : SUCCESS in 1h 18m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ebf9d3341154ceb843c28893f7625cf : SUCCESS in 1h 20m 50s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/84c047cdac0443f4be9e79668a22a01f : FAILURE in 1h 33m 19s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/497fa424429041dabe89a58949c9caf0 : SUCCESS in 1h 33m 33s","accounts_in_message":[],"_revision_number":5},{"id":"f82544f8b03faa761e729ba076d79ae8e93999a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-19 08:57:17.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"1e45da9e67ea1092e1b83c89d6055b7d8d825cf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 10:59:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5799bd8cabc243139bb247ff5812ec46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4828b5752ae4f03bf3abce22c1c87cb : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b250a5ce1fb34a66b21627a0b0ea4be7 : SUCCESS in 4m 39s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ddcc555dad542e8ab5cef59d6ad3242 : SUCCESS in 24m 12s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e06fbd1b4054e198883f10193b62351 : SUCCESS in 1h 38m 18s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/29b5af4b5ea64599b561314ccd74c18d : SUCCESS in 1h 19m 34s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b889b43d78f943ed9abee7bedc88dd3c : SUCCESS in 1h 25m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb49be3945f246c6a7456a5dbbd28f11 : SUCCESS in 1h 16m 03s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/15d4a15b903a407b8e3b1ff918d0123f : SUCCESS in 1h 53m 11s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/863da3308e5e4ae7beefa9443eb2c1c5 : FAILURE in 1h 39m 52s","accounts_in_message":[],"_revision_number":6},{"id":"3221b1d850729414acbc5713272dbe6eb788cb58","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-19 15:45:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b7f6de94a9622061e1ad80ae161c64ceae3b2205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 17:37:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86cc05e7fdcc4b30842d74858b56fef8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2f17522eee4421f94ceed2134706543 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7489fc87336f4a269401e6a5a029381d : SUCCESS in 3m 43s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b9d31defd4d4d94bbb1b06e1ad4ffc4 : SUCCESS in 13m 18s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d25b59f0ecd6493593ffe223e1dfbc5c : SUCCESS in 1h 25m 16s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/76ee5c2386a240ed80455a97b52cc879 : SUCCESS in 1h 14m 51s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/06cd75f336eb4268aa457d31996d1510 : SUCCESS in 1h 19m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1a2439e6b374a3394bff59370753647 : SUCCESS in 1h 18m 58s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7112988f92d42fb90dabbcc4f85deb6 : SUCCESS in 1h 46m 46s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d20e8a696c54414bf1a5223f1133822 : SUCCESS in 1h 37m 42s","accounts_in_message":[],"_revision_number":6},{"id":"ee8f5da0b10ef0f90c46cb2ac8981295bc169dcd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-19 17:42:36.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f617b3acf67a963c67813a43e865fd319075def9","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-07-20 07:47:26.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"68a675680d890fe29f807379d9b95cd31052cb1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 07:49:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b79daf341ff95d290fa514508b29bc9b8140551d","author":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"date":"2022-07-20 07:56:04.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"bf5cc79c265c2c835e3bb5462192717bc96f92f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 09:27:43.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0935d015ac28473d9a70b303a5f50439\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/510ff825c37447c7b7b951e0e2b4c5ef : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a55b1f3749d74225839e06c753bf5c2d : SUCCESS in 4m 31s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c43099490294491cb0348eaf7d3766a3 : SUCCESS in 18m 33s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e6bdce675ce4cf78ecd839e8290460b : SUCCESS in 1h 26m 46s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/06f1f88359754e2e83752780871906ea : SUCCESS in 1h 11m 45s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b65c2c5b993b4b219b203e35b6989183 : SUCCESS in 1h 12m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ded39b1ef3b4625ae63d2ec39dfff07 : SUCCESS in 1h 14m 20s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c07a04770dc54d169b3241a0972a810b : SUCCESS in 1h 34m 11s","accounts_in_message":[],"_revision_number":6},{"id":"cf498736f2a17b40fed50789bc9e887cd91da3c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 09:27:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"a825b6a1fc010b9b41412d41cd432aeff14e9507","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 09:28:55.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2906d52486bb40c38ea7592346cd5597\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d323c2476af9445dad8f5cf58dca31d0 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"04224a4e2890cca15fe2ebd3ff30e21ae768d568","revisions":{"fef21c94092d7886b5e30d4ce715e5f888f60b7f":{"kind":"REWORK","_number":1,"created":"2022-06-24 08:27:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/35/847435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/35/847435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/1"}}},"commit":{"parents":[{"commit":"7585db20e60f8397d3429c2afe10d2dca7f4f3f3","subject":"Use separate tmp directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7585db20e60f8397d3429c2afe10d2dca7f4f3f3"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-17 13:37:59.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-06-24 08:27:06.000000000","tz":0},"subject":"Bump mariadb version.","message":"Bump mariadb version.\n\nThis bump will fix:\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27376\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27377\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27380\n\nChange-Id: I90286b01b5578df04447155c69c92f823388185a\n(cherry picked from commit 1013ee3a5dc2b48fdba63c6e2df8b5e1fb7860ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fef21c94092d7886b5e30d4ce715e5f888f60b7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fef21c94092d7886b5e30d4ce715e5f888f60b7f"}]},"branch":"refs/heads/stable/xena"},"bd0def663f36b6017159c653298f1019b1523d2b":{"kind":"REWORK","_number":2,"created":"2022-06-24 08:27:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/35/847435/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/35/847435/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/2"}}},"commit":{"parents":[{"commit":"7585db20e60f8397d3429c2afe10d2dca7f4f3f3","subject":"Use separate tmp directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7585db20e60f8397d3429c2afe10d2dca7f4f3f3"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-17 13:37:59.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-06-24 08:27:25.000000000","tz":0},"subject":"Bump mariadb version.","message":"Bump mariadb version.\n\nThis bump will fix:\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27376\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27377\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27380\n\nChange-Id: I90286b01b5578df04447155c69c92f823388185a\n(cherry picked from commit 1013ee3a5dc2b48fdba63c6e2df8b5e1fb7860ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/bd0def663f36b6017159c653298f1019b1523d2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/bd0def663f36b6017159c653298f1019b1523d2b"}]},"branch":"refs/heads/stable/xena"},"d222c9baf4703128373fecd4b07e2b22263d4c43":{"kind":"REWORK","_number":3,"created":"2022-06-24 08:29:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/35/847435/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/35/847435/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/3"}}},"commit":{"parents":[{"commit":"7585db20e60f8397d3429c2afe10d2dca7f4f3f3","subject":"Use separate tmp directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7585db20e60f8397d3429c2afe10d2dca7f4f3f3"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-17 13:37:59.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-06-24 08:29:43.000000000","tz":120},"subject":"Bump mariadb version.","message":"Bump mariadb version.\n\nThis bump will fix:\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27376\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27377\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27380\n\nChange-Id: I90286b01b5578df04447155c69c92f823388185a\n(cherry picked from commit 1013ee3a5dc2b48fdba63c6e2df8b5e1fb7860ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d222c9baf4703128373fecd4b07e2b22263d4c43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d222c9baf4703128373fecd4b07e2b22263d4c43"}]},"branch":"refs/heads/stable/xena"},"fd8a09b8e6ed61a695e71c04aab36ca0d30c5470":{"kind":"REWORK","_number":4,"created":"2022-07-14 06:12:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/35/847435/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/35/847435/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/4"}}},"commit":{"parents":[{"commit":"7585db20e60f8397d3429c2afe10d2dca7f4f3f3","subject":"Use separate tmp directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7585db20e60f8397d3429c2afe10d2dca7f4f3f3"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-17 13:37:59.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-07-14 06:12:01.000000000","tz":0},"subject":"Bump mariadb version.","message":"Bump mariadb version.\n\nThis bump will fix:\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27376\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27377\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27380\n\nWe also modify workaround applied for 10.6 upgrade wrt to bug [1]\nas it has been added extra tools to help with checking state of upgrade.\n\nNew flag --check-if-upgrade-is-needed is checking if any upgrade is\nalready running and waits until it\u0027s finished.\nIt exits with rc 0 if upgrade is required and 1 if not.\n\nIf upgrade is required, we fall into rescue and perform upgrade.\n\nChange-Id: I90286b01b5578df04447155c69c92f823388185a\n(cherry picked from commit 1013ee3a5dc2b48fdba63c6e2df8b5e1fb7860ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fd8a09b8e6ed61a695e71c04aab36ca0d30c5470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fd8a09b8e6ed61a695e71c04aab36ca0d30c5470"}]},"branch":"refs/heads/stable/xena"},"6d29bab76ef3bd4c54534abe4df22ac6bbab1a11":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-07-14 06:13:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/35/847435/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/35/847435/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/5"}}},"commit":{"parents":[{"commit":"7585db20e60f8397d3429c2afe10d2dca7f4f3f3","subject":"Use separate tmp directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7585db20e60f8397d3429c2afe10d2dca7f4f3f3"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-17 13:37:59.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-07-14 06:13:25.000000000","tz":0},"subject":"Bump mariadb version.","message":"Bump mariadb version.\n\nThis bump will fix:\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27376\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27377\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27380\n\nWe also modify workaround applied for 10.6 upgrade wrt to bug [1]\nas it has been added extra tools to help with checking state of upgrade.\n\nNew flag --check-if-upgrade-is-needed is checking if any upgrade is\nalready running and waits until it\u0027s finished.\nIt exits with rc 0 if upgrade is required and 1 if not.\n\nIf upgrade is required, we fall into rescue and perform upgrade.\n\n[1] https://jira.mariadb.org/browse/MDEV-27068\n\nChange-Id: I90286b01b5578df04447155c69c92f823388185a\n(cherry picked from commit 1013ee3a5dc2b48fdba63c6e2df8b5e1fb7860ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6d29bab76ef3bd4c54534abe4df22ac6bbab1a11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6d29bab76ef3bd4c54534abe4df22ac6bbab1a11"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"},"04224a4e2890cca15fe2ebd3ff30e21ae768d568":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-07-19 08:57:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/35/847435/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/35/847435/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/35/847435/6"}}},"commit":{"parents":[{"commit":"7585db20e60f8397d3429c2afe10d2dca7f4f3f3","subject":"Use separate tmp directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7585db20e60f8397d3429c2afe10d2dca7f4f3f3"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-17 13:37:59.000000000","tz":-240},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-07-19 08:57:17.000000000","tz":0},"subject":"Bump mariadb version.","message":"Bump mariadb version.\n\nThis bump will fix:\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27376\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27377\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27380\n\nWe also modify workaround applied for 10.6 upgrade wrt to bug [1]\nas it has been added extra tools to help with checking state of upgrade.\n\nNew flag --check-if-upgrade-is-needed is checking if any upgrade is\nalready running and waits until it\u0027s finished.\nIt exits with rc 0 if upgrade is required and 1 if not.\n\nIf upgrade is required, we fall into rescue and perform upgrade.\n\n[1] https://jira.mariadb.org/browse/MDEV-27068\n\nChange-Id: I90286b01b5578df04447155c69c92f823388185a\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-rabbitmq_server/+/850233\n(cherry picked from commit 1013ee3a5dc2b48fdba63c6e2df8b5e1fb7860ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/04224a4e2890cca15fe2ebd3ff30e21ae768d568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/04224a4e2890cca15fe2ebd3ff30e21ae768d568"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
