)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~847782","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fwallaby~I90286b01b5578df04447155c69c92f823388185a","project":"openstack/openstack-ansible-galera_server","branch":"stable/wallaby","topic":"mariadb-bump-stable/wallaby","attention_set":{},"removed_from_attention_set":{"819":{"account":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"last_update":"2022-07-14 08:32:16.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-07-14 08:32:16.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-07-13 15:47:13.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"I90286b01b5578df04447155c69c92f823388185a","subject":"Bump mariadb version.","status":"MERGED","created":"2022-06-27 13:06:34.000000000","updated":"2022-07-14 08:33:15.000000000","submitted":"2022-07-14 08:32:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847782-mariadb-bump-stable/wallaby","meta_rev_id":"cc780b018f67e4d6389ef58a045a27c371fd24ae","_number":847782,"virtual_id_number":847782,"owner":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-14 08:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-07-14 06:07:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-14 06:58:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-14 06:58:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-27 13:06:34.000000000","updated_by":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"CC"},{"updated":"2022-06-27 14:04:31.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-06-27 15:09:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-14 06:58:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"1ef40c5c907ef64145bc159f0fbee53b5379827a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"date":"2022-06-27 13:06:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"babfdf58c2d3e48e8cafb70d6ea82fe50405bb23","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-27 14:04:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"03e347ef535aa810fac4a74d1fbab0144d6c32e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 15:09:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baa1109ace6f458183ef7d5929242004 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d64d7feb5674ed4bb61fa9495e9daf1 : SUCCESS in 4m 42s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f763b43beb5f4aa7824cb9f3ec5446ec : SUCCESS in 13m 27s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ac779cc1b844d2d8cfb78a516656170 : POST_FAILURE in 1h 53m 17s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/da7ed7ef20ea4e10ac60c4adc6b5c5a8 : SUCCESS in 1h 08m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/26034b787c464e19a288d46ec22dc117 : SUCCESS in 1h 13m 39s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6ce6523c9a24b81841b580177406f6c : POST_FAILURE in 1h 32m 28s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/098e0c6da1da47b18489233e171b9004 : FAILURE in 22m 56s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e44da4fa87ac473395c70a01745fa973 : FAILURE in 1h 03m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b42ce473f6204bf4d500ffd95727609f05b08dca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-06-27 17:49:29.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3137002029c6739063c540ea28af5a2bef378c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-06-27 17:49:41.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0d9aa4e3d103fbee0ce9b700132a100c52b1fe86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 19:27:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a76ad5b59c3f4833ad7312c9b9e7e6de : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b3da5e14d2c42ba88cfbab629f0c25e : SUCCESS in 3m 44s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb854e7e78224d9fb5ff859047105b32 : SUCCESS in 19m 31s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c0dabb1475f4036b1dd8bbbccedf918 : SUCCESS in 1h 30m 52s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/cfd4d1485fd54cd583ae173803043c94 : SUCCESS in 1h 11m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/62f0e6daa8184d8fb7db65d473e973f5 : SUCCESS in 1h 04m 13s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b801fa8fa994021914423b55cdd2835 : SUCCESS in 1h 15m 28s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7f01dc855634b9aa41b58f3458557c7 : FAILURE in 29m 41s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cea074963b4e41b9b69b53780d121b27 : FAILURE in 1h 14m 55s","accounts_in_message":[],"_revision_number":3},{"id":"93a5825e0d30c57458847c93efb16120072548b2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-13 07:43:47.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fb97e8c0a529faadab4958d8a422f4af12165feb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 09:13:29.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67014aefafc84039aeacf697bd4d53d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89372a57a8b74962a665527d210dd690 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a26ae92fb3349c19cfdad51016bdc71 : SUCCESS in 3m 12s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce19bdc45fe242f2aa2ae865c250dd32 : SUCCESS in 12m 46s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/46aa35c302a5491ab938ab0856fe5e89 : SUCCESS in 1h 23m 16s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e921b286ffa3401695ed05af9d146e70 : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a15feba55e3f44a6b66603ac3a427af2 : SUCCESS in 1h 11m 27s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/842d3dff10584a08ae32ae1f938af4b0 : SUCCESS in 1h 08m 18s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7be4d129477146128359062066fb8aae : FAILURE in 23m 46s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4f8f29cbb2245e8abca1c8cf1a95362 : FAILURE in 1h 08m 29s","accounts_in_message":[],"_revision_number":3},{"id":"f7943ef9e6cfa099cfac8737782f6b0260464946","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-13 12:39:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ded9786a3f56f7449b4e0bcd1533bcde0229e3c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-13 13:09:03.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_28619\u003e\n","accounts_in_message":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"_revision_number":3},{"id":"9f433a7dfa8ae7c6527a855c1a438656d021eaab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-13 13:12:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"42a818a0065621753048987c7935b04f4a2b7d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 15:47:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b53016386be43749e862a059a26ca02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/357081a1fd044aa4b96bf34805f1413f : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b867acede72f4d709baba7a5b409a450 : SUCCESS in 4m 24s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e14d8add34bf4e7c955d6384c762b164 : SUCCESS in 38m 57s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/345658ff6e2b469fa08ff346c7e26cdf : FAILURE in 1h 25m 05s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/89a6d9bf15c845e99c9545d63cee4cb4 : FAILURE in 1h 13m 07s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/80c81ddbcfda41628428bebebe008f49 : FAILURE in 1h 10m 58s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29cc0db3d14b40c09a0dd127da0fd5a0 : FAILURE in 1h 06m 59s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/189506cb9d5744e6a1d92c1476286b27 : FAILURE in 23m 24s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71a7879a4242467ca2a144add96df65e : SUCCESS in 1h 49m 46s","accounts_in_message":[],"_revision_number":4},{"id":"e3c29b6ae4d52cd6e87a178f15a90efebf0c3938","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-13 19:37:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"772947cb1b39ea4719506f9ac47398661ae1922e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 21:09:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b53457a039c4a63b475d5c1acd085de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b2551198ab74c2a9955f3087a5600ca : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a02e0070da514d2aa4017ad0d0ce4c5d : SUCCESS in 2m 58s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1ae858309994f119e79d92a477ad174 : SUCCESS in 18m 51s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7e4e58346cc45f6956e5c1ded49e69f : SUCCESS in 1h 30m 01s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/ed9ec00d91f84a2e90e1b0f90c27a51e : SUCCESS in 1h 21m 51s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/93cfd0e851304b568a401dd009fd7914 : SUCCESS in 1h 15m 44s (non-voting)\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bffb04d37e74fa0aa5201be1f6eb128 : SUCCESS in 1h 14m 31s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/76686b4794ec421da7727bc0e8da7eca : FAILURE in 30m 48s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfa2bb4ee76749bbbd84f3fcd9c95687 : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":4},{"id":"80e4a16ae027a8b76816d41dfc62377e60a1aa90","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-07-14 06:07:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8584c861f97b28116098573ff67362f5401620f2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-07-14 06:58:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"10f21b3413cced6c280360bec325f1d18201804b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 07:00:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a1f789881cbd5642c652660c42a987346f3da9f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 08:32:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b09f28bd2244260a84078e5ee4f7af4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333be2d79a2d4c24a68c596abb320cc5 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/481939c53a2d4198b88560fa52992228 : SUCCESS in 3m 55s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20150a3ac6ee4c0a979937b5484e92aa : SUCCESS in 20m 48s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b978311f02214bc1b1d22098fbf3f58c : SUCCESS in 1h 21m 49s\n- openstack-ansible-deploy-infra_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/69e591fbccad4f418df8531c70adeae4 : SUCCESS in 1h 08m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4e7287d80844d7abb544854cf6b5939 : SUCCESS in 1h 23m 05s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4404f8fb2c914233ba879f63d9539f2e : SUCCESS in 1h 23m 50s","accounts_in_message":[],"_revision_number":4},{"id":"040b78f6dbd12aa6af0e0880bd57c1225330ee72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 08:32:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"cc780b018f67e4d6389ef58a045a27c371fd24ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 08:33:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b63464eba0644dd391219c9cd0e37733\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72cc0aac15ca4d56858ab2df809aa00c : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"848e979791e37ae9fb4a6892e7510c9ec0167b79","revisions":{"11389031ba4d1c35c382947ae9b44fd2d38509b1":{"kind":"REWORK","_number":1,"created":"2022-06-27 13:06:34.000000000","uploader":{"_account_id":819,"name":"Adrien Cunin","email":"adrien@adriencunin.fr","username":"adri2000"},"ref":"refs/changes/82/847782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/82/847782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/1"}}},"commit":{"parents":[{"commit":"661d1379d954791c2af8f78199f8d2b17201ac72","subject":"Use unix socket while granting access for the backup service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/661d1379d954791c2af8f78199f8d2b17201ac72"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-17 13:37:59.000000000","tz":-240},"committer":{"name":"Adrien Cunin","email":"adrien.cunin@springernature.com","date":"2022-06-27 13:03:33.000000000","tz":120},"subject":"Bump mariadb version.","message":"Bump mariadb version.\n\nThis bump will fix:\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27376\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27377\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27380\n\nRelated-Bug: #1979728\n\nChange-Id: I90286b01b5578df04447155c69c92f823388185a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/11389031ba4d1c35c382947ae9b44fd2d38509b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/11389031ba4d1c35c382947ae9b44fd2d38509b1"}]},"branch":"refs/heads/stable/wallaby"},"b181a15eb27058d851eed40c2dcb6e3cba79012f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-06-27 17:49:29.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/82/847782/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/82/847782/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/2"}}},"commit":{"parents":[{"commit":"661d1379d954791c2af8f78199f8d2b17201ac72","subject":"Use unix socket while granting access for the backup service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/661d1379d954791c2af8f78199f8d2b17201ac72"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-17 13:37:59.000000000","tz":-240},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-06-27 17:49:29.000000000","tz":0},"subject":"Bump mariadb version.","message":"Bump mariadb version.\n\nThis bump will fix:\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27376\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27377\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27380\n\nRelated-Bug: #1979728\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/847668\nChange-Id: I90286b01b5578df04447155c69c92f823388185a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b181a15eb27058d851eed40c2dcb6e3cba79012f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b181a15eb27058d851eed40c2dcb6e3cba79012f"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"},"52e3f3ecb2b54bef24dba2dc4c7c497edc700229":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-06-27 17:49:41.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/82/847782/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/82/847782/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/3"}}},"commit":{"parents":[{"commit":"661d1379d954791c2af8f78199f8d2b17201ac72","subject":"Use unix socket while granting access for the backup service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/661d1379d954791c2af8f78199f8d2b17201ac72"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-17 13:37:59.000000000","tz":-240},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-06-27 17:49:41.000000000","tz":0},"subject":"Bump mariadb version.","message":"Bump mariadb version.\n\nThis bump will fix:\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27376\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27377\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27380\n\nRelated-Bug: #1979728\n\nChange-Id: I90286b01b5578df04447155c69c92f823388185a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/52e3f3ecb2b54bef24dba2dc4c7c497edc700229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/52e3f3ecb2b54bef24dba2dc4c7c497edc700229"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"},"848e979791e37ae9fb4a6892e7510c9ec0167b79":{"kind":"REWORK","_number":4,"created":"2022-07-13 13:12:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/82/847782/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/82/847782/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/82/847782/4"}}},"commit":{"parents":[{"commit":"661d1379d954791c2af8f78199f8d2b17201ac72","subject":"Use unix socket while granting access for the backup service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/661d1379d954791c2af8f78199f8d2b17201ac72"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-05-17 13:37:59.000000000","tz":-240},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-07-13 13:12:03.000000000","tz":120},"subject":"Bump mariadb version.","message":"Bump mariadb version.\n\nThis bump will fix:\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27376\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27377\nhttps://nvd.nist.gov/vuln/detail/CVE-2022-27380\n\nWe also modify workaround applied for 10.5 upgrade wrt to bug [1]\nas it has been added extra tools to help with checking state of upgrade.\n\nNew flag --check-if-upgrade-is-needed is checking if any upgrade is\nalready running and waits until it\u0027s finished.\nIt exits with rc 0 if upgrade is required and 1 if not.\n\nIf upgrade is required, we fall into rescue and perform upgrade.\n\n[1] https://jira.mariadb.org/browse/MDEV-27068\n\nRelated-Bug: #1979728\n\nChange-Id: I90286b01b5578df04447155c69c92f823388185a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/848e979791e37ae9fb4a6892e7510c9ec0167b79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/848e979791e37ae9fb4a6892e7510c9ec0167b79"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
