)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~858555","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I4181691ba3b23c5195b3cee3699637ece94187db","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2022-10-07 11:40:21.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2022-10-07 10:07:10.000000000","reason":"\u003cGERRIT_ACCOUNT_32666\u003e replied on the change","reason_account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-10-07 11:40:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4181691ba3b23c5195b3cee3699637ece94187db","subject":"Bump mariadb version to 10.6.10","status":"MERGED","created":"2022-09-20 16:30:18.000000000","updated":"2022-10-07 11:41:27.000000000","submitted":"2022-10-07 11:40:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"858555","meta_rev_id":"b70b41bd7443473de56f0c85b95a8ab99426b1cf","_number":858555,"virtual_id_number":858555,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-07 11:40:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2022-10-07 10:07:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-07 09:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2022-10-07 10:07:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2022-10-06 08:35:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 18:26:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-20 19:22:06.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2022-09-27 19:53:03.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-10-05 09:36:42.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"CC"},{"updated":"2022-10-07 10:07:10.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"3031898e8508d37fd569085cf695d192e2951e82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-20 16:30:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f7997f73218a89e9966499a4dfffffa1e35e35c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 18:26:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/613f1fad6a7e4abeb65eda0b2dbbc2f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67daa2e9f18a454786ba30a1fdb1e80b : SUCCESS in 5m 04s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f864b2e285145448e6c15f61e9c7248 : SUCCESS in 17m 52s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4eb66867f0646a3a4ba26456c9d8eef : SUCCESS in 1h 18m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e74525095b204954872ccb2777e34dc0 : SUCCESS in 1h 06m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8ec9424350b46a29b440dcff20b62be : SUCCESS in 1h 10m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e43f3dbccb0c47f0875b796a4266d624 : SUCCESS in 1h 01m 51s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab44b79ecccd4afa90e8653ee3657c16 : SUCCESS in 1h 47m 43s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/058c53cbf03f4fb1922a02827f99d5f0 : SUCCESS in 1h 48m 27s","accounts_in_message":[],"_revision_number":1},{"id":"27c3d60d53682efec99692c887df259259a51a4f","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-09-20 19:22:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19be2ad85e8dbbe9fe5eaed453399726a3e250ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-21 07:46:47.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e7a097e409b58444ddd4c774bd8c32b50975ea94","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-21 08:00:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8dfd445dad8e9395fd9832218c29d64f4eaeaa4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 09:23:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d829a2c9dcb4e8781ef2b68146fa3f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c1b222ef2464a5c925701e524ead5ff : SUCCESS in 5m 06s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06d8a2c400714931b2addf7eb295b9ec : SUCCESS in 18m 08s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f336b1211f7412cb72cebdccbe8a3f7 : SUCCESS in 1h 20m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/df28c68663384b75a6b51c04e2193bdb : SUCCESS in 56m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c77a849bb1624528824884b1f4bb6d48 : SUCCESS in 59m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d83c6efe8aaf4075b7b33b18c3fa59c6 : SUCCESS in 1h 02m 15s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/91f8eb17092b4169a6d5c7495805cdbb : SUCCESS in 1h 28m 01s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef11423708d14f10ae452cc6ba07242b : SUCCESS in 1h 17m 23s","accounts_in_message":[],"_revision_number":2},{"id":"520956c09a90d6082b5e568930d29056eff63927","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-09-27 19:53:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"96877ca257deb28e950fc755eb74943e33443deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-30 18:29:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f677b30fb5e6408b95896d52e087293265ba9175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 20:13:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5aee233e55b4ef7affae577c969ff64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b004a3890de442e875708aecffa9788 : SUCCESS in 6m 24s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfdc6a833cb3405583cb6b46b18c2723 : SUCCESS in 16m 13s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac778c70785f455591566987129c3584 : FAILURE in 44m 03s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/04e0ea5898d44979bdb3d18ad07f15a1 : SUCCESS in 56m 53s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/887132b8ba90448da97d8f5e3295ad9f : SUCCESS in 1h 13m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea0ab2faba5049969b8a2ae619e1d182 : SUCCESS in 1h 12m 08s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/437f819f67be4a1581fa19701c8145f0 : FAILURE in 1h 01m 03s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e83aa5a34264b8ea0a8cdc4fb092d2e : SUCCESS in 1h 40m 37s","accounts_in_message":[],"_revision_number":3},{"id":"b34add04743b646bfab078c2e940638ef26e69d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-03 12:41:55.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"96ae0d78bd5d86f1efaef6b3980d253164a792d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 14:15:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/639196fc338d40bd9b9a78ebe2360135\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/695180bffb944bba859baf5a19522222 : SUCCESS in 5m 52s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d72f8fb8b824642ab23b9cba8021f0e : SUCCESS in 15m 31s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ec8a7bd29f44fb0bea49d33e7c9f833 : FAILURE in 1h 08m 30s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ec5f663d79d949a0a79303c71a57a360 : SUCCESS in 1h 17m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d3f40e0069854e66b161485d51775b36 : SUCCESS in 1h 06m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57c7cf8632de4118bfdfeb0dc2098089 : SUCCESS in 58m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e52cd220ecc450498a73b99c6ab1302 : SUCCESS in 1h 09m 49s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/02441ebdf78041618fb8fd71dcc56364 : FAILURE in 1h 23m 14s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78bc96a77f6f4066b17610af006347ee : SUCCESS in 1h 14m 07s","accounts_in_message":[],"_revision_number":4},{"id":"d2b62e90774c116db69b31d2d08cb736832c693e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-04 12:18:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"983c0cb196d6b3704632f0ed715f7aa03e217299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 13:52:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eab5b0ae33c944b9b3b4eb8ee621eaeb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e99e89d1c134196ba7b552f32978d73 : SUCCESS in 6m 26s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4b93643378d4bc09c7d3340e2323806 : SUCCESS in 15m 36s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1e255c8efd44057a358003a29fbaa94 : SUCCESS in 1h 22m 06s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/97a0d1df1c15490cac8fb01d543547f1 : SUCCESS in 1h 18m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/02e12ac14dc140ffabad3739050fa76a : SUCCESS in 1h 11m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72d1d6e37fb7452487910932b084b4eb : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f9d340c80fe45cd8215812379da4e9d : SUCCESS in 56m 19s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd2aed1b1f3f43ef8f904bc5a8027000 : FAILURE in 1h 29m 52s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57bb6ddd41024a82bee8ea5a29f2ffc1 : SUCCESS in 1h 22m 01s","accounts_in_message":[],"_revision_number":4},{"id":"dda96dfd2a8725df235c08cd1de20714fb3e625c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-04 15:14:23.000000000","message":"Patch Set 4: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"70fbf6e33ecb94f7dd08f680cfed8657633528f3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-10-04 15:18:48.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5d404651217fd3c817242ac10efb5e16d79dd029","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-10-05 09:36:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"40bbadc67bb1cb6b7cfac4cb6ecda950f97a8426","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-10-05 10:55:10.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2b4bf5154099e44c0f74fd32f88cda42514e6b70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-05 11:26:00.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"7c20f1f5f82faa52edf15ccc12ac91115ba8f118","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-05 11:26:34.000000000","message":"Patch Set 5: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0aa87f842de8356be071a6da7b68c2ec84d103cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 13:07:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/58bed6a1e4724abc8c9ebc899e855905\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6920f75312594e8b8badfe32462e7ec2 : SUCCESS in 5m 49s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5788a90054d45a88bffba1e31ef9933 : SUCCESS in 14m 40s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/646e993466494b469d0d3d24baffd21f : FAILURE in 44m 52s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2d21dd41465c45e5a5e6ab11560bc229 : FAILURE in 40m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c93ae2fe3f114b88bb6bd76141911c41 : SUCCESS in 1h 06m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb2ac9ffb9d948a689cef469c01a94bd : SUCCESS in 1h 11m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d26d619da43c43c2b2b40b2d1a0ddbeb : SUCCESS in 1h 10m 48s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3372b59c6bd049faa48cceaff0b96d44 : FAILURE in 1h 20m 10s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f25847f17f345eb9c84b66afc5ca8db : SUCCESS in 1h 35m 21s","accounts_in_message":[],"_revision_number":5},{"id":"55e019e4b3a97f18338f9eea981e0a3af9eeb0a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-06 08:35:43.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"711abbfdcb5476d061e64882ee5439eaa3c50843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 10:27:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ae7a969e69e494db7d503e4ba505d1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b465277543e438883d7eb971cc742a9 : SUCCESS in 4m 32s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b0ad07346c547c0adc98049f60aedf7 : SUCCESS in 14m 45s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a66ea1196924fb28c77c07ad5940049 : SUCCESS in 1h 20m 41s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/133826fb05c0460a89da774f4c279785 : SUCCESS in 1h 19m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/496a6ecc92cb469ba78f3d1fa010ba7a : SUCCESS in 1h 02m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddb40d426ef14b7f87724de94ed39f4c : SUCCESS in 1h 02m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82d2efa924b64c4eb7f8c3c03016a078 : SUCCESS in 1h 08m 06s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b8dd1a583b640b9ba8fdc5f8da1e244 : FAILURE in 1h 33m 35s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28c61a898f784de3bfb99f2e0d41758d : SUCCESS in 1h 24m 20s","accounts_in_message":[],"_revision_number":6},{"id":"f8eb0fb8bbcc42e9c175aaf8ccb82e987f57316f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-10-07 09:32:15.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"fa72a39eba1649b3891e9a411395a4256804aa7d","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-10-07 10:07:10.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e6de0916bf326131a41a364c7b9565eb8d6ae4ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 10:09:23.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"6d0d914e4f67644f0cc62ba43359eb6087c126c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 11:40:21.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/847650b4c11d49239ee555a92a0a6748\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45efbacf8c184bb8a6c16e1d23ef7faf : SUCCESS in 4m 48s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/930c9789ddf94034928f8760b416374e : SUCCESS in 10m 11s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/cedc4f6a812943749fc5896ee5bd76d7 : SUCCESS in 1h 10m 19s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f21fef679b42488eadd33505bd65e8b9 : SUCCESS in 1h 13m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a1559f7ab1204d909f93a4ac63f62e34 : SUCCESS in 59m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9513968f659e48818bfde69eed625060 : SUCCESS in 1h 01m 39s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dec4dca8d96466fa5a481e592b7d2aa : SUCCESS in 1h 30m 50s","accounts_in_message":[],"_revision_number":6},{"id":"7aa8dcc4a0b9c68766a8acd6dac609fecedf7eef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 11:40:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"b70b41bd7443473de56f0c85b95a8ab99426b1cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 11:41:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5173835adb504bf5b0803fe96abd00fd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba88da0aa5f945d39163a4209926c7b8 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e04aeacc58c196c4fb3d49116cb93fa74f7fba31","revisions":{"ab0e7be8cadb53ae1734bd81b9349860bcbcc523":{"kind":"REWORK","_number":1,"created":"2022-09-20 16:30:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/858555/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/55/858555/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/1"}}},"commit":{"parents":[{"commit":"5200b50cf650fb5ad5e0733b9e0ead207dbf6c6a","subject":"Add the ability to specify custom additional galera users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5200b50cf650fb5ad5e0733b9e0ead207dbf6c6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-20 16:30:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-20 16:30:10.000000000","tz":120},"subject":"Bump mariadb version to 10.6.9","message":"Bump mariadb version to 10.6.9\n\nWith that we also able to use MariaDB provided repo for Ubuntu 22.04\n\nChange-Id: I4181691ba3b23c5195b3cee3699637ece94187db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ab0e7be8cadb53ae1734bd81b9349860bcbcc523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ab0e7be8cadb53ae1734bd81b9349860bcbcc523"}]},"branch":"refs/heads/master"},"098fe15f7f0e69b76b809e7a6debdda3d7064442":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-21 07:46:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/858555/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/55/858555/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/2"}}},"commit":{"parents":[{"commit":"5200b50cf650fb5ad5e0733b9e0ead207dbf6c6a","subject":"Add the ability to specify custom additional galera users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5200b50cf650fb5ad5e0733b9e0ead207dbf6c6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-20 16:30:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-09-21 07:46:46.000000000","tz":0},"subject":"Bump mariadb version to 10.6.10","message":"Bump mariadb version to 10.6.10\n\nWith that we also able to use MariaDB provided repo for Ubuntu 22.04\n\nChange-Id: I4181691ba3b23c5195b3cee3699637ece94187db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/098fe15f7f0e69b76b809e7a6debdda3d7064442"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/098fe15f7f0e69b76b809e7a6debdda3d7064442"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b84873b3317a914f3c2df73b7e0bdd396b80f5b3":{"kind":"REWORK","_number":3,"created":"2022-09-30 18:29:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/858555/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/55/858555/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/3"}}},"commit":{"parents":[{"commit":"5200b50cf650fb5ad5e0733b9e0ead207dbf6c6a","subject":"Add the ability to specify custom additional galera users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5200b50cf650fb5ad5e0733b9e0ead207dbf6c6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-20 16:30:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-30 18:29:10.000000000","tz":120},"subject":"Bump mariadb version to 10.6.10","message":"Bump mariadb version to 10.6.10\n\nWith that we also able to use MariaDB provided repo for\nUbuntu 22.04 and CentOS Stream/Rocky 9\n\nChange-Id: I4181691ba3b23c5195b3cee3699637ece94187db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b84873b3317a914f3c2df73b7e0bdd396b80f5b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b84873b3317a914f3c2df73b7e0bdd396b80f5b3"}]},"branch":"refs/heads/master"},"274bef5dfcd3f021e79708c8d8dd45d18b548941":{"kind":"REWORK","_number":4,"created":"2022-10-03 12:41:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/858555/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/55/858555/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/4"}}},"commit":{"parents":[{"commit":"5200b50cf650fb5ad5e0733b9e0ead207dbf6c6a","subject":"Add the ability to specify custom additional galera users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5200b50cf650fb5ad5e0733b9e0ead207dbf6c6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-20 16:30:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-03 12:41:55.000000000","tz":0},"subject":"Bump mariadb version to 10.6.10","message":"Bump mariadb version to 10.6.10\n\nWith that we also able to use MariaDB provided repo for\nUbuntu 22.04 and CentOS Stream/Rocky 9\n\nChange-Id: I4181691ba3b23c5195b3cee3699637ece94187db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/274bef5dfcd3f021e79708c8d8dd45d18b548941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/274bef5dfcd3f021e79708c8d8dd45d18b548941"}]},"branch":"refs/heads/master"},"7ed463964d8b67f1e87ecc71cac8cbb0e7707a20":{"kind":"REWORK","_number":5,"created":"2022-10-05 11:26:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/858555/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/55/858555/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/5"}}},"commit":{"parents":[{"commit":"5200b50cf650fb5ad5e0733b9e0ead207dbf6c6a","subject":"Add the ability to specify custom additional galera users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5200b50cf650fb5ad5e0733b9e0ead207dbf6c6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-20 16:30:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-05 11:26:00.000000000","tz":0},"subject":"Bump mariadb version to 10.6.10","message":"Bump mariadb version to 10.6.10\n\nWith that we also able to use MariaDB provided repo for\nUbuntu 22.04 and CentOS Stream/Rocky 9\n\nChange-Id: I4181691ba3b23c5195b3cee3699637ece94187db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7ed463964d8b67f1e87ecc71cac8cbb0e7707a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7ed463964d8b67f1e87ecc71cac8cbb0e7707a20"}]},"branch":"refs/heads/master"},"e04aeacc58c196c4fb3d49116cb93fa74f7fba31":{"kind":"REWORK","_number":6,"created":"2022-10-06 08:35:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/858555/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/55/858555/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/858555/6"}}},"commit":{"parents":[{"commit":"5200b50cf650fb5ad5e0733b9e0ead207dbf6c6a","subject":"Add the ability to specify custom additional galera users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5200b50cf650fb5ad5e0733b9e0ead207dbf6c6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-20 16:30:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-06 08:35:42.000000000","tz":0},"subject":"Bump mariadb version to 10.6.10","message":"Bump mariadb version to 10.6.10\n\nWith that we also able to use MariaDB provided repo for\nUbuntu 22.04 and CentOS Stream/Rocky 9\n\nChange-Id: I4181691ba3b23c5195b3cee3699637ece94187db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e04aeacc58c196c4fb3d49116cb93fa74f7fba31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e04aeacc58c196c4fb3d49116cb93fa74f7fba31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[]}
