)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~860655","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fyoga~I4181691ba3b23c5195b3cee3699637ece94187db","project":"openstack/openstack-ansible-galera_server","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-10-18 13:31:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4181691ba3b23c5195b3cee3699637ece94187db","subject":"Bump mariadb version to 10.6.10","status":"MERGED","created":"2022-10-07 11:54:40.000000000","updated":"2022-10-18 13:32:37.000000000","submitted":"2022-10-18 13:31:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860655","cherry_pick_of_change":858555,"cherry_pick_of_patch_set":6,"meta_rev_id":"b3bd84d58b7687bc8bf01fb55afb31f806b36867","_number":860655,"virtual_id_number":860655,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-18 13:31:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2022-10-17 17:33:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2022-10-14 08:00:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2022-10-18 10:58:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2022-10-17 17:33:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"CC":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-07 14:05:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-10 16:32:50.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2022-10-17 17:33:21.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2022-10-18 10:58:08.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"69cfa98d3333f12c3fb745b591dabf09738cd1b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-07 11:54:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a111e18f1dfeeda7ecdeb99c7564d6cbc58e3327","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-07 12:01:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"35b58bd2753a0d88fa377ef431f8a3258103d0de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 14:05:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b6470696f0e42ab8a922959c747aba3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c7a27184b3d4cc0a857055b30b51542 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0aceb669e0d4d44baab9db876571888 : SUCCESS in 4m 47s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62a1b80773604e6d84cfe0e6184caa4f : SUCCESS in 16m 26s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef9dbf73fb594aefbd07a2be886f55b5 : SUCCESS in 1h 26m 21s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7fc01697f1b4fd6a63fbecb853c745d : SUCCESS in 1h 19m 12s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1a36e90e9113409b84fe2b5c2d20020b : SUCCESS in 1h 10m 37s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e15e0789e9b94a9c9a262dedd11f0842 : SUCCESS in 1h 20m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cfc61dacf02450eb36778c49afe0e11 : FAILURE in 1h 05m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5a28f2f5b604f5fafb741501d4f994c : SUCCESS in 1h 04m 18s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a607e7b0eb44d89bd8beb0f51a6bcf5 : SUCCESS in 1h 46m 24s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a21edebfb3548d4bc50bcba5d455cd5 : SUCCESS in 1h 31m 56s","accounts_in_message":[],"_revision_number":2},{"id":"8ff9ddeb8513be2d28d5f15d155cc84626d400a3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-10-10 16:32:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f8a3eb2f6a165d1dccdb961c18f23e7ffaa200c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 18:18:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ed434a212e24fc7a53e31c2af26d095\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/316aaf6d5f7c46b5898bb64a47d7b3df : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a2af77e99194dcd8c0c2c5ccfc7df30 : SUCCESS in 3m 23s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8865e22e4a4643288ea28809eb35fe6f : SUCCESS in 12m 08s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e7b315fad89449f9654170e5d1f91fc : SUCCESS in 1h 02m 12s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bfe4ab425bc429f86e105206d7aaf43 : SUCCESS in 1h 18m 36s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/11e4737cc65940bf934aaf932d0df2a9 : SUCCESS in 1h 09m 22s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5ab4e34d0f6d40d5863d3b4202ca933a : SUCCESS in 1h 05m 40s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b8a4a25fa7434265bdf6a451e9269a5f : SUCCESS in 1h 17m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04bf63f248c4450ba2ef3ab2211ac357 : SUCCESS in 1h 09m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84db084a7e6f4c098e66d3eedfd99e8c : SUCCESS in 53m 20s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a82170c2d39d464da47fd8571c21f47d : SUCCESS in 1h 43m 13s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c50de000ce7a41d9886b2a1111b53bf8 : SUCCESS in 1h 19m 53s","accounts_in_message":[],"_revision_number":2},{"id":"56688c2830a118b9a108f961b341a0831844ce52","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-14 08:00:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5d1f298bc8e3cf50e3c340f56900bcfdab2412c6","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-10-17 17:33:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ff3cb4cccbe6d806521bd22accf7d45d590642bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 17:35:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5200566013b1e3fa8c1d497b8bf9de241d2f9e95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 19:16:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f0d2bb5cc094c31b5605583d5476670\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2947b86a4864decafb1c276d0631e88 : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e57bacc712dd4a89b54ef77f390401d5 : SUCCESS in 2m 47s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ed34efa452b437b9d09dcf9c3a931a8 : SUCCESS in 10m 06s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5a432917ded482cad58cd62c175a19e : SUCCESS in 1h 12m 38s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa39ae67ca714b86adc3f188924954a2 : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/267adf34fd8a44d6ab39c34dd7de71da : SUCCESS in 1h 11m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7323addde58f4572885c3598694dae19 : SUCCESS in 1h 06m 55s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b3543a5f82e64054bd43566b1e01160a : FAILURE in 59m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbf7870c3d3f491f8ae9bab5073e0e27 : SUCCESS in 53m 56s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39cd6b715c134018907f0913669c7a9c : SUCCESS in 1h 34m 32s","accounts_in_message":[],"_revision_number":2},{"id":"5ac99eef9e9a8c26bb1afde0594c9267207f8c94","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-10-17 20:24:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba9d223a0e5af9e57ee3e162b3ad599f966bd51a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 22:13:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f59cdef6ba94109bfbcb929b8291c62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/399b113d1585450497f875c601f7ffc9 : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a9c7b23bbb74da3a12c730c6ba310fe : SUCCESS in 2m 50s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2d406e066714f6787d70bad99b98859 : SUCCESS in 10m 02s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc298526fb2f4acbb5d1d5adee681514 : SUCCESS in 1h 16m 09s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/82a502b941f244aea64265ac0b30c1a2 : SUCCESS in 1h 11m 49s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ba413e43a5a740c09d0c2fb515bcb94a : SUCCESS in 1h 16m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b0e98d3dd44a48f498bb1a4c6f7d47be : SUCCESS in 1h 07m 34s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/078c03a465344d0aab6f3b0b4005f385 : SUCCESS in 1h 12m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52627fdfec4e43f9a1ec6f1af713ef84 : SUCCESS in 1h 04m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4436fc135dce4842951d95cd4fe71eee : SUCCESS in 1h 02m 21s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4592f2ca839e4252aff6825ff2e0d5bc : FAILURE in 1h 41m 09s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db7c67db12e14e9e94737fc790668d05 : SUCCESS in 1h 36m 16s","accounts_in_message":[],"_revision_number":2},{"id":"0c44274fe583f31037a22c231cfb3fb81fc33189","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 22:15:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"37ae8f2bd2004b5f576145d2029e486394a88bb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 00:02:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/959b728111ca48928933812f15d8f8ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793e6703576f459ca510c7ca3e7132a2 : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69a443e7c4954901b8c41631dda66616 : SUCCESS in 2m 45s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d84dc6c7683040528de1db334ef0b8a7 : SUCCESS in 9m 43s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/760d01eade50481381540559d335fced : FAILURE in 57m 40s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cfcc3c9d83341f1bc4836242b4bec9c : SUCCESS in 1h 24m 33s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d02bf15d133044aa950336bce31afd1b : SUCCESS in 1h 23m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f2c220d4ce7647e1ab2365d4a20f5581 : SUCCESS in 51m 54s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/80d1b6ad7af8407da537ad6e98e88538 : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a646823928b6480d937ebd8e53a7f1ca : SUCCESS in 57m 01s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/280bcda99c2a42bfbb433b5bb762666f : SUCCESS in 1h 46m 15s","accounts_in_message":[],"_revision_number":2},{"id":"47cf2a238028e0c90d1a2f5dad5bafc9c38e2b25","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 10:12:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d6b741747d2d001e108f4af2ba0ab2bdad9aeab","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-10-18 10:58:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ba51b543cf5bdcb6379ac1a76e30f911c2bc49a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 11:47:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90e995dbb6414bbc9b92c0c26f344c97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e6198967f340118e26c30a9c01c49a : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b377c29b04d48d6b4a5f8af6a97b3a1 : SUCCESS in 3m 02s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67aa843cbb894a368fc11b3fcabafbd0 : SUCCESS in 9m 54s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/08a748c064ce4c3ebaee2c021bcf4607 : SUCCESS in 1h 18m 38s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b25d5210bb27420f99c7903b68884d0c : SUCCESS in 1h 08m 25s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2d30e352b3f4f4d9f0d6da7dbaabf3e : SUCCESS in 1h 18m 39s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e1eb8d9678124426a31add36054cbbf9 : SUCCESS in 54m 34s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d13d00f44b8745afba8916dd924a5858 : SUCCESS in 1h 20m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a19deb293c42462cb5e4a8593b74e8f3 : SUCCESS in 1h 00m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bffca48006445d7941ad3141e511229 : SUCCESS in 1h 00m 40s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bff3031536bd457095961d3cc7f78fcd : SUCCESS in 1h 31m 32s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6739c8e37354b148c0e9103efa6ba3e : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":2},{"id":"23a431744969817b54ddb0d7ab7311749c94fb2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 11:49:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"90c0dcdb2e360d201eaae050e014db0dfa4112e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 13:31:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc02ba7bf12a49aca10bcdb55700631b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb32f4553fa84a0abde149fc63454fbb : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6235ece723804b559ffc4998b0d5d548 : SUCCESS in 3m 05s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70e510181e4940e8880dc78da272f964 : SUCCESS in 11m 24s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/3451b04ab76f43b7b47dcb0e5fbf499b : SUCCESS in 1h 35m 25s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b525b3cafcd84a8c88f4d3082c363fba : SUCCESS in 1h 18m 30s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4397801281de4018bc119bb5497b8816 : SUCCESS in 1h 22m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/105d3ccfb4844affa53c4c6100e28198 : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/9e80a6c334a9430ab9efc5bb39df6e79 : SUCCESS in 1h 22m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9fa025723724201b8e941c13962a80d : SUCCESS in 59m 03s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10d78f1e8dc646bca275d96b5b3272b7 : SUCCESS in 1h 26m 30s","accounts_in_message":[],"_revision_number":2},{"id":"45cb0b91876df638265a24a0e17258bb2690a012","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 13:31:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b3bd84d58b7687bc8bf01fb55afb31f806b36867","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 13:32:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd6065efc650474d89948f14336393f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8db93e0e5b234921a248d688874c8cf9 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44ff77548277333f5f23e4781e3f1fda613a3056","revisions":{"fa0b6d7fabc4e6f208947d781286a829305ab865":{"kind":"REWORK","_number":1,"created":"2022-10-07 11:54:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/860655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/55/860655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/860655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/860655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/860655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/860655/1"}}},"commit":{"parents":[{"commit":"af86aaa58d045f4ceebf9762cd7d1507374d4140","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/af86aaa58d045f4ceebf9762cd7d1507374d4140"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-20 16:30:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-07 11:54:40.000000000","tz":0},"subject":"Bump mariadb version to 10.6.10","message":"Bump mariadb version to 10.6.10\n\nWith that we also able to use MariaDB provided repo for\nUbuntu 22.04 and CentOS Stream/Rocky 9\n\nChange-Id: I4181691ba3b23c5195b3cee3699637ece94187db\n(cherry picked from commit e04aeacc58c196c4fb3d49116cb93fa74f7fba31)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fa0b6d7fabc4e6f208947d781286a829305ab865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fa0b6d7fabc4e6f208947d781286a829305ab865"}]},"branch":"refs/heads/stable/yoga"},"44ff77548277333f5f23e4781e3f1fda613a3056":{"kind":"REWORK","_number":2,"created":"2022-10-07 12:01:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/860655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/55/860655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/860655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/860655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/860655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/55/860655/2"}}},"commit":{"parents":[{"commit":"af86aaa58d045f4ceebf9762cd7d1507374d4140","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/af86aaa58d045f4ceebf9762cd7d1507374d4140"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-20 16:30:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-07 12:01:17.000000000","tz":120},"subject":"Bump mariadb version to 10.6.10","message":"Bump mariadb version to 10.6.10\n\nWith that we also able to use MariaDB provided repo for\nUbuntu 22.04 and CentOS Stream/Rocky 9\n\nChange-Id: I4181691ba3b23c5195b3cee3699637ece94187db\n(cherry picked from commit e04aeacc58c196c4fb3d49116cb93fa74f7fba31)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/44ff77548277333f5f23e4781e3f1fda613a3056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/44ff77548277333f5f23e4781e3f1fda613a3056"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
