)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~863515","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I002a60cf1091e05bf1b0b5d066df97795b5123fc","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"I002a60cf1091e05bf1b0b5d066df97795b5123fc","subject":"Remove mariadb-connector-c during upgrades","status":"ABANDONED","created":"2022-11-03 12:00:42.000000000","updated":"2023-02-27 10:40:11.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"78583c01e29fb5befc9e7176c64c8ccc2e592dc0","_number":863515,"virtual_id_number":863515,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-03 13:29:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":-1,"date":"2022-11-04 09:35:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-11-04 09:35:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-03 13:29:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb3819414ad029a2fc0e8670fb59045d01a2ff97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-03 12:00:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd985336f8b0a1c48f055f8ad89e7838ea68ff5f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-03 12:01:00.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"270b90ef010d574cea871c360fd86a6f3136872d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 13:29:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12161a42236c44aba45fa040686c1592\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1229ec53152e454eb99fc9162678ddef : SUCCESS in 5m 25s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d2f3fed13dc4b8890a898013ab40d2f : SUCCESS in 16m 40s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0fd5bd2e6fe4189b37d80e51ff54b72 : SUCCESS in 1h 24m 01s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4ab1b816a5fa42b593a276432e964404 : SUCCESS in 1h 24m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f5b8c00e7b0b40e1839d2f4895e18dbf : SUCCESS in 1h 17m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91f8d16244a14b2997708e1bc14b11bf : SUCCESS in 1h 15m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d61fc86fcfe444090498020f18b9af0 : SUCCESS in 1h 18m 34s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/adba125247174ef49d44c0535724cea3 : FAILURE in 1h 10m 10s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a1e860e3bba4d7d975bfc0336e4215f : SUCCESS in 1h 10m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9f0bad3c5983659da2962de0f0e5403a74f9bb61","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-04 09:31:55.000000000","message":"Patch Set 1: Backport-Candidate-1","accounts_in_message":[],"_revision_number":1},{"id":"89567788168029e59f5df3464b450856f1319b15","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-04 09:35:02.000000000","message":"Removed Backport-Candidate-1 by \u003cGERRIT_ACCOUNT_28619\u003e\n","accounts_in_message":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"_revision_number":1},{"id":"f1821967919119bb92a19c9bd804b2b2958667df","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-04 09:35:05.000000000","message":"Patch Set 1: Backport-Candidate+1 Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"78583c01e29fb5befc9e7176c64c8ccc2e592dc0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-27 10:40:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e64446c43230468bfbb895c3d2b4b6c3d69ef97d","revisions":{"e64446c43230468bfbb895c3d2b4b6c3d69ef97d":{"kind":"REWORK","_number":1,"created":"2022-11-03 12:00:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/863515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/15/863515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/863515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/863515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/863515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/863515/1"}}},"commit":{"parents":[{"commit":"7c00dbf3d2b929c1fefc3a841e812e3fbd9d63cc","subject":"Merge \"Bump mariadb version to 10.6.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7c00dbf3d2b929c1fefc3a841e812e3fbd9d63cc"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-11-03 12:00:34.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-11-03 12:00:34.000000000","tz":60},"subject":"Remove mariadb-connector-c during upgrades","message":"Remove mariadb-connector-c during upgrades\n\nmariadb-connector-c package is being installed with native mariadb,\nas CentOS 9 Stream support was added at times when MariaDB was not\nproviding repo for it\u0027s installation. Now, since we bumped version\nto 10.6.10 and install MariaDB for CentOS from MariaDB repo, upgrade\npath is broken, since mariadb-connector-c is conflicting with\nMariaDB-shared that provides same functionality. We need to add\nmariadb-connector-c to the packages that are removed during upgrade not\nto break upgrades.\n\nChange-Id: I002a60cf1091e05bf1b0b5d066df97795b5123fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e64446c43230468bfbb895c3d2b4b6c3d69ef97d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e64446c43230468bfbb895c3d2b4b6c3d69ef97d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
