)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~871527","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ia7d3d1cbfa3aea934d10262a8556952e58e82953","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-01-25 19:59:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia7d3d1cbfa3aea934d10262a8556952e58e82953","subject":"Do not forcefully restart socket","status":"MERGED","created":"2023-01-23 18:17:48.000000000","updated":"2023-01-25 20:00:35.000000000","submitted":"2023-01-25 19:59:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871527","meta_rev_id":"6fed3ae7a4d4998d49134d99065be3d9e60e9472","_number":871527,"virtual_id_number":871527,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-25 19:59:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-01-25 15:46:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-24 22:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-25 18:06:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-23 20:05:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-24 22:35:59.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-01-25 15:46:48.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"65ea60c965789fa13a1620c28f5bdd5611fd387b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-23 18:17:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01695996c0ad9a175fe5883c62fe0a9202626c54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 20:05:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b2ee87ea2d54f979bb8ac3f68c031c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c3606a8ab140bca9e6250038790a5c : SUCCESS in 4m 47s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28254d6b4d5945e99b0b24a9d2440a41 : SUCCESS in 12m 02s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/576e3f845b854c529ad958579bcf5994 : SUCCESS in 1h 28m 21s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/12d71d7f783545beb4b8ca2833c56d96 : SUCCESS in 1h 36m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/079f34e38b3e442aa6c30cb86dcc1a29 : SUCCESS in 1h 18m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8adacf0c51d344ee87db8ed29b426399 : SUCCESS in 1h 23m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e8087f1a219461d9e315fc4aa78fbb1 : SUCCESS in 1h 23m 00s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8359adcf2b384b48884f04a9a12ee9b7 : FAILURE in 1h 22m 15s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91398e1a34d34358927be2137c37ef3d : SUCCESS in 1h 37m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a97a0185ffc34efb36dcf55731973972e15593b7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-01-24 22:35:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"02e7d23fd9ab64bb931006c1e4697126351c316b","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-01-25 15:46:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cb549a9fc5d9bf4f7c3de88af11f9c9b3940c909","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-01-25 18:06:18.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0ea5841972f626ab078f63eddb6c9bc60dfd985","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 18:09:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea33abf33ad639bbf83101f1af6f37656844651f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 19:59:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16dfaef3129e4c0f8fdf57e9956fa80c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/963af27fd1344a8fbc5e1ec8f9fbd679 : SUCCESS in 4m 12s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5a8450c7300492792e92863ba6ecccd : SUCCESS in 12m 18s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a359c2c2b7b749238183ae0c1ca1ade8 : SUCCESS in 1h 27m 02s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2ba0335fa6054c5889259931633b5d3a : SUCCESS in 1h 45m 26s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/75fdf9778e164f11a2b2add05492c655 : SUCCESS in 1h 03m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d30aaac35fdc42faab4d7221b473a021 : SUCCESS in 1h 09m 56s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ae84ea2751f4278b10bd4c0153767e5 : SUCCESS in 1h 33m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3a6899f497aeaf59ffecacb715ddc593316cd282","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 19:59:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6fed3ae7a4d4998d49134d99065be3d9e60e9472","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 20:00:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d25082162d1a4403b9e7a02d36147306\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d7d51fd59240c2bfbb66421baa517f : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bfe6dffee072a14dd838f42df7b53adb9ff2e7a0","revisions":{"bfe6dffee072a14dd838f42df7b53adb9ff2e7a0":{"kind":"REWORK","_number":1,"created":"2023-01-23 18:17:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/27/871527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/27/871527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/27/871527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/27/871527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/27/871527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/27/871527/1"}}},"commit":{"parents":[{"commit":"1eb6f73fe654ddca04d4a0d288f8aa7286baf79d","subject":"Merge \"Remove \"warn\" parameter from command module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1eb6f73fe654ddca04d4a0d288f8aa7286baf79d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2023-01-23 18:17:39.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2023-01-23 18:17:39.000000000","tz":60},"subject":"Do not forcefully restart socket","message":"Do not forcefully restart socket\n\nWith state:restarted for socket it will be restarted on each playbook\nrun, even when it\u0027s not needed. Instead, we should restart socket\nonly when it\u0027s changed.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/871526\nChange-Id: Ia7d3d1cbfa3aea934d10262a8556952e58e82953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/bfe6dffee072a14dd838f42df7b53adb9ff2e7a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/bfe6dffee072a14dd838f42df7b53adb9ff2e7a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
