)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~874733","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fyoga~Ia4cde2153813e68419d261cd94e3017523177142","project":"openstack/openstack-ansible-galera_server","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-03-07 09:44:01.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ia4cde2153813e68419d261cd94e3017523177142","subject":"Allow maridbcheck socket to FreeBind","status":"MERGED","created":"2023-02-22 08:07:41.000000000","updated":"2023-03-07 13:19:51.000000000","submitted":"2023-03-07 13:18:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874733","cherry_pick_of_change":873334,"cherry_pick_of_patch_set":1,"meta_rev_id":"6da56be9144fec48aaad94519e21434fbd995691","_number":874733,"virtual_id_number":874733,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-07 13:18:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2023-02-28 15:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2023-02-28 14:28:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-28 12:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2023-02-28 15:43:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2023-02-28 14:28:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 10:04:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 16:58:59.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2023-02-28 12:31:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-02-28 15:43:24.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"01ed908a305f276b439ac753e3011088a29beabc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-22 08:07:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bfe43ddf4405338163710695d58eeceedd7df802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:04:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc16523889e1456faad3515c9bf81b1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e2fd415967542808673caa4098352e7 : SUCCESS in 4m 28s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d344d3081a834d409fb5c489f19166ec : SUCCESS in 11m 57s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a8017474a9b406dbf7c00ec2589e20b : SUCCESS in 1h 28m 09s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8800bc28140f4f618139d46690b707ce : SUCCESS in 1h 21m 21s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0e0f7c75c7dd499991229306c07968dc : RETRY_LIMIT in 8m 13s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fe83e8a49c0546c59958b3eeba6d02d7 : SUCCESS in 1h 03m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/c862fdc77f824f47b9ac89b31e574ace : SUCCESS in 1h 08m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5cf828151b44ba69d826e0a42328dd5 : SUCCESS in 52m 53s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/707eed964bf34f719628749c0569f8ec : SUCCESS in 56m 36s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/78ac310dbb9c4067bc1df9188b5ee8d6 : SUCCESS in 1h 52m 26s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dac8aacf4b24b2f9d21245edda10cb0 : SUCCESS in 1h 11m 33s","accounts_in_message":[],"_revision_number":1},{"id":"635f277cfa3ab113b05fdc3c9ea29544b59c31a1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-23 17:50:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79c0d07fcc44b3dd802b3f22ef1f9b949396045f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 20:11:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4dd0c4ab7794ac1a31a07fd98e41772\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4b4581d0085444e9ca41dd5533dd169 : SUCCESS in 5m 41s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2e0231736724d3e9c7fb63dc73ed6ea : SUCCESS in 10m 49s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/82f595c145504bf9956751e145c0bc0a : SUCCESS in 1h 24m 26s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/df35b71646a444278ae4638e82335401 : SUCCESS in 1h 18m 29s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/576f936f27944f34bc3b93cc704d8550 : SUCCESS in 1h 17m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ac98705883eb461aad8912d1ee868601 : SUCCESS in 1h 34m 42s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b74916b8cc0d485f8705a9ca79f6324f : FAILURE in 1h 10m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be8c6c0a37aa46b99a5faadacac9732f : SUCCESS in 1h 10m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a3797802a924efc99ec598486deb669 : SUCCESS in 1h 18m 05s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9cd31df6477e4472bfcd0d015019e201 : SUCCESS in 2h 11m 20s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/342b89a2e0674c86a9408294a93d30d9 : SUCCESS in 1h 37m 05s","accounts_in_message":[],"_revision_number":1},{"id":"00de7c5f12493006c26718a9ff748114dfa7ba33","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-27 10:32:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47041f7124570abeff07574ba22ae68982280f11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:44:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e47a5b7152794dfb93d178245c334b7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d24c03552f74bd89f08f287c2377606 : SUCCESS in 4m 17s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db0f146ac4c14f9ba64fa4a2b6d1298c : SUCCESS in 9m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/03f3f5660f874595b2336ff1e036829e : SUCCESS in 1h 12m 10s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b68d570d6bf472686f33e52decd7956 : SUCCESS in 1h 08m 05s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ebfcd8253ed34800b41d2ba2feabc8bf : SUCCESS in 1h 13m 44s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/2b2b43a12bdb45f8b1499256f872a014 : SUCCESS in 56m 31s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/498eda5dc01041babfce2bc8387e37fc : SUCCESS in 1h 17m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/268e82c2162541508b5e203a5f264911 : SUCCESS in 1h 04m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03f665bdbd074daebc4a60a522689a2c : SUCCESS in 1h 03m 25s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a455bb3d5de24752a9a25ebb6158d536 : SUCCESS in 1h 54m 23s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3abd349fbe148269f2776d143fa4ebc : FAILURE in 7m 47s","accounts_in_message":[],"_revision_number":1},{"id":"53c5a07435d66014b184da17aa34ff5fbe525263","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-27 13:39:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ff94387931a392e9080c4f0b341ce555a70e84f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 15:55:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81edb7a01bde4ef49960546acef77937\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e1426150df48658b4b673260528639 : SUCCESS in 5m 18s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/683ad63e509247d18c33063e25642c43 : SUCCESS in 11m 08s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4d0dd7a316c42b8959df0cfd26f6bb8 : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5df16d7001f543c7b0e6cdda69dbc37d : SUCCESS in 1h 16m 07s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/71021c9c67f44dec8ede2bae8aa058e7 : SUCCESS in 1h 10m 10s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7b4fd14a8dcb40bda18fe368265a94a1 : SUCCESS in 1h 08m 58s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/37ae391a52e14e3ab4e346c989dc6b6c : SUCCESS in 1h 25m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94b977a21ed9414090c3d7934baf9890 : SUCCESS in 1h 00m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdbe7b221e664489976c053d9fb44245 : SUCCESS in 1h 09m 59s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f3983151a3142ebaeb63b844c3cda63 : SUCCESS in 1h 54m 57s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/038090be5b0a463697a025956c187ce3 : FAILURE in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"dc3d65a4d3406fffe2d8576b0f87e409adcdecd3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-02-27 16:58:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d015ddef6ea43456af094e394b337bb1a0ca22c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-28 09:16:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cbb1d0bddadb700562a087c0e84ac0071de00f79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 11:23:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1ad5bedfc96407e87918543da0c63eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4afff818f9a0497fbf31c877c7f736a2 : SUCCESS in 4m 21s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d711cb03085842dabbd0cc6251e586b8 : SUCCESS in 12m 08s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a50ced63909b4d578431d7760d590a7f : SUCCESS in 1h 19m 54s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfce644a0efa4f5f94c4a20723d2da4d : SUCCESS in 1h 18m 56s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/de156dbded44423cabc28390c6e01199 : SUCCESS in 1h 25m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/433d0f28cc1d4f4f85fea7481f2828fa : SUCCESS in 1h 05m 43s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/7096dcdca934465d8001740c001fe8c0 : SUCCESS in 1h 18m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0509db079944fd9881a1fa65fd78255 : SUCCESS in 58m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f90e0e8816304ceea663b93ac5c7e24c : SUCCESS in 1h 02m 45s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/373988848eb34867b3770aa121a183bb : SUCCESS in 1h 45m 59s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa33f0d3a09e441d9ce45df2b83c0a50 : SUCCESS in 1h 25m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1d87e286c42e7bb61cc3ee4f8483e4cee5126136","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-02-28 12:31:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7b6b00307ee2a813714282edaed6aec22aca5426","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-28 14:28:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b0b8ebd46ebdb0e984bf791741404a2018baa43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 14:30:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34c66b2fd79c2cfb322df9158c009dee83841667","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2023-02-28 15:43:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b2f9c0e58a3607a8d6df64ba585a4b08eb202ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 16:46:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87b98b2306864da4b60139d2ecd6a021\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b15dc607cb40ec90378f2cbde64df5 : SUCCESS in 4m 15s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d28c7a465121401c9e7d69999240b923 : SUCCESS in 12m 41s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd09107f37a4422f9d081a3bd36a0954 : FAILURE in 49m 13s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2d358e8300c43229888dcf35a78e718 : FAILURE in 1h 02m 43s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5d333382d1a948cb977c2ba95367b086 : FAILURE in 44m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/38c3e084807b418cb97daa33515351b8 : SUCCESS in 1h 02m 48s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/f3dcd806e23848d09ad4b7e305f37996 : FAILURE in 47m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dacdf10a9c1b45dc991542bcecfde328 : SUCCESS in 59m 37s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f61b43f40844299a62343a5522c6e18 : SUCCESS in 2h 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"66bf3079b98fbe22795640d8bf1825a6b8fd331b","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2023-03-01 15:16:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d86dcbd2372931ef7a7caee6f97e32d4fd2baff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 16:53:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a439bac518c0472c876c1d5974dbf658\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74425471319849e1ab271b8b29bb3c88 : SUCCESS in 5m 25s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ab36240b2634dfe860fb9ad7ad0376a : SUCCESS in 12m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/30ce4d94bf5242cfb3da11d5d664aeaf : FAILURE in 50m 17s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbacf47fa28546439fa56e0741b76b3e : FAILURE in 47m 52s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b654ae69bd5042359afab641bd9d391c : FAILURE in 48m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1dab9a2156cd472a8f86e02ce66a14ba : SUCCESS in 1h 00m 28s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/27d22d07c4cb41a5b85e8c13fe095c57 : FAILURE in 45m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d84fa4c1653a4ca8a127d52b344ec515 : SUCCESS in 1h 27m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55f5d74cac094ccdb764383118fc3824 : SUCCESS in 1h 29m 58s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a971c014a99c462580295805321c607d : FAILURE in 44m 27s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40c0a7bc8090401ab75416cdeaecf04c : SUCCESS in 1h 31m 17s","accounts_in_message":[],"_revision_number":1},{"id":"2fef11d3987f3221fdde464b2430166fb3a3de3c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-03-07 09:44:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"328c067b48e45f30848cb26de8e369ec78fc5e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 11:43:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21903d88cffd450c99290fd3e8fe11cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5245190775c443a99f823675c51fdc50 : SUCCESS in 4m 29s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6c3e2ea8f714a0c8efe9a54d00d55e3 : SUCCESS in 10m 29s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/98621c410bee496c8561e5fabba51fd1 : SUCCESS in 1h 09m 56s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a5c8b97ad8d4006baafe070936786fc : SUCCESS in 1h 19m 08s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/536fa599a6c846edb5fc114d7938b135 : SUCCESS in 1h 12m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/43cd9c0a5c5a43ae9560dec7167e771a : SUCCESS in 1h 05m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/f66ca153d3d24db8b9aafb0da96febe0 : SUCCESS in 1h 06m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a3dbd8d58ed46dcae77466aee38bf7e : SUCCESS in 58m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5a99d03602c44599ffc9eac47ef47e1 : SUCCESS in 1h 00m 03s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/73def76bde9c4477b6afddaa7c489d18 : SUCCESS in 1h 37m 04s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81761e4d0a7a4cdfae3b91bc495efac3 : SUCCESS in 1h 24m 16s","accounts_in_message":[],"_revision_number":1},{"id":"68521a6ae02446377d19f42ab0d9afa6fbd51c9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 11:45:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6035767a68bae254acd9e9a944d48ff9224b5cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 13:18:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0463db6b6334abe9c2a9e18bb226a5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7cfdbc53acb4c57b81158142180bff2 : SUCCESS in 4m 43s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea3becc6abfb403f90049ef553a29496 : SUCCESS in 10m 45s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/33c910cbeea14e318a8fae2ba4081a01 : SUCCESS in 1h 05m 43s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/72ba55a6d0764287a4a22ef6e3fccf63 : SUCCESS in 1h 23m 04s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6ce8bfef266748fba19b2c94795b4dd4 : SUCCESS in 1h 03m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/45c6224464254feaab76aadc2b324756 : SUCCESS in 1h 06m 06s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/bae14b2159f34d0b8c9a6645c724ccb8 : SUCCESS in 1h 05m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbfb8c21ba9e430b91699a89f7d7fa3c : SUCCESS in 53m 38s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b408bfe3c9e4f5cb4b275054e4ba639 : SUCCESS in 1h 17m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c8854492354fe4adb373ce74c481062c0e3aea43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 13:18:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6da56be9144fec48aaad94519e21434fbd995691","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 13:19:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d9ef7ebfc1b4208b57e87cd937f4260\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ec097d95d924435a8ddbc145bbe88e8 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4acaf657873452e0720a1b3f5ba2f889ab88d96e","revisions":{"4acaf657873452e0720a1b3f5ba2f889ab88d96e":{"kind":"REWORK","_number":1,"created":"2023-02-22 08:07:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/33/874733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/33/874733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/33/874733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/33/874733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/33/874733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/33/874733/1"}}},"commit":{"parents":[{"commit":"bedc860b2fe68c3dc5e084750693c91d218f6982","subject":"Prevent mariadbcheck.socket to wait for network.target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/bedc860b2fe68c3dc5e084750693c91d218f6982"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-09 21:19:36.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-02-22 08:07:41.000000000","tz":0},"subject":"Allow maridbcheck socket to FreeBind","message":"Allow maridbcheck socket to FreeBind\n\nOnce we\u0027ve removed network.target from wanted targets for\nmariadbcheck.socket, it started to fail to startup intermitently in LXC\ndeployments, since it was trying to bind on IP address that is not\nbrought up yet. At the same time we can\u0027t wait for IP being up, as\nOVS while providing network, waits for socket.target as it needs\nto have ovsdb started up, so waiting for network.target does\ncreate circular dependency.\n\nTo avoid that we\u0027re allowing socket to bind on IP even when IP is not\nUP yet. Other possible solution would be to bind on 0.0.0.0.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/872896\nChange-Id: Ia4cde2153813e68419d261cd94e3017523177142\nCloses-Bug: #2003631\nRelated-Bug: #2002653\n(cherry picked from commit 8a8d29ea490fba6695e3356831846466f6991089)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4acaf657873452e0720a1b3f5ba2f889ab88d96e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4acaf657873452e0720a1b3f5ba2f889ab88d96e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
