)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~875381","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I430acf61fd4fdacdead19d0c5cc2765e017eb3c7","project":"openstack/openstack-ansible-galera_server","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-04-04 19:53:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I430acf61fd4fdacdead19d0c5cc2765e017eb3c7","subject":"Upgrade MariaDB to 10.11","status":"MERGED","created":"2023-02-27 10:46:08.000000000","updated":"2023-04-04 19:53:56.000000000","submitted":"2023-04-04 19:53:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875381","meta_rev_id":"219a259acb82f124b8df775469ba2cba4ffb41cc","_number":875381,"virtual_id_number":875381,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-04 19:53:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2023-04-04 18:15:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-04 18:00:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2023-04-04 18:15:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 12:27:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-04 18:00:03.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-04-04 18:15:22.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"a675e6a27a8e27d91742e1148528af728792d83d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-27 10:46:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d3caefbb7201799b9bc5ba63bd453446c0ae01c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:27:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14f678d81ecf4a03869baf495c447d76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/420afb675f6f496a923870e708a3990b : SUCCESS in 8m 32s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f185d5a71314c06a626b69e2fa1112f : SUCCESS in 16m 23s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/963c27642320425ba0ed77d1a8dfabbe : FAILURE in 45m 40s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9bf159957214b27b2fe9d79f00eece5 : FAILURE in 44m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/9a17ce84e8694222a50a8e6b811d4ebb : FAILURE in 33m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd01191ed3014baf89e1e11264fb5c22 : FAILURE in 29m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac5c4d816d1f4ad6a4cef7ffb62dcb45 : FAILURE in 29m 52s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e70e38f8f8ae4450b224607eefc280c6 : FAILURE in 1h 20m 49s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31fee818cf0743f1a85259b02a066b6b : FAILURE in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9812744ccca740304407c43dfefabb980cdf8d9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-03-31 10:49:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"81c36ac4c3bf9e439e68bf54af50145a7b18d08d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-03-31 11:05:54.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f749569dfcfec2ac2854699cb5db1db7f3eab4b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 13:02:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b012d50529c94964a751af79ac89e1cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f49dd6a9edc49ffadb2a1d6d457f6c4 : SUCCESS in 4m 28s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f49832383d047ec9ea02bd3c1e17fd2 : SUCCESS in 9m 49s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a04d39c77d874893ba0151bc0d5909cc : FAILURE in 38m 44s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a31307622e914b8e8f5996c77974451c : SUCCESS in 1h 39m 47s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/01c039a737174d93a872cc952dc8bdec : FAILURE in 47m 19s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bce48908d00d4faba601bd6f7f6405bf : FAILURE in 57m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72e134f1d4c6464292cf4fe1f1fff004 : FAILURE in 47m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d8fc602ec8f4b1196195ada45e465e2 : FAILURE in 55m 18s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb21f10597a547dd836b3b1e751da4f4 : FAILURE in 32m 49s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03cba4c50d2f4bf7a143f364692eb832 : FAILURE in 1h 11m 55s","accounts_in_message":[],"_revision_number":3},{"id":"b480e9d53fafd12cdf2a96ec6411cc6eefe296b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-03-31 17:21:35.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5ef95c4d8e26d05dfb4a0ddb4ae55de07426b3ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-03-31 17:22:15.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"4adc27213e42874a847d9b0fc54b5a7b71ef62ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 18:48:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cefae26305e2459aad3dcde63df33888\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f0d4487b54a49c7b20af2f4a9b21522 : SUCCESS in 4m 26s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63dad6cb484c4e4387e700d19b0ba195 : SUCCESS in 9m 22s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/bccc76028e1440ddae5acddbc5784322 : FAILURE in 37m 50s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/626442570e8f4d75b03056235564cf55 : SUCCESS in 1h 21m 54s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5e6a5c4db6d8493382d89def306854d2 : SUCCESS in 1h 02m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc4bdb3da5274c04afb9898731776782 : SUCCESS in 1h 08m 50s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0df3d4226d78463b9fe4679fcda96c5c : SUCCESS in 1h 17m 43s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8342844b2eb4c6db8094173d47c893b : SUCCESS in 1h 10m 46s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5da36ec203e44635b033be4764b88c5e : FAILURE in 32m 53s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eac2bad29864920801e62bf46ff78a0 : SUCCESS in 1h 18m 16s","accounts_in_message":[],"_revision_number":5},{"id":"664bc5943118d1c1d55fa6a2c881fd902c9d494c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-04 18:00:03.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"5b267e23e655bbbbf0b1a70569425e87bbf3022f","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2023-04-04 18:15:22.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3e028fb8c20cf9576720e84956a5d879164389c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 18:17:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4d535f0f6f418dcf380582c551bddd189367edb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 19:53:02.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0d88af5e38b4bed9984735fe2836b14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9562cc0baadb44219100424bd1937ba6 : SUCCESS in 4m 46s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38bd01dd76404e7fb72b5266fa8f9a2f : SUCCESS in 12m 06s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c045e0b45224c719da77879465b5f18 : SUCCESS in 1h 29m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e5d40501ed1b4797be109263d6bbd5a0 : SUCCESS in 1h 03m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/496be170e1774da38ae9cbe7b855b695 : SUCCESS in 1h 13m 38s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5311b0ad84e40adb0f9bd7e1e5de80d : SUCCESS in 1h 19m 14s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69c7396477c440efb0337d4d471fdab8 : SUCCESS in 1h 09m 59s","accounts_in_message":[],"_revision_number":5},{"id":"f4943a68ae23993562d8deb87459d55b03187d0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 19:53:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"219a259acb82f124b8df775469ba2cba4ffb41cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 19:53:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecacb6ad5189439b9f5f994ba08f9065\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/132a04166b904653a62ab8ee17833f57 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"515bec4165b211ae7f5b7cfc3cfcd10b793d358d","revisions":{"73d1538a2e9df8b0e7b9dc1d3e7e7e97ff5914df":{"kind":"REWORK","_number":1,"created":"2023-02-27 10:46:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/875381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/81/875381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/1"}}},"commit":{"parents":[{"commit":"8a8d29ea490fba6695e3356831846466f6991089","subject":"Allow maridbcheck socket to FreeBind","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8a8d29ea490fba6695e3356831846466f6991089"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-27 10:46:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-27 10:46:00.000000000","tz":60},"subject":"Upgrade MariaDB to 10.11","message":"Upgrade MariaDB to 10.11\n\n10.11 is the next LTS release of MariaDB which has been released\nrecently. Let\u0027s switch to using new LTS from 10.6 that we\u0027re using\nfor quite a while now.\n\nChange-Id: I430acf61fd4fdacdead19d0c5cc2765e017eb3c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/73d1538a2e9df8b0e7b9dc1d3e7e7e97ff5914df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/73d1538a2e9df8b0e7b9dc1d3e7e7e97ff5914df"}]},"branch":"refs/heads/master"},"c2aca5717fcd7e9368c54a7076cadb5af4df8e69":{"kind":"REWORK","_number":2,"created":"2023-03-31 10:49:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/875381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/81/875381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/2"}}},"commit":{"parents":[{"commit":"8a8d29ea490fba6695e3356831846466f6991089","subject":"Allow maridbcheck socket to FreeBind","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8a8d29ea490fba6695e3356831846466f6991089"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-27 10:46:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-03-31 10:49:21.000000000","tz":120},"subject":"Upgrade MariaDB to 10.11","message":"Upgrade MariaDB to 10.11\n\n10.11 is the next LTS release of MariaDB which has been released\nrecently. Let\u0027s switch to using new LTS from 10.6 that we\u0027re using\nfor quite a while now.\n\nChange-Id: I430acf61fd4fdacdead19d0c5cc2765e017eb3c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c2aca5717fcd7e9368c54a7076cadb5af4df8e69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c2aca5717fcd7e9368c54a7076cadb5af4df8e69"}]},"branch":"refs/heads/master"},"6df7fd4aff5520f1b3dfce2d929d6b75adf590db":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-31 11:05:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/875381/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/81/875381/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/3"}}},"commit":{"parents":[{"commit":"cc7deb1f0e8306482aaa323f5d11f12f449757e6","subject":"Update MariaDB GPG keys for RPM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cc7deb1f0e8306482aaa323f5d11f12f449757e6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-27 10:46:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-03-31 11:05:54.000000000","tz":0},"subject":"Upgrade MariaDB to 10.11","message":"Upgrade MariaDB to 10.11\n\n10.11 is the next LTS release of MariaDB which has been released\nrecently. Let\u0027s switch to using new LTS from 10.6 that we\u0027re using\nfor quite a while now.\n\nChange-Id: I430acf61fd4fdacdead19d0c5cc2765e017eb3c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6df7fd4aff5520f1b3dfce2d929d6b75adf590db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6df7fd4aff5520f1b3dfce2d929d6b75adf590db"}]},"branch":"refs/heads/master","description":"Rebase"},"e984394e9c89f6034ae907680f0b0fdfa0c05e5f":{"kind":"REWORK","_number":4,"created":"2023-03-31 17:21:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/875381/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/81/875381/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/4"}}},"commit":{"parents":[{"commit":"cc7deb1f0e8306482aaa323f5d11f12f449757e6","subject":"Update MariaDB GPG keys for RPM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cc7deb1f0e8306482aaa323f5d11f12f449757e6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-27 10:46:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-03-31 17:21:35.000000000","tz":0},"subject":"Upgrade MariaDB to 10.11","message":"Upgrade MariaDB to 10.11\n\n10.11 is the next LTS release of MariaDB which has been released\nrecently. Let\u0027s switch to using new LTS from 10.6 that we\u0027re using\nfor quite a while now.\n\nChange-Id: I430acf61fd4fdacdead19d0c5cc2765e017eb3c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e984394e9c89f6034ae907680f0b0fdfa0c05e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e984394e9c89f6034ae907680f0b0fdfa0c05e5f"}]},"branch":"refs/heads/master"},"515bec4165b211ae7f5b7cfc3cfcd10b793d358d":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-03-31 17:22:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/875381/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/81/875381/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/81/875381/5"}}},"commit":{"parents":[{"commit":"cc7deb1f0e8306482aaa323f5d11f12f449757e6","subject":"Update MariaDB GPG keys for RPM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cc7deb1f0e8306482aaa323f5d11f12f449757e6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-27 10:46:00.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-03-31 17:22:15.000000000","tz":0},"subject":"Upgrade MariaDB to 10.11","message":"Upgrade MariaDB to 10.11\n\n10.11 is the next LTS release of MariaDB which has been released\nrecently. Let\u0027s switch to using new LTS from 10.6 that we\u0027re using\nfor quite a while now.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879069\nChange-Id: I430acf61fd4fdacdead19d0c5cc2765e017eb3c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/515bec4165b211ae7f5b7cfc3cfcd10b793d358d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/515bec4165b211ae7f5b7cfc3cfcd10b793d358d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
