)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~886180","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I28c6a0e0b41d4d29c3e79e601de45ea373dee4fb","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"mariabackup_compression","attention_set":{},"removed_from_attention_set":{"34150":{"account":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"last_update":"2023-06-27 17:34:41.000000000","reason":"Change was submitted"},"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2023-06-27 17:34:41.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-06-20 13:38:24.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-06-21 10:04:33.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I28c6a0e0b41d4d29c3e79e601de45ea373dee4fb","subject":"Add optional compression to mariabackup","status":"MERGED","created":"2023-06-15 14:21:34.000000000","updated":"2023-06-27 17:35:41.000000000","submitted":"2023-06-27 17:34:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886180-mariabackup_compression","meta_rev_id":"e8362968d45b44ecb94ec15fbb40e571ea3e7e7f","_number":886180,"virtual_id_number":886180,"owner":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-27 17:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-06-20 13:38:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-21 10:04:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-21 10:04:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 16:23:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-16 07:29:03.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-06-16 07:38:04.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-06-16 07:38:04.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-06-20 14:51:57.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"8f6477ca3685b68de3d613ab3587c3212070ed4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"date":"2023-06-15 14:21:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e7c77cbeb12812d4caa7e84d95a5a627edb1e7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 16:23:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efc0b7752a7844f8acb1a27a32fd90ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/572b1d4a54eb43f59d0f4259c7fd2ac6 : SUCCESS in 4m 45s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bbb2f11b4fc43e98b34f5e8f8e12c69 : SUCCESS in 11m 57s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/74e82acc0fa848d8bf4b33d4c6d50f8f : FAILURE in 42m 20s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/18a3e82d8b4a41f8a9f3cd27c63fcd2e : SUCCESS in 1h 31m 55s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cabb861d367e4fc7a8c9e7acae70b9c2 : SUCCESS in 1h 40m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/573bec5e77b7412c8a6010671f17f3bb : SUCCESS in 1h 24m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f20b81a9e12247459d89b98208e2cbdd : SUCCESS in 1h 16m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c87f8e976374aeb9eac2b078eaac974 : SUCCESS in 1h 13m 42s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac0c0a26ba6d4dde92c41c0585077136 : SUCCESS in 1h 15m 33s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/274cc67cb081400ea4216f74a236c789 : SUCCESS in 1h 20m 59s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/291d12ae65764ad6b901bc5e18fa7e0a : SUCCESS in 1h 55m 23s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee9b9d350441425cb56345aa00b338a6 : SUCCESS in 1h 46m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ecb69c6e19e2adf04445b39a645e72163c6b8323","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-06-19 14:27:24.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0eeee801f966055e9e5f7a256ab966688b0a5d83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"date":"2023-06-20 09:03:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e92970a9d96da0ce30cf61e7c78b21e4c717d953","author":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"date":"2023-06-20 09:05:11.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"86c35a0a365cd4116b35939b921fc5478fcfcc89","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-06-20 10:00:43.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e9437817120e59e694a300e9cb37b33e642847a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 11:17:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61462b83433649abb78142a1a980dcc1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9945cc5ffba74f5c8ceedd1d34c28640 : SUCCESS in 4m 33s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43072d55bdcf4092b741cae4f5e9b9a4 : SUCCESS in 12m 06s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a703c72f1e349698684dae7fafeef6f : FAILURE in 40m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/854d3261e1934fe09997c24580ce788d : SUCCESS in 1h 51m 02s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cdf8c7c79ccb4fd5bab0b14b7f425f57 : SUCCESS in 1h 22m 06s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e7d23faa8ebc450d95609421ade77ec6 : SUCCESS in 1h 04m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a730ab0308a142baa59b36557b6f267e : SUCCESS in 1h 06m 53s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/121039bdfe3e44ce922f61ae049bed53 : SUCCESS in 1h 10m 11s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4abd223a55eb4f7ea25bcd6fc6bd3d63 : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/729da441dfb54f35a2b0e488a36c487d : SUCCESS in 1h 26m 12s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ded0564aabc14cc1805229af558deb8a : SUCCESS in 2h 06m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/517f81676a964d5997e8f6109ba3f977 : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4d4bc5cedb82c0d855f6ae7a9cef70a2dbe9585d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-06-20 12:14:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"868467de09f035a88cc2f755d2a9c333cdd96b46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"date":"2023-06-20 13:35:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"da2c7a8973f39738d5136943d1024ba9aecb95a0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-06-20 13:38:24.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"10402d7189d27ddc3f476768ff92f3fd41b91e6b","author":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"date":"2023-06-20 13:46:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ca8627d0f5485958a839293b5b7187d5ebe9396c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 15:31:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/759dd52172d046bbba60f29c95237292\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df3d170310848198e523b8d5e8be1d7 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb8321da5e2d4ae299150db6e04a414f : SUCCESS in 3m 48s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc301bc6ec50402182930c8b1d26d166 : SUCCESS in 23m 22s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/70b0c3f0be4e4a5a9b61ef44db7f22e8 : FAILURE in 44m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2cab29889e064f08b439286ce6c13ad9 : SUCCESS in 1h 26m 45s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cc7a4a5e094d4f1fb27b38161a2aa72b : SUCCESS in 1h 33m 45s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5d91ecad0737421a950d145903cb33e0 : SUCCESS in 1h 12m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa17a9ead9294709810772beb96c76d8 : SUCCESS in 1h 15m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be6b61d1eea84cf4ae63b0bbbf285b56 : SUCCESS in 1h 34m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83431391bbb444a1b159f11394f84bf7 : SUCCESS in 1h 23m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80aa00d843af4a2987905fbabd98104b : SUCCESS in 1h 14m 00s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c528783bf5b546a8935be19a0dbaa4f9 : SUCCESS in 1h 52m 21s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02c485c4c3984155b2c3010f2d4a8159 : SUCCESS in 1h 41m 18s","accounts_in_message":[],"_revision_number":3},{"id":"244bcb9e7fb85deebe7dbf6d8f5a4bccc4eb624d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-06-21 10:04:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"05da51e630f30ed900c353818f0d32615aaa3c91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 10:06:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ad7e82e4554b95928efc1a6e00c02a4a149f6200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 12:40:50.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/594ba316b70e40bd8b85e4f9dafb223d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa75d6178124a7693bcda87650492a0 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca4f388a82a641a5aa5aab8cf45952f6 : SUCCESS in 3m 01s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7eb34afa38c8465d9cc9f7a2342675a3 : SUCCESS in 11m 47s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2d1ab777656748169693b78d014a7c40 : FAILURE in 1h 41m 40s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/da2ec6a348bc4653bbac9b832742baaf : SUCCESS in 1h 26m 06s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/349d869f3d3b48eeaeea7fec214f2c9c : SUCCESS in 1h 23m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4bbc1b7ff934da8b6a88b8fd98f7876 : SUCCESS in 1h 14m 10s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d71a5f1d5ce4c82978bc1d2c0226521 : SUCCESS in 1h 30m 03s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/436435e000ab40668c6750c88c633c91 : SUCCESS in 1h 12m 39s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d3cd795918824e479b8b95959be7fbfd : SUCCESS in 2h 24m 44s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcfe566b34024994850f00ab77ca6eba : SUCCESS in 1h 53m 03s","accounts_in_message":[],"_revision_number":3},{"id":"bd87459bf543ebd13b9019f5dc198d066a7c5fc1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-06-26 16:03:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5ef9fab316ae056974b9135b1e6ac0fef24f85ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 19:15:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb5e9c91fb4b4c1e8f6f4250785ee854\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c80d8ed2ea043ed83e263b88b927310 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd8a68818c6242ecace106bf12d68e8f : SUCCESS in 3m 14s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96be81cc564a432d9e4385f3e1f15953 : SUCCESS in 12m 11s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c69628f04c744c0f97d4ace532e09ba2 : FAILURE in 43m 47s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c3a44e71f89e48bda31a4e7fb036882a : SUCCESS in 1h 39m 49s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8d62b0326dfe480d9386e70513a6dbb9 : SUCCESS in 1h 21m 51s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/db77b2de3de3485fb313c27d6c3f70b0 : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15a5d250ad46417786b20f9ca0c17f52 : SUCCESS in 1h 19m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37ebad5b255d40f98dc31df7b098b752 : SUCCESS in 1h 17m 25s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49533d8b6d7b450aba5f2eb2a6c11bb0 : SUCCESS in 1h 22m 55s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61b3be6e15fb40cc92e10bdab21bfbc1 : SUCCESS in 1h 27m 01s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9598d676645e493f949b9baed46de0df : TIMED_OUT in 3h 09m 46s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2260ad8f7fbc478ebece7d10048d7491 : SUCCESS in 1h 46m 53s","accounts_in_message":[],"_revision_number":3},{"id":"238290bb72f73c0b1ac4452ebca40b592541100b","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-06-27 12:47:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dcc585278e94144b7529f60808c96c45f421d3a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 15:03:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a244887777cf4031b1e1bdfacb3a27b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/098df6c36b9b49fdaddaf7cc463beb53 : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65b2ef4a69de4c238873337981a2ee8d : SUCCESS in 3m 15s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/695e1b7d55f64143af32e42269f07671 : SUCCESS in 12m 59s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/86980f9c7af34e7e9a7eb62a98cabf3d : FAILURE in 43m 48s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5bf275bd39124358af31e32a5409fb73 : SUCCESS in 2h 04m 10s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/be988f434364443db62962351711fc07 : SUCCESS in 1h 42m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3a49eabce34b4e2da0817523df3b3f98 : SUCCESS in 1h 40m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e79f66b4c7c4c5c8530e80d288729f4 : SUCCESS in 1h 10m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca3439765fd54430a4b93c95229ec59b : SUCCESS in 1h 25m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9457069d9534f36b3a270a600afa615 : SUCCESS in 1h 16m 27s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81e84d3a9a3a4809bdb890144f6614b5 : SUCCESS in 1h 42m 37s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4781cf7e705c40e4bc5d8e50d118da79 : SUCCESS in 2h 12m 57s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afc21cb11d9d474b832829e2b16aee17 : SUCCESS in 1h 36m 26s","accounts_in_message":[],"_revision_number":3},{"id":"2d4e12545672aa9b2b5584a40ac2455ce20d6ee5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 15:05:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"efcc8a46c88c5d2a45804ad5a6b0ea1aca47f1c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 17:34:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f2e80e57b1b49a9b6ab70a39bc72ccc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c533935f2720487a8a8b5ebcae679eee : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed23ada1a004487ba16ea23a18db33b0 : SUCCESS in 3m 55s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65fda7dae9c547ba93323a6a349922b5 : SUCCESS in 12m 01s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4f2a8b12eab84cd79b8d73d7bb0a36a5 : SUCCESS in 1h 33m 15s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1dde27e810f1400290c08bdc789ab841 : SUCCESS in 1h 57m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/822b4d70a7ce494eacd57259c3a27798 : SUCCESS in 1h 05m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c44e38859a704354b90bab8ffb4a3375 : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36394ebc70274735871368c428fb822e : SUCCESS in 1h 19m 35s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7995fb042b2f408b90e5a5e41d64bb81 : SUCCESS in 1h 17m 30s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4114c52d6b6549e7b1f186d70a64ed2a : SUCCESS in 2h 20m 36s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcc2a89db4b14f35a890cc56962443c0 : SUCCESS in 1h 35m 35s","accounts_in_message":[],"_revision_number":3},{"id":"9c0eb45d17b7c918d2e2e8c22501faaebbb847a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 17:34:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e8362968d45b44ecb94ec15fbb40e571ea3e7e7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 17:35:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3a12b1e101d4329a1bad22576456c1c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7492aee2fd584e85a552b4346c3ca386 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"60009ed7cebe9c082592fd564b1577068ef94b6c","revisions":{"afa7049f789f823ca20a845d4a044c1b618c7533":{"kind":"REWORK","_number":1,"created":"2023-06-15 14:21:34.000000000","uploader":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"ref":"refs/changes/80/886180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/80/886180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/886180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/886180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/886180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/886180/1"}}},"commit":{"parents":[{"commit":"92b5711b94d734d77665875bb99deb09cda72b25","subject":"Define backup randomized delay in defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/92b5711b94d734d77665875bb99deb09cda72b25"}]}],"author":{"name":"Simon Hensel","email":"simon.hensel@inovex.de","date":"2023-06-15 13:47:49.000000000","tz":120},"committer":{"name":"Simon Hensel","email":"simon.hensel@inovex.de","date":"2023-06-15 14:19:49.000000000","tz":120},"subject":"Add optional compression to mariabackup","message":"Add optional compression to mariabackup\n\nAs database backups can grow substantially in size, compressing backups\nhelps to preserve disk space.\nWhile the mariabackup utility offers no compression by itself, we can\nstream the backup into a compression tool to create an archive [1].\nThe xtrabackup_checkpoints file, which contains metadata on a backup,\ngets stored alongside the archive, allowing to create incremental\nbackups from non-compressed backups and vice-versa [2].\nOne thing to note, is that compressed backups cannot be prepared in\nadvance, this step must be manually carried out by the user.\nBackup compression is disabled by default and different compressors\ncan be chosen (zstd, xz, ...), with gzip being the default.\n\n[1] https://mariadb.com/kb/en/using-encryption-and-compression-tools-with-mariabackup/\n[2] https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/#combining-with-stream-output\n\nChange-Id: I28c6a0e0b41d4d29c3e79e601de45ea373dee4fb\nSigned-off-by: Simon Hensel \u003csimon.hensel@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/afa7049f789f823ca20a845d4a044c1b618c7533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/afa7049f789f823ca20a845d4a044c1b618c7533"}]},"branch":"refs/heads/master"},"8a372ff1b1b097c56461326e878fc8e31a9809af":{"kind":"REWORK","_number":2,"created":"2023-06-20 09:03:18.000000000","uploader":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"ref":"refs/changes/80/886180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/80/886180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/886180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/886180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/886180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/886180/2"}}},"commit":{"parents":[{"commit":"92b5711b94d734d77665875bb99deb09cda72b25","subject":"Define backup randomized delay in defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/92b5711b94d734d77665875bb99deb09cda72b25"}]}],"author":{"name":"Simon Hensel","email":"simon.hensel@inovex.de","date":"2023-06-15 13:47:49.000000000","tz":120},"committer":{"name":"Simon Hensel","email":"simon.hensel@inovex.de","date":"2023-06-20 09:03:04.000000000","tz":120},"subject":"Add optional compression to mariabackup","message":"Add optional compression to mariabackup\n\nAs database backups can grow substantially in size, compressing backups\nhelps to preserve disk space.\nWhile the mariabackup utility offers no compression by itself, we can\nstream the backup into a compression tool to create an archive [1].\nThe xtrabackup_checkpoints file, which contains metadata on a backup,\ngets stored alongside the archive, allowing to create incremental\nbackups from non-compressed backups and vice-versa [2].\nOne thing to note, is that compressed backups cannot be prepared in\nadvance, this step must be manually carried out by the user.\nBackup compression is disabled by default and different compressors\ncan be chosen (zstd, xz, ...), with gzip being the default.\n\n[1] https://mariadb.com/kb/en/using-encryption-and-compression-tools-with-mariabackup/\n[2] https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/#combining-with-stream-output\n\nChange-Id: I28c6a0e0b41d4d29c3e79e601de45ea373dee4fb\nSigned-off-by: Simon Hensel \u003csimon.hensel@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8a372ff1b1b097c56461326e878fc8e31a9809af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8a372ff1b1b097c56461326e878fc8e31a9809af"}]},"branch":"refs/heads/master"},"60009ed7cebe9c082592fd564b1577068ef94b6c":{"kind":"REWORK","_number":3,"created":"2023-06-20 13:35:22.000000000","uploader":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"ref":"refs/changes/80/886180/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/80/886180/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/886180/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/886180/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/886180/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/886180/3"}}},"commit":{"parents":[{"commit":"92b5711b94d734d77665875bb99deb09cda72b25","subject":"Define backup randomized delay in defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/92b5711b94d734d77665875bb99deb09cda72b25"}]}],"author":{"name":"Simon Hensel","email":"simon.hensel@inovex.de","date":"2023-06-15 13:47:49.000000000","tz":120},"committer":{"name":"Simon Hensel","email":"simon.hensel@inovex.de","date":"2023-06-20 13:34:39.000000000","tz":120},"subject":"Add optional compression to mariabackup","message":"Add optional compression to mariabackup\n\nAs database backups can grow substantially in size, compressing backups\nhelps to preserve disk space.\nWhile the mariabackup utility offers no compression by itself, we can\nstream the backup into a compression tool to create an archive [1].\nThe xtrabackup_checkpoints file, which contains metadata on a backup,\ngets stored alongside the archive, allowing to create incremental\nbackups from non-compressed backups and vice-versa [2].\nOne thing to note, is that compressed backups cannot be prepared in\nadvance, this step must be manually carried out by the user.\nBackup compression is disabled by default and different compressors\ncan be chosen (zstd, xz, ...), with gzip being the default.\n\n[1] https://mariadb.com/kb/en/using-encryption-and-compression-tools-with-mariabackup/\n[2] https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/#combining-with-stream-output\n\nChange-Id: I28c6a0e0b41d4d29c3e79e601de45ea373dee4fb\nSigned-off-by: Simon Hensel \u003csimon.hensel@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/60009ed7cebe9c082592fd564b1577068ef94b6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/60009ed7cebe9c082592fd564b1577068ef94b6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
