)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~887143","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fzed~I28c6a0e0b41d4d29c3e79e601de45ea373dee4fb","project":"openstack/openstack-ansible-galera_server","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"32755":{"account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"last_update":"2023-08-31 10:46:40.000000000","reason":"Change was submitted"},"34150":{"account":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"last_update":"2023-07-14 06:51:48.000000000","reason":"\u003cGERRIT_ACCOUNT_32755\u003e replied on the change","reason_account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-30 15:10:43.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"34411":{"account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"last_update":"2023-08-30 12:52:54.000000000","reason":"\u003cGERRIT_ACCOUNT_34411\u003e replied on the change","reason_account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-06-29 13:46:53.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I28c6a0e0b41d4d29c3e79e601de45ea373dee4fb","subject":"Add optional compression to mariabackup","status":"MERGED","created":"2023-06-28 07:58:22.000000000","updated":"2023-08-31 10:47:41.000000000","submitted":"2023-08-31 10:46:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"887143","meta_rev_id":"8ddf16d23931b354cd6d5c85e40e40d7bfcaa0e0","_number":887143,"virtual_id_number":887143,"owner":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-31 10:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-08-09 06:24:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2023-08-30 12:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2023-08-30 12:52:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-28 07:58:22.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"state":"REVIEWER"},{"updated":"2023-06-28 08:01:33.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-06-28 08:01:33.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-06-28 09:45:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 08:29:40.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"b6ab77dc5920d7b490edfbb8d347ad09e4042b7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-06-28 07:58:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03785d382c70666622b1fc87aeb424595ffb5542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 09:45:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/28a281a4d01346c1b87721449bfdb703\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6b9e39754fb454ba0c11dcb2e8e1ba6 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b94a6f03199f441d81f9a37c71713147 : SUCCESS in 3m 00s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7738131b4d2b4855a6bbe07335bd0477 : SUCCESS in 10m 12s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f41cdc172cca4cc7b3ffc2c093b7578f : FAILURE in 36m 20s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/605b2a638dbb438e9e524e5fcdbe7000 : FAILURE in 1h 23m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/38341d232d5c4dc492baa4f81d26462c : FAILURE in 1h 07m 21s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4df8f03e857140cf85db2b208870e397 : FAILURE in 1h 10m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/403f8dfc1186415181937a99879e24b4 : FAILURE in 1h 06m 49s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fb68c9dfae3c4f0cb4ac86fd7779a073 : FAILURE in 1h 40m 08s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bb58fb0dd1a4c618c672989544f4b7e : FAILURE in 1h 27m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ff9abb8af57ff5946ffaa5af9ddd1a95aa78cf49","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-06-28 10:03:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3871bd3f596e007f8aede26b8d2c483df5a5196d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 11:54:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4ef321a0fa542d3af7613f4175b1c20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b226db3ffaa408d979b6301883fab71 : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/285f04be957f4c41bc3ad07de17dda1e : SUCCESS in 3m 08s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a96095ed9f74ac98244a97e29b0efdd : SUCCESS in 11m 03s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4902513288e544ad8f7ec86e4fa1fd2d : FAILURE in 39m 00s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a93fba49a4e4cdd9506a2ccf14379cb : FAILURE in 1h 37m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f56fb630f29c4182962a38b845bd5296 : FAILURE in 1h 25m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/179cbb52c78d4910bf2848d31009af26 : FAILURE in 1h 13m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b83862f984a84c439aeb77f7685ffe13 : FAILURE in 1h 07m 28s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/319642161ca84835974c0fdedf902af3 : FAILURE in 1h 42m 13s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c42f5e4e68c449b58b542004df868caf : FAILURE in 1h 32m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1c112eb2119717daaba83ac76cdf67a500225f27","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-06-28 15:15:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d820c2511e00dd3f19a52d696066f5ecb038be28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 17:39:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75fe599821754401be479f253ed00674\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae3b1e2d71984a1fbc8ca73cdb65d808 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/028bc5b988b4448fb35ebf141bd73c86 : SUCCESS in 3m 46s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d2f41bdd6a94703b5ca9879ed3fb9ff : SUCCESS in 10m 17s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/01e893599a4642f7b89e9b43de9b3ef0 : FAILURE in 46m 20s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/addc45e4ae414a6985a811b3de83bcd0 : FAILURE in 1h 52m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b6d0a8bd77b44866b2c9d3d0c9701ec8 : FAILURE in 1h 20m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aea309bd317946b888a3a0371d6eda7b : FAILURE in 1h 08m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/507513ff5858483eb622b03676dd1523 : FAILURE in 1h 02m 02s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/53f952577b19404bb16214a729938533 : FAILURE in 2h 15m 52s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/165b592d4bee4994a642ed9cd80d9ecd : FAILURE in 1h 18m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4e583fe8bce4b57b997b0e5d78cf75a7f5fc38a2","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-06-29 13:37:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"190d042a91c75b99d39ce4ca2f0459cc9f742181","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-06-29 13:46:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23b8989edbd0e19ce383a073f1358961e499b4ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:04:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d62145ceea194b0c947729308d5ae402\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/551a5aa0e0b24d6fafd29076d4664f39 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cd3752c47524c9dac50635f2d716b29 : SUCCESS in 2m 58s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4274306ab1724b78ba12118307f84ce9 : SUCCESS in 12m 51s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/105da90ccc3c443ba118db309a2b724d : FAILURE in 39m 28s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/acb6c5994e274812a6fe1db76e1e3dc5 : FAILURE in 1h 19m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3ce4b97923f7443097efc9b7b3070458 : FAILURE in 1h 46m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4b7b2540f074cd283235eafce94c34c : FAILURE in 1h 51m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a06a8c0f9c94cce8b7cbf87827a8594 : FAILURE in 1h 50m 55s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e808741c95e74d24b7b358a116d42381 : FAILURE in 2h 16m 43s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95049198639a4076a6a55f001c6f3bbb : FAILURE in 1h 18m 46s","accounts_in_message":[],"_revision_number":1},{"id":"86583e649c9ce2ee1e34692f9c0dc7e7d47cd1f2","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-07-14 06:51:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6cfdffdd916dc34cc9ba5fc4ae89fe7d52614ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 08:36:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c860d1e39a944f7b29f6cd6070925b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5f4029d4a7e4bffb9cb2246210ea49c : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdeb86b195cb4f4fa86bccb8b0d1167b : SUCCESS in 3m 28s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e16cf0f1bdae4285b9a4bfef775fd5a0 : SUCCESS in 10m 41s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f91aea321d84632a48c3b2a17c8ab7c : FAILURE in 38m 52s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0e1bec8c80d84443871ac6bf9a56b8ff : FAILURE in 1h 19m 21s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/295f0eca37ef40ba84ff6718f6b00ce9 : FAILURE in 1h 00m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c0a1f361d20434eb1ead7de65af8ec4 : FAILURE in 1h 14m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9d75a6348004038961b3a950442334f : FAILURE in 1h 10m 11s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fb54d511315d46789de535b6e5ec4fe5 : FAILURE in 1h 37m 59s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/536cffcb3388432a95f45d5f0540a991 : FAILURE in 1h 22m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a989b0d4e1c0e2fd145771979aca13429e0a596d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-20 07:44:11.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2eef8a352c8238e3c20c6ab41f56083f2a0d518f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 09:42:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbfc794431dd4569832d046318cef939\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78bac47cfe3b4bbe8a73f9d287429560 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9044f129ee2408987734857844f0590 : SUCCESS in 4m 10s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5916325a3e634f299cf18eab78f12d21 : SUCCESS in 10m 04s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/52441093de6948eaa3f3b21c083f7cc6 : FAILURE in 40m 23s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/92e5b171464a40b49b7eb3e8dbd0c7d1 : SUCCESS in 1h 24m 53s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8215a6bc03f04ee39f590c75e139f5ec : SUCCESS in 1h 13m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b8178a229924115be7309e8eaa04239 : SUCCESS in 59m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c7b90d6995f4f039d7a46d6ff7f7563 : SUCCESS in 1h 10m 20s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/07563dfcd2344307b27b8dc21ce99bc0 : SUCCESS in 1h 47m 14s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb03ade84ca74905a52bced34242c14e : SUCCESS in 1h 27m 43s","accounts_in_message":[],"_revision_number":2},{"id":"8c732db8144b7331fd6e7d74c3a49f4c0daee523","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-09 06:24:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a874c2a43530fb63b82d8d613e05a749e5f182dc","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2023-08-30 12:52:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"33d5d9d7f81799831f100fa175f1d266231c9452","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 12:54:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a996f6be738998f838772f23076d9ca98b11b2b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 14:50:16.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ad7d97dfd6e439b8c9279e181230fd1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd5706e9dae4c6f9c6577c0d5b60343 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5012d4b29084f28b887609f194eacd3 : SUCCESS in 3m 26s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc2256be4ce749c2990628efa918c1fb : SUCCESS in 11m 39s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aa825fb9156844199f08ede7e4db3e39 : SUCCESS in 1h 22m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8f9329f38f3a41b29a8c03568b784d42 : FAILURE in 1h 03m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5e544f5d7b045c78f5e50ffc1836a5d : SUCCESS in 1h 16m 09s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0bebc6c7ee8e45a79d2c87c141a40e93 : SUCCESS in 1h 51m 19s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ef9fd3dcbbe4d589b98bd3d457295bb : SUCCESS in 1h 36m 37s","accounts_in_message":[],"_revision_number":2},{"id":"cbfb39c1b13952424397c8939eaf46e116721cfa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-30 15:10:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1fa67033d145e2e33e09f05913c33c92a4e76e56","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-31 06:24:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7c50bd3d9742a63d86848b3ebda4c4777fde2b1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-31 06:24:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"80af052b6e4a6887c5dfa82beaf304fd2a651e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 08:46:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aae8850bba5a48558ab40bf01d6cc9a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad4f986943c84e298c60f29e07853553 : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4b5ed3d54d94a9aa802b309a962db4e : SUCCESS in 3m 59s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfbc5b0c28c043659c5027b943185179 : SUCCESS in 11m 00s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe1b6072e45b4e15950a621143f2b178 : SUCCESS in 1h 19m 12s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/172f7c2daa394784be87eea4470f26bd : SUCCESS in 1h 22m 25s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f66de8d9322d4562a9bfe97a32789d97 : SUCCESS in 1h 05m 23s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3c8aa292fcc4b33831e603f1fd406a9 : SUCCESS in 1h 04m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6d94e3f8e8f4eed994e3f2bbdee268f : SUCCESS in 1h 05m 40s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0fe665baf36e489da41fb974cdd15195 : SUCCESS in 2h 15m 15s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6154f97d5c04b48b2513521dca0c449 : SUCCESS in 1h 25m 08s","accounts_in_message":[],"_revision_number":2},{"id":"54f7190a5f53915ee882a1b556856212ef2103d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 08:48:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18b648a63fa47c64fd83398041729cb352d5c93c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 10:46:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79bb529def3c473884ae678a170a278f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e690421beec4d96ad0fe072d6d832cf : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d629804f42c84bda8310577a900eb29f : SUCCESS in 3m 10s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/819d7466830a4f32a973cedb4ae171c9 : SUCCESS in 11m 18s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d8d2d6d7a87c41069d8cba72aded6b42 : SUCCESS in 1h 24m 31s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7ba52e3987b44699bb2f004b86ae54a8 : SUCCESS in 58m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e10392756de1499898c5c868243a561e : SUCCESS in 1h 08m 02s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3365ce45dfa1450cb9bdb5a2a3ab256a : SUCCESS in 1h 52m 29s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33cfd52a25474d309f3e1e5b67f29814 : SUCCESS in 1h 29m 50s","accounts_in_message":[],"_revision_number":2},{"id":"f607ac76c30ecf7ed335cf800be109cda3e8c44c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 10:46:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8ddf16d23931b354cd6d5c85e40e40d7bfcaa0e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 10:47:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23bc9ab4943f4953b6b70baa6a5c4529\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43022af64091471eaaae30d09129c605 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"03ad7d6ace52cfef9976315d918b8f58e94d1539","revisions":{"e21d80924c08aa0af809389bc068217d491b6080":{"kind":"REWORK","_number":1,"created":"2023-06-28 07:58:22.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/43/887143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/43/887143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/43/887143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/43/887143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/43/887143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/43/887143/1"}}},"commit":{"parents":[{"commit":"0f343a83156bcd058da32cb8c7fe00d3ef77dfcf","subject":"Define backup randomized delay in defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0f343a83156bcd058da32cb8c7fe00d3ef77dfcf"}]}],"author":{"name":"Simon Hensel","email":"simon.hensel@inovex.de","date":"2023-06-15 13:47:49.000000000","tz":120},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2023-06-28 07:58:16.000000000","tz":120},"subject":"Add optional compression to mariabackup","message":"Add optional compression to mariabackup\n\nAs database backups can grow substantially in size, compressing backups\nhelps to preserve disk space.\nWhile the mariabackup utility offers no compression by itself, we can\nstream the backup into a compression tool to create an archive [1].\nThe xtrabackup_checkpoints file, which contains metadata on a backup,\ngets stored alongside the archive, allowing to create incremental\nbackups from non-compressed backups and vice-versa [2].\nOne thing to note, is that compressed backups cannot be prepared in\nadvance, this step must be manually carried out by the user.\nBackup compression is disabled by default and different compressors\ncan be chosen (zstd, xz, ...), with gzip being the default.\n\n[1] https://mariadb.com/kb/en/using-encryption-and-compression-tools-with-mariabackup/\n[2] https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/#combining-with-stream-output\n\nChange-Id: I28c6a0e0b41d4d29c3e79e601de45ea373dee4fb\nSigned-off-by: Simon Hensel \u003csimon.hensel@inovex.de\u003e\n(cherry picked from commit 60009ed7cebe9c082592fd564b1577068ef94b6c)\n(cherry picked from commit 41801ab04d3c64d61b881d410691e05e0312114d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e21d80924c08aa0af809389bc068217d491b6080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e21d80924c08aa0af809389bc068217d491b6080"}]},"branch":"refs/heads/stable/zed"},"03ad7d6ace52cfef9976315d918b8f58e94d1539":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-07-20 07:44:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/43/887143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/43/887143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/43/887143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/43/887143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/43/887143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/43/887143/2"}}},"commit":{"parents":[{"commit":"0f343a83156bcd058da32cb8c7fe00d3ef77dfcf","subject":"Define backup randomized delay in defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0f343a83156bcd058da32cb8c7fe00d3ef77dfcf"}]}],"author":{"name":"Simon Hensel","email":"simon.hensel@inovex.de","date":"2023-06-15 13:47:49.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-07-20 07:44:11.000000000","tz":0},"subject":"Add optional compression to mariabackup","message":"Add optional compression to mariabackup\n\nAs database backups can grow substantially in size, compressing backups\nhelps to preserve disk space.\nWhile the mariabackup utility offers no compression by itself, we can\nstream the backup into a compression tool to create an archive [1].\nThe xtrabackup_checkpoints file, which contains metadata on a backup,\ngets stored alongside the archive, allowing to create incremental\nbackups from non-compressed backups and vice-versa [2].\nOne thing to note, is that compressed backups cannot be prepared in\nadvance, this step must be manually carried out by the user.\nBackup compression is disabled by default and different compressors\ncan be chosen (zstd, xz, ...), with gzip being the default.\n\n[1] https://mariadb.com/kb/en/using-encryption-and-compression-tools-with-mariabackup/\n[2] https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/#combining-with-stream-output\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/888437\n\nChange-Id: I28c6a0e0b41d4d29c3e79e601de45ea373dee4fb\nSigned-off-by: Simon Hensel \u003csimon.hensel@inovex.de\u003e\n(cherry picked from commit 60009ed7cebe9c082592fd564b1577068ef94b6c)\n(cherry picked from commit 41801ab04d3c64d61b881d410691e05e0312114d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/03ad7d6ace52cfef9976315d918b8f58e94d1539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/03ad7d6ace52cfef9976315d918b8f58e94d1539"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
