)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~887144","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fyoga~I28c6a0e0b41d4d29c3e79e601de45ea373dee4fb","project":"openstack/openstack-ansible-galera_server","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"32755":{"account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"last_update":"2023-10-06 19:44:54.000000000","reason":"Change was submitted"},"34150":{"account":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"last_update":"2023-10-06 19:44:54.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-09 06:24:38.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-10-06 19:44:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I28c6a0e0b41d4d29c3e79e601de45ea373dee4fb","subject":"Add optional compression to mariabackup","status":"MERGED","created":"2023-06-28 07:59:32.000000000","updated":"2023-10-06 19:46:04.000000000","submitted":"2023-10-06 19:44:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887144","meta_rev_id":"3f5f4bab1f02ee064e479aa381bc0dcf4f4d44d2","_number":887144,"virtual_id_number":887144,"owner":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-06 19:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":2,"date":"2023-08-30 02:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":2,"date":"2023-09-28 17:29:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":1,"date":"2023-09-28 17:29:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-28 07:59:32.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":34150,"name":"Simon Hensel","email":"simon.hensel@inovex.de","username":"shensel"},"state":"REVIEWER"},{"updated":"2023-06-28 08:01:28.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-06-28 08:01:28.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-06-28 09:27:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 02:07:53.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"89053f40f69088a4a39c4190a769e65943b3a1db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-06-28 07:59:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5784f63a7e5e442d84cd58d3aa7ec308e983fa28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 09:27:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de3d3d1df6ba41cba1488349dbd93c07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d380b2cefb498e9f8d5ee247ed33ba : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cddbcb7a0214afe8a3cf4a1f3067273 : SUCCESS in 3m 40s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e577eab031704300934c2f1a129b9089 : SUCCESS in 14m 20s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a51f93541415457496d45005f7fd5b62 : FAILURE in 1h 03m 49s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fc2c58bd84446bfa50d4e566de69c41 : FAILURE in 34m 04s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ef245a8fdc2c4227847857a3c86d94c5 : FAILURE in 1h 18m 05s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/df8039aa4b344d52b8321c58b966b7ca : FAILURE in 1h 13m 42s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/7e410a4c048b42d5bc93e6252d41e8a4 : FAILURE in 1h 18m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a5a495174ab41a98994c239e506cf4c : FAILURE in 1h 07m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0700cfc3ab74ce0ba00e63bfc92cc22 : FAILURE in 1h 05m 22s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ec6f446d99946cdbd1e0b286b6e1662 : FAILURE in 1h 05m 04s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81d24e377de143899da7cbc5adab59c8 : FAILURE in 45m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4587c81ddc6540c074db64951d0619369b78d2df","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-09 06:24:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"095320931b867ec31cb64add638a8c8198456692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 08:13:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/572abc19d14c4bd199343368bbf89d61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a6f6898b5854d15af85901bc2186d1e : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21ca4ac33def445e8c22912279e9ac17 : SUCCESS in 2m 58s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f099c059b6f94036b8ad4fd58587b95a : SUCCESS in 10m 15s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/22cc88ea16574278abf84a7213cda967 : SUCCESS in 1h 20m 39s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe05537361d34b96af9deb94b486ac67 : FAILURE in 34m 19s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d39559bcd78749009a3b907c937a19e3 : SUCCESS in 1h 22m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5f00153b45ed4aa082bbbc0e01334b89 : SUCCESS in 1h 00m 21s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/932e7a9c9ccf46408e487804c8a04be9 : SUCCESS in 1h 06m 27s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19cfdb41bcd644dfb680e46e2ea5b19d : SUCCESS in 1h 06m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04556d5328e24ca7b9bd4af6113959c9 : SUCCESS in 55m 57s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3c219d4d40f4c449eca6b978f713460 : FAILURE in 1h 03m 06s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6cc386ea20b4b89a873153e4f4ed5c8 : SUCCESS in 1h 39m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9aec943d315449e53d6aa06d16c2261c3685fbea","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2023-08-30 02:07:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"75729ec69eba933a912125205f5c6e975ccf0e2d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-28 17:29:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1904eef025181329083605dfdc0c3431bf2a8b61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 17:31:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"745dd49e67042783c1ba2e4a04034fa55d0fb58d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 19:09:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f95d0b17650444fa15b5278ba317875\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d77e700fb7ff4013af93484b02b680b4 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9848f04be473413586853d0feed2d68c : SUCCESS in 2m 53s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da6d173263924ad4b061a3170f030782 : SUCCESS in 14m 47s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a962506e4e9d4487bd2e574fd995a736 : SUCCESS in 1h 31m 10s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6ce04d3ad5ae4251ac12c4bda4f3dca7 : SUCCESS in 1h 32m 26s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/056fd1179f9f4e39ac6f6f77b0b5a00e : SUCCESS in 1h 04m 27s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4fd1ae4fd66e49b29f9a03fff34e1957 : FAILURE in 1h 12m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/337d57226af248bc811aecd472a62f4b : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":1},{"id":"cba7cd4bba342f055370684d15a6052a8d38b306","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-29 07:27:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b69d90b250da0a4789364d8e79d2113a790e1488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 09:06:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/598c8f84bb7f4e68bd15a250c72696f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dea34db8b4b422ebcfc5fe9415ce87c : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acaf4af7bacd4f3b9ca0bb4df0d93f03 : SUCCESS in 3m 58s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c869ef17c454316918bd58df90aea19 : SUCCESS in 10m 40s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d4c50bbc3cc4ca2a86e32e1be381b30 : SUCCESS in 1h 26m 23s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7d758ecaa3b442080480fb38cc0a3e2 : SUCCESS in 1h 25m 38s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4b7067cf8dce4b318505dc01e948244f : FAILURE in 24m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f6932bb609e04ab693ee97c27eb0c594 : SUCCESS in 1h 00m 08s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b2a7a98d2e18484b9a13d22e3fdb9c27 : SUCCESS in 1h 06m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54afc926451a40c4967f6ba2588c7ee0 : SUCCESS in 1h 04m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/daa6df3e5d694eeb82ea9a60d7b7a6b4 : SUCCESS in 1h 12m 07s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfb4b90798aa4aae8c942791078d3f39 : FAILURE in 1h 00m 55s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00ad5bcfd01b458f9509d42b57d49798 : SUCCESS in 1h 31m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b9f8319b65082bc2edbebfe86afe11822f36218","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-02 06:53:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79d3518339e0055fc1a3e2654192af54fab38ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 08:02:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0420deebfe147be94f85294269aff5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b90aa386c4b54c54b2183c43875a109c : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8aaa56e7ff50435fb11ece786634d422 : SUCCESS in 3m 41s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29f2dce9701d4cd2923e75999a1dab02 : RETRY_LIMIT in 10m 06s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/37eb7b512b224a068b02e288ba118df8 : RETRY_LIMIT in 8m 17s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/44dddb8de0f94f7a8fd22fadb588a6d5 : RETRY_LIMIT in 16m 13s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6226bb819d17474b9ce3cfcd1e97c8e7 : RETRY_LIMIT in 10m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b42516457e964c149a98a80e822f7f9e : RETRY_LIMIT in 5m 54s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/2f486615722d4910b26e8b43467a4413 : RETRY_LIMIT in 11m 18s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e66c91ce5034b6eb9d71639ad73ea2d : RETRY_LIMIT in 6m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/285b9ae6868f463f837f98ec939449f7 : RETRY_LIMIT in 6m 25s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/62bdd5d45e504caa8bdd3958477ddda6 : FAILURE in 8m 01s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d5f2ff854bc41e3854597804b0c2276 : FAILURE in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77dca01f3dd67125f59e61e1bcd4e35b221db7aa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-06 16:08:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"324f41c69a82b014f96ea2faf09314e0f0197a93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 17:44:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bcb6b1be2354b10bb385aacbef80e2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c91d1d322d04179ad170d9fbb0a8194 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b1814a6463b454e8e74d18fa280dcbe : SUCCESS in 4m 41s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e05981a0ca704a578ec6fc2f5f7d8813 : SUCCESS in 13m 47s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b95cfce084ef4c60844b4d778222084b : SUCCESS in 1h 28m 02s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6530d1f4fe0d4229a1ad63bd3bcc8aeb : SUCCESS in 1h 23m 28s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/23bad8bf8b3446798461325157fe7677 : SUCCESS in 1h 13m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1bad8e46a7d44fecafdcfaa4949356d8 : SUCCESS in 1h 02m 24s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d0c06ab16be34de78fac82ed7b1e200d : SUCCESS in 1h 04m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b01dd688bb54eaeb259a20fc37f2809 : SUCCESS in 1h 07m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42e478c4858a4c6eae4a77035b4e67e1 : SUCCESS in 1h 11m 47s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae30dd0f0bf44e14aefe43922855519b : FAILURE in 12m 12s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3d4bcf7f65c47c4a1c0e28d523e3275 : FAILURE in 7m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39c8899702bf7cd5bbe80ac2175eabc02f3eff33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 17:46:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"309c0062429eceb1dd58c6235754259f9296afe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 19:44:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a95d877ea1d4fd8b85614c684fa8342\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed916f6de674a89a7a640f021a6987e : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9231d581208b46318397da8a22da0cad : SUCCESS in 4m 23s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d1e05e2cf734e31afab68a9ee4c93e2 : SUCCESS in 15m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/63a1d084287b46d89de0d7e52b03d0b5 : SUCCESS in 1h 30m 16s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a4d27b538f246b9b0ce9306e43ea643 : SUCCESS in 1h 11m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a28bb82e6e4746788f63918f206eac2d : SUCCESS in 1h 12m 38s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/af90704e6d2f4914abc59f84f99c5eb4 : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9409d7acc034ac4ae43d2ea497d9889 : SUCCESS in 1h 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"4a086da5e086179d945f5e4a5192339725b637d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 19:44:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3f5f4bab1f02ee064e479aa381bc0dcf4f4d44d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 19:46:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68af96867ae0493e8be6b49f096cb2f0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49b0212ddb154d04946f4c9ad7888e21 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a307f9fc299f0c76c7a145ccf4bbf84684bf482","revisions":{"5a307f9fc299f0c76c7a145ccf4bbf84684bf482":{"kind":"REWORK","_number":1,"created":"2023-06-28 07:59:32.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/44/887144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/44/887144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/887144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/887144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/887144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/44/887144/1"}}},"commit":{"parents":[{"commit":"a3e7a1f41bf86a2dbb70b919ae084af2b6a94275","subject":"fix indentation for condition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a3e7a1f41bf86a2dbb70b919ae084af2b6a94275"}]}],"author":{"name":"Simon Hensel","email":"simon.hensel@inovex.de","date":"2023-06-15 13:47:49.000000000","tz":120},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2023-06-28 07:59:26.000000000","tz":120},"subject":"Add optional compression to mariabackup","message":"Add optional compression to mariabackup\n\nAs database backups can grow substantially in size, compressing backups\nhelps to preserve disk space.\nWhile the mariabackup utility offers no compression by itself, we can\nstream the backup into a compression tool to create an archive [1].\nThe xtrabackup_checkpoints file, which contains metadata on a backup,\ngets stored alongside the archive, allowing to create incremental\nbackups from non-compressed backups and vice-versa [2].\nOne thing to note, is that compressed backups cannot be prepared in\nadvance, this step must be manually carried out by the user.\nBackup compression is disabled by default and different compressors\ncan be chosen (zstd, xz, ...), with gzip being the default.\n\n[1] https://mariadb.com/kb/en/using-encryption-and-compression-tools-with-mariabackup/\n[2] https://mariadb.com/kb/en/incremental-backup-and-restore-with-mariabackup/#combining-with-stream-output\n\nChange-Id: I28c6a0e0b41d4d29c3e79e601de45ea373dee4fb\nSigned-off-by: Simon Hensel \u003csimon.hensel@inovex.de\u003e\n(cherry picked from commit 60009ed7cebe9c082592fd564b1577068ef94b6c)\n(cherry picked from commit 41801ab04d3c64d61b881d410691e05e0312114d)\n(cherry picked from commit e21d80924c08aa0af809389bc068217d491b6080)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5a307f9fc299f0c76c7a145ccf4bbf84684bf482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5a307f9fc299f0c76c7a145ccf4bbf84684bf482"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
