)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~887861","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Id5ae73222a1109ad13b0b70ba3d02063d931ff90","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"osa/core-2.15","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-07-07 06:49:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id5ae73222a1109ad13b0b70ba3d02063d931ff90","subject":"Remove warn argument for command/shell","status":"MERGED","created":"2023-07-06 16:18:58.000000000","updated":"2023-07-07 06:50:45.000000000","submitted":"2023-07-07 06:49:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887861-osa/core-2.15","meta_rev_id":"ef1954d36da7f4eb0ae7f06c300073f0458d43b0","_number":887861,"virtual_id_number":887861,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-07 06:49:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2023-07-06 21:51:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-07 04:24:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-07 04:24:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-06 19:38:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-06 21:51:26.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-07-07 04:24:00.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"e56fa89041587d91eadebc47df59fa4ab36255d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-06 16:18:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9f5a2c543afbbef5534b4d231ea3d3be4c25b87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 19:38:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/983a81a869d1453b911e926649d416eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/390a858f2d154809800dced9849af923 : SUCCESS in 8m 19s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aacee20623a148ac89dd8b8ef552b3c9 : SUCCESS in 11m 34s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/534fd2a8d9424f649b90ac43901b658d : FAILURE in 51m 49s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7ebb4fe2321b434dae2152d33f09c5a6 : SUCCESS in 1h 52m 00s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5914aa4c51084ec0bd950615a4acda12 : SUCCESS in 1h 52m 34s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/427199d2e2d34818941d6218d3a20363 : SUCCESS in 1h 13m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e53f0d239134cb6bd452e3b0df894ed : SUCCESS in 1h 23m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b42de0bea8246d489d8256f9fb1b453 : SUCCESS in 1h 25m 48s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9122af658ce547fb9a18accf104a9b6a : SUCCESS in 1h 25m 56s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3af31f93c7e9487cb0722f524b37ecc7 : SUCCESS in 1h 15m 03s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5ee54795668943c6882903a8c94404a2 : TIMED_OUT in 3h 06m 03s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38af3fd861704e7c9fc212ab8417fdc1 : SUCCESS in 1h 48m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a2d26fdf17a781552dc7a64fa6ccdfdbfa7175c4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-06 20:05:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d90626fc1405672bcc4a3ecd50e5df16e3271a88","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-07-06 21:51:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4590030bc70217ba04477b8082f38b58bd65b76e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 22:22:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6e1571081ad4a01a8417aa509fe48d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734e7ed53fb142b191f24fe2f17df0a7 : SUCCESS in 8m 31s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/583f211791de4a97ae2998e063db3bd8 : SUCCESS in 14m 20s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/aacd2f92aa694683942eef36419e3b52 : FAILURE in 47m 55s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a1ee3a80966c43bb814ff6d546b276c4 : SUCCESS in 1h 48m 05s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6e75201400024abba9699372379096b2 : SUCCESS in 1h 50m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a4891e9358e74110b21f73b3bb5eb097 : SUCCESS in 1h 20m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05e9f8b034154cc2a4a3cb0553be67bb : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1eed12c1f244477691e5ec2dd0cf4e1e : SUCCESS in 1h 14m 57s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c665ea9449054dd4a688d4136c29d422 : SUCCESS in 1h 18m 54s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e3de2f1e0c8491a80ecc8744e28e2a7 : SUCCESS in 1h 13m 50s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/673522c827b44511af5ac33ebe25fd9a : SUCCESS in 2h 06m 46s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65d9ed0d9d334803bf7d0d3d2933b083 : SUCCESS in 1h 26m 18s","accounts_in_message":[],"_revision_number":1},{"id":"52aaaaf5c7d28cf90f0aa4ffbacc4ab7448e16a5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-07-07 04:24:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ccf660fd80e6b47f68a8514accfee528759204e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 04:26:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a946326a0ec9779e4b900cd4e2bc604997588959","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 06:49:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45ecab4c79344fe6a84eee8356bb539b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b4927cc1fad4af782ab2995b2d40fad : SUCCESS in 4m 36s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e62624c155a74baf8c9be24ce7064e7c : SUCCESS in 26m 40s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/da15ac5cc57349c9b748841f0117035b : SUCCESS in 1h 34m 02s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/25a1b2ef0daf4799bc873d7451c2b112 : SUCCESS in 1h 41m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8a3bec261ad34383a42bc60ae3dea2f2 : SUCCESS in 1h 05m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b91c30fccafc4bea83c0d44684502628 : SUCCESS in 1h 05m 30s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/579a2045a7124313806c1a64ca739fac : SUCCESS in 1h 11m 53s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7c7f0d9d58a4571a35eae102e92e0ea : SUCCESS in 1h 35m 06s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b06e5138172a427e8a79b87b9b8011ca : SUCCESS in 2h 10m 12s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01e10be8e07e4c6d82c42559b72d285c : SUCCESS in 1h 30m 03s","accounts_in_message":[],"_revision_number":1},{"id":"64ed03eb1715190747dc4bbf69016d9af5ec22ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 06:49:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ef1954d36da7f4eb0ae7f06c300073f0458d43b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 06:50:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c2569e78cdf4a9791d327dac44e6c78\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f3441dbbf1a4fee8d5ebf080c91581e : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cef3aa94f6a39b62f8a3fc337d6b00b4dacb7226","revisions":{"cef3aa94f6a39b62f8a3fc337d6b00b4dacb7226":{"kind":"REWORK","_number":1,"created":"2023-07-06 16:18:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/61/887861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/61/887861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/61/887861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/61/887861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/61/887861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/61/887861/1"}}},"commit":{"parents":[{"commit":"60009ed7cebe9c082592fd564b1577068ef94b6c","subject":"Add optional compression to mariabackup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/60009ed7cebe9c082592fd564b1577068ef94b6c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-06 16:18:48.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-06 16:18:48.000000000","tz":120},"subject":"Remove warn argument for command/shell","message":"Remove warn argument for command/shell\n\nSince ansible-core 2.14 you can\u0027t use warn as module argument.\n\nChange-Id: Id5ae73222a1109ad13b0b70ba3d02063d931ff90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cef3aa94f6a39b62f8a3fc337d6b00b4dacb7226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cef3aa94f6a39b62f8a3fc337d6b00b4dacb7226"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
