)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~896405","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2F2023.1~Ib15c60dc577b376b1f761c4266eea89c4cb0be9f","project":"openstack/openstack-ansible-galera_server","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"32755":{"account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"last_update":"2023-09-28 11:44:55.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2023-09-28 11:44:55.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-09-28 09:34:27.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ib15c60dc577b376b1f761c4266eea89c4cb0be9f","subject":"Added vars to override systemd for mariabackup","status":"MERGED","created":"2023-09-25 14:01:12.000000000","updated":"2023-09-28 11:45:56.000000000","submitted":"2023-09-28 11:44:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896405","meta_rev_id":"fdeddc07c3b52b20526367ff277060e30933b898","_number":896405,"virtual_id_number":896405,"owner":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-28 11:44:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-09-26 15:29:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-28 09:34:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-28 09:34:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"CC":[{"_account_id":36350,"name":"David","display_name":"David Hitze","email":"ubuntuone@hitze.cologne","username":"dhitze","status":"inovex GmbH"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 14:01:12.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":36350,"name":"David","display_name":"David Hitze","email":"ubuntuone@hitze.cologne","username":"dhitze","status":"inovex GmbH"},"state":"CC"},{"updated":"2023-09-25 14:03:17.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-09-25 14:03:17.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-09-25 15:47:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-26 15:29:50.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"93a39307485e69aef896fc57797fbcbb0224ff2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-09-25 14:01:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"164f9da7d3df7633e15a8d86f182301307ccf999","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-09-25 14:03:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"004af3d503b0a2e2b45e77a41580833871855e96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 15:47:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9eb73077fea43b3b52333f26e2a76df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a057d52cd71f4f79bf01fd5e6548ff2b : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d66747c8f4334f68a8a698cda75f3954 : SUCCESS in 3m 47s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aafdb15d5fa14780baa13d30d79d1ab4 : SUCCESS in 10m 04s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b8518621fc44037bcf5febcf20e380b : SUCCESS in 1h 32m 58s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/157a340a4e5146c5bf68943145e7f74c : SUCCESS in 1h 32m 41s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8d7577e661a24ef48fa1e9e3915b23eb : SUCCESS in 1h 39m 19s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/66a025c807c2432e82e92a0ef52977d0 : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8159f7ee7bc48e883c4d3aa01d6f76c : SUCCESS in 1h 01m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20a7e8dbbcd4461b82fcc3eeabcdf6ae : SUCCESS in 1h 14m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97e99cfd87aa4c71a9681b55b2d17e63 : SUCCESS in 1h 04m 14s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/488127d7f18a482b9b104610a71fa315 : SUCCESS in 1h 15m 13s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2900f2705f9449f9bdca7976bffa0735 : SUCCESS in 1h 18m 48s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c9b75fb90de45e494ed0a3574e0e3d7 : FAILURE in 1h 11m 26s\n- openstack-ansible-upgrade_yoga-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c11b3eab78014ff19f4b1887f4d0e6ed : SUCCESS in 1h 21m 58s","accounts_in_message":[],"_revision_number":1},{"id":"db06b22911d1407773c269009ca75de276a6e393","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-09-26 12:26:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc6d21aaaff86bff80ff9f5703323fdfc015a1e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 14:31:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5838ca2dc7be4acfa2df6e7c047097e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ad69ae7c6834ca298495e9f42477a5f : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/088138c3776b449a9dd5aa45a6f5a4eb : SUCCESS in 3m 09s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c35e1f89265b4e51bd35adf6d986d662 : SUCCESS in 11m 31s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8841bf6b4d664818981705032240802b : SUCCESS in 1h 32m 10s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b2ae04e7745c47d49db6138cacbeb7e5 : SUCCESS in 1h 33m 12s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b94ae5fad3454ab8b2c391cfba25b01c : SUCCESS in 1h 33m 59s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d58a8300d3d14ff0b8dc175b50e745f7 : SUCCESS in 1h 14m 52s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bfdeae1f05347f5a4e7e3c820f63732 : SUCCESS in 1h 11m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c56b3c18cb8b40b7b4064c8b9590293e : SUCCESS in 1h 14m 42s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/474da461ad384d37a4e62c2c8743313a : SUCCESS in 1h 24m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be1222ac0845483a8cad6f5d89d2f9de : SUCCESS in 1h 19m 36s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8d3f132e75274223ae4e1c1a8b8e4d25 : SUCCESS in 1h 57m 06s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1be71bd0feaa4011a3ae68c739e9efd4 : SUCCESS in 1h 34m 30s\n- openstack-ansible-upgrade_yoga-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/622e6272894440bd9ea0a8a4fd7e8857 : SUCCESS in 1h 32m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a28f3680aded66f5a941c116ef42346630f3e7b6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-26 15:29:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a07212c8b36888e6edc1bff6fcb17ce9b5b7bae5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-28 09:34:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"45589289c2cb3a1a542cdf90c790cf05e70632cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 09:36:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47a512d323f558136884f9bfd1a025de2d45ee27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 11:44:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43837acad6794475813da5b24297daf8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b1f5fb884f4485b69036538d325bb8 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a93b742c85ac4001b8549ed9da144bda : SUCCESS in 4m 03s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0999a4c5e35f4fd2a078d3a52a4a9d5a : SUCCESS in 20m 37s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d458ef9956db4df095c0159ef6867aa5 : SUCCESS in 1h 39m 21s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b63be49bdd214f94a85c2fe6258eb81d : SUCCESS in 1h 34m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5fc132a1f7b7497caf2364e8b872889d : SUCCESS in 1h 13m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a55487484054a27a3255ae15048d3fe : SUCCESS in 1h 21m 31s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb1deb89eeb54d538cca73ba2c1ca79e : SUCCESS in 1h 36m 56s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15ad9a696afe49f3bde5d49a80a8418d : SUCCESS in 1h 13m 03s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/90b1d464b3454fce80e13e0fa8644532 : SUCCESS in 1h 43m 51s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65dd112cb2174586937e9325b65acbc2 : SUCCESS in 1h 40m 48s","accounts_in_message":[],"_revision_number":1},{"id":"aacdf040cee963d9045c66b1a47507c13ed2d769","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 11:44:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fdeddc07c3b52b20526367ff277060e30933b898","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 11:45:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3059065fcff14a6193ba5333adcf9cb3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ef275d151094a4cba64e3943b1f4cb4 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5f4e4d737f4a2f8fc6ded3639904b7e429048d9","revisions":{"b5f4e4d737f4a2f8fc6ded3639904b7e429048d9":{"kind":"REWORK","_number":1,"created":"2023-09-25 14:01:12.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/05/896405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/05/896405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/896405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/896405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/896405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/05/896405/1"}}},"commit":{"parents":[{"commit":"cc10018c2cb14847eda98d3b8598b8d216f48188","subject":"Replace libgcc1 with libgcc-s1 for Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/cc10018c2cb14847eda98d3b8598b8d216f48188"}]}],"author":{"name":"David Hitze","email":"ubuntuone@hitze.cologne","date":"2023-09-11 08:52:14.000000000","tz":120},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2023-09-25 14:01:02.000000000","tz":120},"subject":"Added vars to override systemd for mariabackup","message":"Added vars to override systemd for mariabackup\n\nAdded variables ``galera_backups_full_init_overrides`` and\n``galera_backups_increment_init_overrides`` that can be leveraged to\noverride default set of systemd unit file for mariadb backups.\n\nChange-Id: Ib15c60dc577b376b1f761c4266eea89c4cb0be9f\n(cherry picked from commit 3e2afc1e4edb152693d4e77a35340791dc80e408)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b5f4e4d737f4a2f8fc6ded3639904b7e429048d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b5f4e4d737f4a2f8fc6ded3639904b7e429048d9"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
