)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~900266","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I09603c55a2f5c45bbb130ac23d3b838086bd2637","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"ca_cert_client_cnf","attention_set":{},"removed_from_attention_set":{"14805":{"account":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"last_update":"2024-07-11 22:49:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I09603c55a2f5c45bbb130ac23d3b838086bd2637","subject":"Include CA cert in client my.cnf","status":"ABANDONED","created":"2023-11-07 06:21:21.000000000","updated":"2024-07-11 22:49:48.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b320854b18e737bd8050870daf59d9de7f75f247","_number":900266,"virtual_id_number":900266,"owner":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-15 00:47:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-07 08:06:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-07 14:36:17.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"}],"messages":[{"id":"7e1a03ecbcdba807fca9951034a19586e5aeb54b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2023-11-07 06:21:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8628421ca5790e0f7e1eb606684e8cc751193204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 08:06:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39876f896a3e411282d22a6e52d4d29a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a96b29ac7f4d6da13ba29b23e84253 : SUCCESS in 4m 43s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58f9aa57beb94e8780f7e28b43a9850a : SUCCESS in 11m 45s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bccd8a002834fa8bb0e6e1032beaf0c : FAILURE in 1h 00m 26s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/889b143f4c2a40cf94bc9b301d86ab5c : FAILURE in 1h 01m 26s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4cb65ebf65e04aa3ac25bdd601c186e3 : FAILURE in 1h 02m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6fc893257d83456ab6d0d959620b3337 : FAILURE in 39m 16s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ded7d89539444f79853208fcb38cbf85 : FAILURE in 53m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/909a90ae9f494ad4b94df7e0dfc339da : FAILURE in 50m 02s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c1343b2056046d9b081405d37e416b1 : FAILURE in 49m 55s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a76211a5196471eb56a5ef2338bd13d : FAILURE in 53m 35s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/130ec7eab2b74f3e8dcb05530dc85e18 : FAILURE in 1h 38m 58s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77a8a68fede94333816fb866fbe87b2e : FAILURE in 1h 05m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3d74f6a3715741544df483c7ebf247238e3b4609","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-07 14:36:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73933c611af0ca6ffe736ad82f38848bb8673461","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2023-11-14 22:35:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"af353007561f68ae926fb586a2fa5341ace98efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 00:47:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37e738ea36f54169a3fd3ddf0eec8a5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3d5cd00eeef47b3b0eb1551172bc9fd : SUCCESS in 4m 13s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bf862a0e5034297b96d1fca80967280 : SUCCESS in 12m 16s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ca21d25f25e4366a91ef04a7b06fee7 : FAILURE in 54m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5e1a3b7663fb41fba3a8320ffecdde2f : FAILURE in 1h 12m 31s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fb3d3c4b17cc452393f4e6b7af52ba40 : FAILURE in 1h 04m 22s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7dc682c48e844eacaea018f8457c63e7 : FAILURE in 42m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/cb05979cc26540e2ad67648925157b17 : FAILURE in 47m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e64b68a20773406bb5484fa18e5d96f5 : FAILURE in 46m 28s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19abb254b825461b9e46e091c187dadc : FAILURE in 1h 00m 50s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8152bd73769942d7a23fd6371f4af5e8 : FAILURE in 47m 18s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5e8f50c17dd2449185dd8673214a3ca4 : FAILURE in 2h 02m 49s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aaf77eeee64f4b4aa2c9e8929d182192 : FAILURE in 1h 10m 47s","accounts_in_message":[],"_revision_number":2},{"id":"b320854b18e737bd8050870daf59d9de7f75f247","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2024-07-11 22:49:48.000000000","message":"Abandoned\n\nno longer needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4fa391416bce9caaa471b0777809de7f43bccb9a","revisions":{"7a81e6e0d3e4102861b0d2482c67b64cbe5196a7":{"kind":"REWORK","_number":1,"created":"2023-11-07 06:21:21.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/66/900266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/66/900266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/66/900266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/66/900266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/66/900266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/66/900266/1"}}},"commit":{"parents":[{"commit":"28ac2fc7ee661f19170037da489a7a1644d7d84b","subject":"Bump galera version to 10.11.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/28ac2fc7ee661f19170037da489a7a1644d7d84b"}]}],"author":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2023-11-06 19:14:00.000000000","tz":-480},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2023-11-07 06:18:48.000000000","tz":-480},"subject":"Include CA cert in client my.cnf","message":"Include CA cert in client my.cnf\n\nTo verify the identify of the server when galera_use_ssl is true,\ninclude the ssl_ca option in client my.cnf file.\n\nChange-Id: I09603c55a2f5c45bbb130ac23d3b838086bd2637\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7a81e6e0d3e4102861b0d2482c67b64cbe5196a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7a81e6e0d3e4102861b0d2482c67b64cbe5196a7"}]},"branch":"refs/heads/master"},"4fa391416bce9caaa471b0777809de7f43bccb9a":{"kind":"REWORK","_number":2,"created":"2023-11-14 22:35:10.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/66/900266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/66/900266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/66/900266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/66/900266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/66/900266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/66/900266/2"}}},"commit":{"parents":[{"commit":"28ac2fc7ee661f19170037da489a7a1644d7d84b","subject":"Bump galera version to 10.11.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/28ac2fc7ee661f19170037da489a7a1644d7d84b"}]}],"author":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2023-11-06 19:14:00.000000000","tz":-480},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2023-11-14 22:33:10.000000000","tz":-480},"subject":"Include CA cert in client my.cnf","message":"Include CA cert in client my.cnf\n\nTo verify the identify of the server when galera_use_ssl is true,\ninstall the CA certificate and include the ssl_ca option in my.cnf on\nclient hosts.\n\nChange-Id: I09603c55a2f5c45bbb130ac23d3b838086bd2637\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4fa391416bce9caaa471b0777809de7f43bccb9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/4fa391416bce9caaa471b0777809de7f43bccb9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
