)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~900890","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fyoga~Ie997393935e04e127893643e4c72d7af07e993ff","project":"openstack/openstack-ansible-galera_server","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-11-16 22:06:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie997393935e04e127893643e4c72d7af07e993ff","subject":"Fix ignored database directories configuration","status":"MERGED","created":"2023-11-14 15:15:01.000000000","updated":"2023-11-16 22:07:11.000000000","submitted":"2023-11-16 22:06:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900890","cherry_pick_of_change":900860,"cherry_pick_of_patch_set":1,"meta_rev_id":"4e8d46e65b661950f5ddee6c45540c97772c979e","_number":900890,"virtual_id_number":900890,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 22:06:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-11-16 18:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-16 18:21:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-16 18:21:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-14 15:15:01.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2023-11-14 17:27:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-16 18:21:57.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"d4c1ae4a68af2647165fbf3608b9bdac8577b3bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-14 15:15:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"975f97352f70f6486e3243fc269b5e4b990b91e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 17:27:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a67d7e8487e1491ead07039205e6c5f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6fadb0857d649d98f38d3ca407db94d : SUCCESS in 6m 22s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5182d0c513044c5a8cd2f04b48cd6e7c : SUCCESS in 11m 10s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9ef7cc684344edba8611e88d37bc506 : SUCCESS in 1h 22m 55s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c3eac53c8e944999e9d7ba7118d62ee : SUCCESS in 1h 17m 05s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/71dcedff7d3d4d25a96e83ae37b4cb70 : SUCCESS in 1h 30m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/791a36ed6eda46e69b45b986762251c6 : SUCCESS in 1h 01m 23s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/cce2635d9f734f0a9e28fb0028179d2d : SUCCESS in 1h 23m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85b534196f3b4701a673af1988360796 : SUCCESS in 1h 02m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/514663948be043e7b4687b124e2e12dd : SUCCESS in 1h 02m 20s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2cc7a6980aa148d2b5db232f603d84ed : SUCCESS in 2h 04m 27s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4deb240b7bb743f691540494dbce62d2 : SUCCESS in 1h 37m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dafa2cd850dbe438a11c44a8d1785ed7e3dfe256","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-16 18:01:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1216deffd47fa33ab62782a6bc1eb70643b851c1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-11-16 18:21:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"395a0aaabef4be0c7b9070466c773c0b0b65480e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 18:23:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50128fca254b64483656b8e0d26362b478c6a668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 22:06:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc6023f1583e43ae85ad0fbb9348147a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c6f95c3006c4ff59d626d05c03e73c5 : SUCCESS in 4m 26s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b6131aa8e1645cd912f5e249655be3e : SUCCESS in 10m 03s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/6364be7825294d2c87e63ebd20b6859c : SUCCESS in 1h 26m 29s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0d42801d3dfb43c29123e13259e33430 : SUCCESS in 1h 35m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a601aaaa10784bea82f5288d25e02e18 : SUCCESS in 53m 32s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/9aa63537d3184b9c93c31248676a5264 : SUCCESS in 1h 22m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0873e602db2243ba94f90ee1de9cecf1 : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":1},{"id":"5cee733d947a8b3ccbebd98b2830fbe44b554014","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 22:06:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4e8d46e65b661950f5ddee6c45540c97772c979e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 22:07:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a9cc5a51ac04faab87bba833fc40657\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15ab0e16c3c24eddad6259b0c8221f32 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b3c36324828b0eef949ecc91f878120a44259d7","revisions":{"9b3c36324828b0eef949ecc91f878120a44259d7":{"kind":"REWORK","_number":1,"created":"2023-11-14 15:15:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/900890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/90/900890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/90/900890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/90/900890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/90/900890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/90/900890/1"}}},"commit":{"parents":[{"commit":"012c5adb2be1706be2a39fc054e2c1251149d332","subject":"Added vars to override systemd for mariabackup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/012c5adb2be1706be2a39fc054e2c1251149d332"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2023-11-14 09:02:08.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-11-14 15:15:01.000000000","tz":0},"subject":"Fix ignored database directories configuration","message":"Fix ignored database directories configuration\n\nConfusingly, the variable ignore_db_dirs is set by passing it\nmultiple times in the configuration file, once per directory.\nIt is then read as a comma separated list, but cannot be set\nin this way.\nhttps://mariadb.com/kb/en/server-system-variables/#ignore_db_dirs\n\nWithout this, the mariadb-upgrade script can fail as it attempts\nto process invalid databases.\n\nChange-Id: Ie997393935e04e127893643e4c72d7af07e993ff\n(cherry picked from commit 229ae217c1e127979cd5370414e6d2b13c910055)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9b3c36324828b0eef949ecc91f878120a44259d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9b3c36324828b0eef949ecc91f878120a44259d7"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
